)]}'
{"id":"openstack%2Fdesignate~689670","triplet_id":"openstack%2Fdesignate~master~If3122f4719b1d52c5972e2d7869b521822de9422","project":"openstack/designate","branch":"master","topic":"move_cfg","hashtags":[],"change_id":"If3122f4719b1d52c5972e2d7869b521822de9422","subject":"Move pecan_debug into cfg section","status":"MERGED","created":"2019-10-21 04:12:07.000000000","updated":"2019-12-16 19:35:54.000000000","submitted":"2019-12-16 19:33:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689670-1576524811063-0548d2f5","meta_rev_id":"44fdb99ddc3dcaf7124ca214afc59bd6db44fc90","_number":689670,"virtual_id_number":689670,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-16 19:33:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-12-16 15:54:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-16 18:21:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-16 15:54:20.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-12-16 19:33:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"23ed42809d9681d80c45c129b1c1427fdbeae8de","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-21 04:12:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"293744dbc9c46047a4f0c9b439e85d50dc6f745d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 05:08:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/92305ffe3ab9479d98b889b54bd21831 : SUCCESS in 35m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/792eb3bd1da64133835a08f032792c08 : SUCCESS in 33m 49s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/ae68b37aa1134617b93f606e978664c4 : SUCCESS in 35m 12s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/f14b0eaaa84e44a9aa648018f79dfbe8 : SUCCESS in 38m 12s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/54e9f0ba979e490987dcb3e95b1a497d : FAILURE in 27m 50s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/50127baae3a34317b6821c1212e7d5f5 : SUCCESS in 55m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/33c006a0629c41f1a68b6cd0f06e7e46 : SUCCESS in 36m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/16d71fc98d4148a58ea25a6aa9e3a876 : SUCCESS in 37m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d39542aae8d447988657e57b170f636e : SUCCESS in 4m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78f737f2985d4818bb9841513581044c : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e8642416949455fa8d8296bc52bf4dd : SUCCESS in 3m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5a24f394752444678da5b7bf8e377fe5 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2136dba08fa478e888600ba26b237a6 : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3735bdddcadf4b8d9a10cba1bdfc24dc : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5db66f3877e4fa8b11b499502d35925 : SUCCESS in 6m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3f815c9798624907b8f3319b352c0c75 : SUCCESS in 32m 00s","accounts_in_message":[],"_revision_number":1},{"id":"0f41ae1925919abc92ad185d45a5447694e33b75","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-21 22:14:06.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"0b91ff8cc8599e98d29bea65aac5407e8d89f180","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 23:29:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/42ff772205db4762835201f56ce1fc99 : SUCCESS in 48m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9aaf9ed07073493fbd20be995301c1e0 : SUCCESS in 46m 17s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/e08de975a7e94a47a5ea98e7188b8a14 : SUCCESS in 34m 09s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/7ce6eda447e542aa805ab30db239ddbe : SUCCESS in 43m 18s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/1092b10b7a7b4763a419492a5167bf40 : FAILURE in 35m 41s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5f42f5a3cf684df7aa9e3915c10955dc : SUCCESS in 1h 13m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5d7001a2d7bc4cf186b026e7d460b5b5 : SUCCESS in 51m 10s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1475f34182c242df9766dff320758c59 : SUCCESS in 43m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53c82c2ab5fd45318bed1c97cb487ec0 : SUCCESS in 5m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7bc0a42d97c4875b806f0eff22af3c9 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dc830344570405c96ba50c95cec3088 : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/10178463e6514beca1dece115ffe265e : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8972b81228b427b85ab5b214bd6d6b6 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5c518e636734db5a9227cf5c72a74ab : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaad1e0a633c4ab7bf4ebb9d48d469f2 : SUCCESS in 7m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/28715b3d9a794c23a16ebdf07360d8f8 : SUCCESS in 31m 40s","accounts_in_message":[],"_revision_number":2},{"id":"26c8d8b6b66786697e4961be7fe1ecd3f1888c44","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-23 06:56:55.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1f688de55c714849e29b7e294b69429930412e61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 08:41:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b4fb5b578b9a4c3aa2c0a99f6cd1eeae : SUCCESS in 42m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5fedcdbc910c407e9943b4690856b5cf : SUCCESS in 47m 42s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/71f1bc6cea6d43269afcc72abc751172 : SUCCESS in 43m 15s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/70a6141753b84184868298f37c7e5c65 : SUCCESS in 41m 57s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/5edf17eca2c84227a345e57ee7ee4e73 : FAILURE in 30m 55s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5ee1c85a5f064c96844a46fc7f85aed3 : SUCCESS in 1h 01m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4e4d4a9ebc5c4678a3b3c57809fbc138 : SUCCESS in 37m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9f25eb0f4df4455cb713b6f681bf5324 : SUCCESS in 45m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/235a94a5b4b447cca60f38cd3140390b : SUCCESS in 5m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ba99bc5b0694189bdbdb0678d13d415 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fe067c8471840f4b8242a4add20465d : SUCCESS in 3m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/11d97e7c1b474174b1132db096f415aa : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/938bea617c814d83b6dd96c7d2877d64 : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3096a080628a49b6b8037277d3c40fd4 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4de5ff012684d24af0469c4791dc3b7 : SUCCESS in 8m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/edbb2901a9e64f729a8817fe61935179 : SUCCESS in 41m 00s","accounts_in_message":[],"_revision_number":3},{"id":"75384d5311a2c86aea86c9fdd177cb66c41eeefa","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-24 08:24:08.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"2135249b054ce0d2bcdc82905db2e0d572647b81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-24 09:18:08.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e9f5b7e871fa4ba7b476e941240a1223 : SUCCESS in 45m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ae6cfde3fd0044f29d469d0aeb898420 : SUCCESS in 48m 54s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/c044da735cd3472f940902a8c82eb31c : SUCCESS in 40m 45s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/d1cf918afe814ad9aaebe68efbd77ca0 : SUCCESS in 53m 18s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/49934fb82caa41858704b8abfa87ea6f : FAILURE in 25m 32s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1e95faf1dbb548e188887cd892b7d188 : FAILURE in 8m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/94c4df99fdc7448e9ac46afd2407ef82 : SUCCESS in 37m 28s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/403c1d00fa44488588252403cbb63605 : SUCCESS in 33m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e33e9d6dcdb42d1bbea180aab3e384d : SUCCESS in 4m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2bd11a451a24ffa8ac03ed0258cf3b1 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c80ad742aae54df7a6632a8341c5f47c : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f1c15ec0f95549769e8aa3cdb61b76df : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f578251eef98401c8be625c423123389 : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d02a8c6e99244902834c1a3479a18910 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98fe2f521dd049f7833982f41b57c7c8 : SUCCESS in 8m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/707455e80ddf44a797e549b9ef1ee478 : SUCCESS in 34m 02s","accounts_in_message":[],"_revision_number":4},{"id":"fbf4c6dbd873894b968adfe62d655d52b266dbad","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-27 18:39:36.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"41d6f0997ba4880ede1d8db60ea9cb505875fc32","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-27 19:48:38.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"850ca009bdfdbd4de6aab74e736d17b1b15b48f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 20:49:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ea963b8cd5854e4baa659f2de1a2d4ad : SUCCESS in 37m 43s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f90b9e22fb334054a452bb1b9421c1eb : SUCCESS in 33m 48s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/90c24de3df234e1db3f2787a72553417 : SUCCESS in 43m 43s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/9b792ba9e0bf47f69fb352259bf7f175 : SUCCESS in 37m 41s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/08333e455fa34cea97a8665593271d99 : FAILURE in 25m 34s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1cd4a17bedd84bc68f05cb8aeca9b4b5 : SUCCESS in 55m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a3e2f40c8cbd4b96a0284237820c6bab : SUCCESS in 34m 20s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/475f9e2e46f348129ba32a3280d173dd : SUCCESS in 44m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e820ce1a21dd4efabe5fcb94a16d47bc : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36d61eb91db142229b34a39f8875fdd5 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d0692d5d2764e30946199b115620e7a : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/43bc48d3e27740598b25ff1c4a674138 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5544285852444867a76faa299d5d380a : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/932d39bd212543309dd2113ffaaf39be : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba8f3284a14f414a8a519666286015af : SUCCESS in 7m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3f2abc12e46c4ee5bc6c29ff97f01468 : SUCCESS in 44m 25s","accounts_in_message":[],"_revision_number":5},{"id":"db12f4bc6e7c728a3ab2e0e94aadcca75c770d8e","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-12-16 15:54:20.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"5a4264fae85eb06af20c3731a1344f02d97030fd","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-12-16 18:21:20.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"236b505e3eeb93c5296df0cab0257ea107bb2ee1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 18:21:34.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"1e2fce7a99ed1ff6390029a23d14244738921d5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 19:33:30.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/28633963cf564f518d2dbe5b77a55b82 : SUCCESS in 45m 54s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/178ebf1e3eef4365b85535a415b4f9b4 : SUCCESS in 49m 08s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d797d606504848c68655e3bc4e01a6ea : SUCCESS in 1h 02m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/62f32ca8dde64664a3adde6bb3bb5063 : SUCCESS in 51m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/94de69bbc23f456e9675c478df9ab6db : SUCCESS in 43m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6f9400e3afb4e379ca4f0811e9492ae : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffec0d2088f443bab5987bcdd5e4e5be : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f3d81551b4949fd9f6a3e2a432f681c : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/50a475d90d7e47689684b89826c2b640 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/620a7aa450354192bff7a6d2ba5f3a88 : SUCCESS in 7m 48s","accounts_in_message":[],"_revision_number":5},{"id":"72611bc72c686c1322101a8e2f66d56f3faf3937","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 19:33:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"b710ad6154b5c489e856583b8917a75f5c48ff2a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 19:35:54.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb583a91b4ce45b5818366798d8445b7 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4301bbcf8a3679ae7b7a10f4fa3a5527cb121cb5","revisions":{"b3aa403334c9652511c6863e077202332740472b":{"kind":"REWORK","_number":1,"created":"2019-10-21 04:12:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/70/689670/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/70/689670/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/689670/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/689670/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/689670/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/70/689670/1"}}},"commit":{"parents":[{"commit":"7dc2f9ac9555a142b2716c3db919554152c09bf0","subject":"Merge \"Enable graceful shutdown of mdns\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7dc2f9ac9555a142b2716c3db919554152c09bf0"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-21 04:11:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-21 04:11:51.000000000","tz":-420},"subject":"Move pecan_debug into cfg section","message":"Move pecan_debug into cfg section\n\nChange-Id: If3122f4719b1d52c5972e2d7869b521822de9422\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b3aa403334c9652511c6863e077202332740472b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b3aa403334c9652511c6863e077202332740472b"}]},"branch":"refs/heads/master"},"cafc8b1f41a1d5a7ba1cd129e9ed11f4740852f0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-10-21 22:14:06.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/70/689670/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/70/689670/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/689670/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/689670/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/689670/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/70/689670/2"}}},"commit":{"parents":[{"commit":"68d8d8f4e5b896ca4c7363972d0ba88a7c5d3d50","subject":"Remove wsgi eventlet monkey patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/68d8d8f4e5b896ca4c7363972d0ba88a7c5d3d50"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-21 04:11:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-21 22:14:06.000000000","tz":0},"subject":"Move pecan_debug into cfg section","message":"Move pecan_debug into cfg section\n\nChange-Id: If3122f4719b1d52c5972e2d7869b521822de9422\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cafc8b1f41a1d5a7ba1cd129e9ed11f4740852f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cafc8b1f41a1d5a7ba1cd129e9ed11f4740852f0"}]},"branch":"refs/heads/master"},"54ebe5a0cc90f430aa8eb16a902f3d0b63c18966":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-10-23 06:56:55.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/70/689670/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/70/689670/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/689670/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/689670/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/689670/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/70/689670/3"}}},"commit":{"parents":[{"commit":"1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11","subject":"Merge \"Fix api version not providing the full url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-21 04:11:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-23 06:56:55.000000000","tz":0},"subject":"Move pecan_debug into cfg section","message":"Move pecan_debug into cfg section\n\nChange-Id: If3122f4719b1d52c5972e2d7869b521822de9422\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/54ebe5a0cc90f430aa8eb16a902f3d0b63c18966"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/54ebe5a0cc90f430aa8eb16a902f3d0b63c18966"}]},"branch":"refs/heads/master"},"f110bdf7c2d93220114aa38fdcf27c36c478255b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-11-24 08:24:08.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/70/689670/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/70/689670/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/689670/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/689670/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/689670/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/70/689670/4"}}},"commit":{"parents":[{"commit":"7951a7d7f5f870f5b97322e252a14eeb303413da","subject":"Merge \"Make designate reproducible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7951a7d7f5f870f5b97322e252a14eeb303413da"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-21 04:11:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-11-24 08:24:08.000000000","tz":0},"subject":"Move pecan_debug into cfg section","message":"Move pecan_debug into cfg section\n\nChange-Id: If3122f4719b1d52c5972e2d7869b521822de9422\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f110bdf7c2d93220114aa38fdcf27c36c478255b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f110bdf7c2d93220114aa38fdcf27c36c478255b"}]},"branch":"refs/heads/master"},"4301bbcf8a3679ae7b7a10f4fa3a5527cb121cb5":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-11-27 19:48:38.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/70/689670/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/70/689670/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/689670/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/689670/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/70/689670/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/70/689670/5"}}},"commit":{"parents":[{"commit":"d9fb34767c5bd9f8f822acc2b91715db45ab68db","subject":"Run grenade old side with python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d9fb34767c5bd9f8f822acc2b91715db45ab68db"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-21 04:11:51.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-11-27 19:48:38.000000000","tz":0},"subject":"Move pecan_debug into cfg section","message":"Move pecan_debug into cfg section\n\nChange-Id: If3122f4719b1d52c5972e2d7869b521822de9422\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4301bbcf8a3679ae7b7a10f4fa3a5527cb121cb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4301bbcf8a3679ae7b7a10f4fa3a5527cb121cb5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
