)]}'
{"id":"openstack%2Fdesignate~690008","triplet_id":"openstack%2Fdesignate~master~Ie8d815391b316d664b0c51099a08d8d32932c555","project":"openstack/designate","branch":"master","hashtags":[],"change_id":"Ie8d815391b316d664b0c51099a08d8d32932c555","subject":"Fix api version not providing the full url","status":"MERGED","created":"2019-10-22 08:22:20.000000000","updated":"2019-10-22 20:57:50.000000000","submitted":"2019-10-22 20:56:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690008-1571777779578-8b4ac2a1","meta_rev_id":"876cd68878fc1eca762961004eeef567d5c20a3a","_number":690008,"virtual_id_number":690008,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-22 20:56:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2019-10-22 11:23:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-22 17:32:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-22 11:23:36.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2019-10-22 20:56:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"559c6867e06dd7772eb160fb70c084bb4dc61f1b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-22 08:22:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9222110f889af01598aec6d24ec7b561ff87a7c9","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-22 08:40:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cd07698d37bbe352cbbb8fc0e3775a0d7631c50c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-22 08:44:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"114a34c493e77fb62ca044740cd26bf381f2435b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 09:58:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e4b1567113074a9d8a9de10c1632aaf6 : SUCCESS in 41m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/97d0f896c6624762b4b1a752aca6cf2e : SUCCESS in 38m 59s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/11a89aeff74945ac9162778ea8d8d024 : SUCCESS in 48m 18s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/514b0e7132594c6fa952330fc9cc8a40 : SUCCESS in 45m 32s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/09a6ca34ce584b63a8ac7308f857dc8f : FAILURE in 40m 00s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1c72dc0ed08f4a5fbed33e25600fe210 : SUCCESS in 1h 08m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4dee2926a2f94eeebdbaa90307ee2e9c : SUCCESS in 38m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a2ae8ce4dfef4d9fa1767abd73369c16 : SUCCESS in 43m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66d9eb481a8441e8b2c7df603c78a017 : SUCCESS in 6m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8c1b02e29cc4632917529619bd3aa64 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/505fd1234b4e4be1bda8d044ddf12ec7 : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b93aa3398ff4402bb53ab9b8e2ab1b95 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/421cb4fdd4ea40dd827367a58f47bfb2 : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5f643f3d698b4adcb1869a25aaf5d4dc : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f000d1868fc45ac87f221a6eb60db9c : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e92af9ccdae641499d3c9dcf5b48de60 : SUCCESS in 5m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dfe0d7be1e4a4358a4579c30ea228cba : SUCCESS in 35m 58s","accounts_in_message":[],"_revision_number":3},{"id":"860608cbe7d0612b3964d8443f1c450661ee6093","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-10-22 11:23:36.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"923bcc2001941f08be8c508b3e30dde991a3acba","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-22 17:32:49.000000000","message":"Patch Set 3: Workflow+1\n\nThe releasenote filename is a little incorrect, but will fix that in a follow up patch.","accounts_in_message":[],"_revision_number":3},{"id":"fcb3c1d50c65cf8c076f9fa8611cac2c9b5ad374","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 17:33:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d95cab1f5791b6ef2dd7f3a1a26ee9ca5034d22f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 18:36:21.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0e39b1fe843c4b9d9cb403b9b08b1247 : SUCCESS in 36m 34s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/815a9b53433f4221aad3236e9903b5e8 : SUCCESS in 38m 32s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/62c958d6b4784fc2b49e9c3a94c94b97 : SUCCESS in 40m 41s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/2d86f503c8994777a4d0771cfd4bd393 : SUCCESS in 34m 37s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fca7b8823bb249a4badb6b4db6bc3b39 : SUCCESS in 58m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/394ba0d5de4a4463b0e78896c31f3d6c : FAILURE in 10m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e17c2ab7697b43309744acd5f9397078 : SUCCESS in 53m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe60e4ce7e444b5c911db75afe13fa9b : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f6eb18c1e914c018fa8cad9f62c6283 : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cd4adb1b2fe840fba4f23ea0c4f76f77 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ca68a4fa3bc47acb5f22a10e8809c5a : SUCCESS in 6m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2de3de08a48a49f0b727ea84a4a436f8 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f51756c6fd2448a98a5ba5769c6e6dd : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6654886019b40d486c8cc4e70cac691 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":3},{"id":"480db03b619740ba27a8c9bd3d683d47826f1c3c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-22 18:37:15.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"0e5b988fc4dc55a8c9c6965f76c218a5d03e22e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 19:43:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cbc2efdf3f604ac3af36890034967b24 : SUCCESS in 45m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5083f28e7d104fe1a1aa27c704ea7a50 : SUCCESS in 41m 47s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/2c9bd8fbd115408a89d60c4e2d413577 : SUCCESS in 41m 41s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/eee5ed289eff42ba9ca074a16a4d250a : SUCCESS in 40m 47s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/2d237767eb5a4781b9eebafea62d7b6d : FAILURE in 32m 16s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0ee513447a0148c6a3c4255853e17367 : SUCCESS in 59m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/efc6d99224cb4a778642f4407cc5d818 : SUCCESS in 38m 51s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/86b7cf01fd244fcfbd3850cf35565e13 : SUCCESS in 39m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75d614d782774916aca5148f8f1cc998 : SUCCESS in 5m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40f24d3cef9f423681fed9412a77a297 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b21b3b26be8740918bc693ba45776468 : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b36660b909084426b93fbf17c64c9431 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5de35536537d4af096d118778921f88c : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fda504422cf5464c8f14c1830364a5f9 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e479a011ddde4f7e8d527399265c80df : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42a115a0fe4844f693aa3fdb0e56d16e : SUCCESS in 6m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c2500c68de4d414ebca8671af03e170c : SUCCESS in 35m 40s","accounts_in_message":[],"_revision_number":3},{"id":"7488f4e0e8925a33379aa5649d84f390c6e73140","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 19:43:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"79ea5dcb851bb7ae8cd29e16bd943f07644798e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 20:56:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"41b199cc7888530dfcb434395c59c575c9fe0ea7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 20:56:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3490313f6a1e49ed8f86249376b70335 : SUCCESS in 41m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6ebb4331fd3049a985187414a14c85ac : SUCCESS in 44m 31s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/a1da6ffc0996437f9633d561d509e603 : SUCCESS in 46m 38s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/26e64c1d26994a18b4626a8863f4ea05 : SUCCESS in 42m 57s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/546be3d3f3ac4d1f8a2411f75492638b : SUCCESS in 1h 06m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e23e64b5060c4dc99eb95a3e33c36090 : SUCCESS in 40m 16s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7665cf8ce18b45a7b867ac213fec4f68 : SUCCESS in 41m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e9511e1afb84244a3f6fcd6e5e15c70 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa70e094a081412ba4ab9fa8d181df69 : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c18581894eae477a9180d65ae4f89652 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/540e97f3f5c14dc5b2dacb5b08875068 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6129b26d758b454f812b07a34b4233af : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06dcd0a4bfef465396ea3799e099b5be : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99886192eeed415bbef200a6160775c2 : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":3},{"id":"876cd68878fc1eca762961004eeef567d5c20a3a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 20:57:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c51dfd369b984f2a882bd3b7cd881719 : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74d0a90fce98417793a088073b65fa89 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"282392158262e1afd1a2e9f7f9d6b9d2bda0fb1c","revisions":{"ceadf1516b7d1d663ee47ec5de499885f805e027":{"kind":"REWORK","_number":1,"created":"2019-10-22 08:22:20.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/08/690008/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/690008/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/690008/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/690008/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/690008/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/690008/1"}}},"commit":{"parents":[{"commit":"68d8d8f4e5b896ca4c7363972d0ba88a7c5d3d50","subject":"Remove wsgi eventlet monkey patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/68d8d8f4e5b896ca4c7363972d0ba88a7c5d3d50"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-22 08:19:54.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-22 08:22:05.000000000","tz":-420},"subject":"Fix api version not providing the full url","message":"Fix api version not providing the full url\n\nWe are currently using host_url to build the version\nhref which causes the /dns portion to get dropped. This\npatch changes host_url to url_root and enables\nenable_host_header by default.\n\nWe also deprecate api_base_uri as this is no\nlonger recommended.\n\nChange-Id: Ie8d815391b316d664b0c51099a08d8d32932c555\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ceadf1516b7d1d663ee47ec5de499885f805e027"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ceadf1516b7d1d663ee47ec5de499885f805e027"}]},"branch":"refs/heads/master"},"7a74feaed55b52b5149ede56d4ead60a03739369":{"kind":"REWORK","_number":2,"created":"2019-10-22 08:40:42.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/08/690008/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/690008/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/690008/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/690008/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/690008/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/690008/2"}}},"commit":{"parents":[{"commit":"68d8d8f4e5b896ca4c7363972d0ba88a7c5d3d50","subject":"Remove wsgi eventlet monkey patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/68d8d8f4e5b896ca4c7363972d0ba88a7c5d3d50"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-22 08:19:54.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-22 08:39:13.000000000","tz":-420},"subject":"Fix api version not providing the full url","message":"Fix api version not providing the full url\n\nWe are currently using host_url to build the version\nhref which causes the /dns portion to get dropped. This\npatch changes host_url to url_root.\n\nWe also deprecate api_base_uri and enable_host_header.\n\nFinally we add a help url that points at the\ndesignate api reference.\n\nChange-Id: Ie8d815391b316d664b0c51099a08d8d32932c555\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7a74feaed55b52b5149ede56d4ead60a03739369"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7a74feaed55b52b5149ede56d4ead60a03739369"}]},"branch":"refs/heads/master"},"282392158262e1afd1a2e9f7f9d6b9d2bda0fb1c":{"kind":"REWORK","_number":3,"created":"2019-10-22 08:44:26.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/08/690008/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/690008/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/690008/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/690008/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/690008/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/690008/3"}}},"commit":{"parents":[{"commit":"68d8d8f4e5b896ca4c7363972d0ba88a7c5d3d50","subject":"Remove wsgi eventlet monkey patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/68d8d8f4e5b896ca4c7363972d0ba88a7c5d3d50"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-22 08:19:54.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-10-22 08:43:54.000000000","tz":-420},"subject":"Fix api version not providing the full url","message":"Fix api version not providing the full url\n\nWe are currently using host_url to build the version\nhref which causes the /dns portion to get dropped. This\npatch changes host_url to url_root. We also enable\nenable_host_header by default.\n\nFinally we add a help url that points at the\ndesignate api reference.\n\nChange-Id: Ie8d815391b316d664b0c51099a08d8d32932c555\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/282392158262e1afd1a2e9f7f9d6b9d2bda0fb1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/282392158262e1afd1a2e9f7f9d6b9d2bda0fb1c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
