)]}'
{"id":"openstack%2Fdesignate~691222","triplet_id":"openstack%2Fdesignate~master~If68e1778d2fbffc0e9d660aa93175bf4b91b9dce","project":"openstack/designate","branch":"master","topic":"fix/invalid-rndc-command-when-hostname-is-used","hashtags":[],"change_id":"If68e1778d2fbffc0e9d660aa93175bf4b91b9dce","subject":"fix invalid rndc command when hostname is used","status":"ABANDONED","created":"2019-10-25 06:55:55.000000000","updated":"2021-12-24 04:29:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d55ef7b4906a04225936df24c12d47c73722c5ce","_number":691222,"virtual_id_number":691222,"owner":{"_account_id":30900,"name":"Gregoire Mahe","email":"gregoire@mahe.lt","username":"gregoiremahe"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-12-16 15:15:12.000000000","_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-25 09:18:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 09:18:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-16 15:15:12.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"}],"messages":[{"id":"b9a6f593e5dc434b018cead7336093bb10cf43da","author":{"_account_id":30900,"name":"Gregoire Mahe","email":"gregoire@mahe.lt","username":"gregoiremahe"},"date":"2019-10-25 06:55:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b8e6b20d84a984e2b986f45fbfd4fa4da4b8035","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 09:18:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/942bdacb974f453fbf2781fbd66ce122 : TIMED_OUT in 1h 14m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/32f27870fcf0457481e6d400dd019985 : SUCCESS in 43m 39s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/fc963f8bec9244aa8004337fa07d8680 : TIMED_OUT in 1h 16m 28s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/33324f1d70144683a1d3cc42f3e2934f : SUCCESS in 42m 00s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/ffba0577c3744d279a17bfec39f1473e : FAILURE in 31m 51s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/28316c7aeef64effafb10083547b8f2a : SUCCESS in 1h 08m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/391e1bed3b45484c93ae58fcf1ad5b8b : SUCCESS in 42m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/96201591d96a4b16ba361f6690286d6c : FAILURE in 1h 23m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05597cbfd1bf45d5bc9bdcb71eecf992 : FAILURE in 6m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7c81a87180e4273a2e3d4294f25ad4c : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98a8b6917e2d4dc3836bfe1b759bd772 : SUCCESS in 6m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9ea39ee7c9a844c4b0fbb0e574fca3a5 : FAILURE in 8m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f51e749737ea42f1ae302b0d364b53d2 : FAILURE in 6m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2558f00ed33743e8897726defdb20e69 : FAILURE in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a96d740056b749f59786a9b26e1492b7 : SUCCESS in 9m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/00369e38f02945c6bfdd4227b150fdda : FAILURE in 48m 21s","accounts_in_message":[],"_revision_number":1},{"id":"f44a7ac1fc43992d49ea61e9c586411aab913ba7","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2019-12-16 15:15:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ec83f0d4e7c945bd30aebef35142a59f7e45962f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 17:00:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/90b0d32f9195418d9dbb95e8fc4eb6fa : TIMED_OUT in 1h 15m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2fe0879ef7964507b928eb136c1a757f : SUCCESS in 38m 54s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/d3a2410af6824646a6ac2b2c085d7cae : FAILURE in 29m 23s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8ba2f4363d1d46ea87c4531f166e0c5f : SUCCESS in 1h 05m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/985502eac77b4eada2af3b876b6fe21b : SUCCESS in 41m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c5956899cf2f42a4bb2597f4e393b04a : FAILURE in 1h 38m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d996082f6982442fbb030a84783cd3be : FAILURE in 6m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af15a9a69fe74d0a9403a63eca1661ed : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24f4de69d2494f08b7e62d2db454286e : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f14d8cdb9b734560a3fc54bca2a83b18 : FAILURE in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/681cf37814d84bfb831681a9d5122f34 : FAILURE in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6bfd22e0b794807a1db17f59269a0cc : SUCCESS in 8m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ab44e23ef8d249a1b637e0580f975379 : FAILURE in 38m 59s","accounts_in_message":[],"_revision_number":1},{"id":"d55ef7b4906a04225936df24c12d47c73722c5ce","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-24 04:29:05.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"093fe6e815500c140ca1d0ce74efb52806180480","revisions":{"093fe6e815500c140ca1d0ce74efb52806180480":{"kind":"REWORK","_number":1,"created":"2019-10-25 06:55:55.000000000","uploader":{"_account_id":30900,"name":"Gregoire Mahe","email":"gregoire@mahe.lt","username":"gregoiremahe"},"ref":"refs/changes/22/691222/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/22/691222/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/22/691222/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/22/691222/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/22/691222/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/22/691222/1"}}},"commit":{"parents":[{"commit":"1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11","subject":"Merge \"Fix api version not providing the full url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11"}]}],"author":{"name":"Gregoire Mahe","email":"gregoire.mahe@corp.ovh.com","date":"2019-10-25 06:55:32.000000000","tz":120},"committer":{"name":"Gregoire Mahe","email":"gregoire.mahe@corp.ovh.com","date":"2019-10-25 06:55:32.000000000","tz":120},"subject":"fix invalid rndc command when hostname is used","message":"fix invalid rndc command when hostname is used\n\nChange-Id: If68e1778d2fbffc0e9d660aa93175bf4b91b9dce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/093fe6e815500c140ca1d0ce74efb52806180480"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/093fe6e815500c140ca1d0ce74efb52806180480"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
