)]}'
{"id":"openstack%2Fdesignate~691523","triplet_id":"openstack%2Fdesignate~master~Ib221f4cf0371e70fc6900582d826ffc1bdfc12b9","project":"openstack/designate","branch":"master","hashtags":[],"change_id":"Ib221f4cf0371e70fc6900582d826ffc1bdfc12b9","subject":"Implement create/delete zone for Akamai v2 API","status":"MERGED","created":"2019-10-27 14:41:38.000000000","updated":"2020-02-12 19:20:59.000000000","submitted":"2020-02-12 19:18:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691523-1581535095310-b9790512","meta_rev_id":"b2e7a7a05df97ee765ace2efbba21b99cb0cccf4","_number":691523,"virtual_id_number":691523,"owner":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-12 19:18:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":2,"date":"2020-02-12 17:13:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":1,"date":"2020-01-16 18:54:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-16 18:54:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":1,"date":"2020-02-12 17:13:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-16 18:54:30.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-28 06:23:11.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-01-06 07:06:37.000000000","updated_by":{"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},"reviewer":{"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},"state":"REVIEWER"},{"updated":"2020-02-12 17:13:00.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2020-02-12 19:18:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"06e2c11b26d3608de4eb344cff9b9c5c1484d08c","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2019-10-27 14:41:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1bebbfbb0be0acb52ae400609117cafbf1e8f41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 15:45:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1e0cd96498004d3f918b4fb312638523 : SUCCESS in 1m 49s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5ba161e5a25e4f8a9317f10b155f4285 : SUCCESS in 39m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2c9a15a4620242d18da61618f1f8a9bb : SUCCESS in 36m 13s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/482c8ede8a1a40d8bcc4a263dc752b94 : SUCCESS in 42m 01s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/ab906e1f98b047c59c6541300c61b47a : SUCCESS in 34m 20s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/6f223536c4244df58696b87f31876e85 : FAILURE in 27m 02s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/409b6b27f1d446e893ccaccc6da299d3 : SUCCESS in 57m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e68d1f569d734d68af315fd2c5ddd028 : SUCCESS in 34m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1c967770d05b4f2c9870e05f12a6a65f : SUCCESS in 37m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1ccac545fa0434f8a14c23e38fd2d3a : FAILURE in 2m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64e60472d8ec49b0b35e3c15d25c14d7 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ddf8593c39948b0b89ca75bb36c5a1f : SUCCESS in 3m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1bb7bd25dd9d4a9482f43aa087b97531 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55dcab7b32224484ae7aff3cc69f15d1 : FAILURE in 2m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2be187333b2a4289ac279e5d6c674179 : FAILURE in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4249480990074c81a42118a26b0b4bad : SUCCESS in 5m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cf14b7fac0bf4fca87d5ff8dbef8e596 : FAILURE in 1m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b3b7621473074708aa02dd29c1950b20 : SUCCESS in 32m 02s","accounts_in_message":[],"_revision_number":1},{"id":"7ebc619579b58bab9a7f1429bf8869bd27968694","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2019-10-27 18:52:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a749e8564078990b876458021e8baf657fac929d","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2019-10-27 18:52:55.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"8f9dc75a171abfe3e85d7343e48e66b59ceb918e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 19:46:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c822be173dec49278cf53839b3d99cd8 : SUCCESS in 2m 24s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/92e4a06b094d42e6adcf2eb68dfd5231 : SUCCESS in 40m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f196623e4bee4640bad83975feee26a0 : SUCCESS in 34m 56s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/d4854fb038304147b8fedd9a65aa4696 : SUCCESS in 42m 17s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/0a275765620d4751b9f3fd36761d80fb : SUCCESS in 36m 18s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/a8957996788645acba423a53399d19c1 : FAILURE in 28m 20s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/88a947fac28f4eb39e09010eca3f9da7 : SUCCESS in 51m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/75631c87f5c04d6b9c8b10f4703ecef1 : SUCCESS in 38m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a5f539777e614ff4a25ad1a56da1c415 : SUCCESS in 31m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cea472e5039b40629f6ca7e82827ae15 : FAILURE in 5m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b854ce5e0cf747c8b4c4cdef229533b7 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/734d7a8c9a254f66925275d1dc0574e9 : FAILURE in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a138c5f6545a4fddba79b19db05050fc : FAILURE in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43a3aea330f3497885742365ecc2960f : FAILURE in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1c9db24f1ee8493da0b598eb3883eb9b : FAILURE in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65449afe8ffb4c3c9377efc3756ba8bb : SUCCESS in 7m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cd80e32344c94eac97c62bf1e5a41188 : FAILURE in 2m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/87b583da6e41495cb5b319ad088d530b : SUCCESS in 30m 07s","accounts_in_message":[],"_revision_number":3},{"id":"b692ed0ffcee27ba3f9d009e98a5ccd27358d65d","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2019-10-27 19:59:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b4abb597df0299d8456715e2945f7219b9f92718","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 20:56:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dc12f38428244277a4311a3ce46b8765 : SUCCESS in 2m 16s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/db1aa04488934e77a6767ae7131bc9bb : SUCCESS in 36m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/368614040cfe4d5f8d54fbe80d595251 : SUCCESS in 37m 34s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/6f29b4b64f2f45d8aa75a530470d8a01 : SUCCESS in 37m 35s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/ad0d39d59aff4dff91042bc5b1f64f27 : SUCCESS in 35m 38s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/008f12e2909049fbb20de5abb3743b15 : FAILURE in 26m 03s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5d1fe8131bca49f5952561c99e69f9b3 : SUCCESS in 55m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6bf51dee12e5410589c9536b620d2cf5 : SUCCESS in 35m 51s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ebc0ab289c8b49a38f9fb21bc352639e : SUCCESS in 38m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/124b2560940e4611a28b4d5aefa56e64 : FAILURE in 4m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13eef19c69244b7089851202b08cef9f : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3340e6e9753941c99629aa965677b641 : FAILURE in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6e10dabdb8aa4c93af8f6a0268db3ce5 : FAILURE in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cde23173e3854aa8b0fa48d27f0e0c11 : FAILURE in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6613957d845e4dada899ae87b55fb3a1 : FAILURE in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00faa8b81c1e4e0d8856ae51b8ea243b : SUCCESS in 6m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/06838e49a45b458699bb0cf5cd6076e0 : FAILURE in 2m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6cfd0fb8a3e6489e8c36a32ef06abea9 : SUCCESS in 34m 29s","accounts_in_message":[],"_revision_number":4},{"id":"e0b088ce8b2785c15b250d91ede98a1b2ca209ba","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-28 02:41:20.000000000","message":"Patch Set 4:\n\n(7 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6e9e33d923352f50ee6e214f3c657b60b4a0ba3e","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2019-10-30 16:05:57.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4cfbc8883600524f67cd4a83acc99ea456e045ec","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2019-10-30 16:06:11.000000000","message":"Patch Set 4:\n\n(7 comments)","accounts_in_message":[],"_revision_number":4},{"id":"0a1a736fd618b8d3d555493d38c1f81a8e10323f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 17:39:35.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a305da35657a4f46aa7b73840d11898b : SUCCESS in 2m 07s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ad1ba9dbb274496d8c4fb3f9a5a0b0a6 : SUCCESS in 32m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f001eb72e44846d3baec90fd3944b2d4 : SUCCESS in 43m 45s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/ae94e78eea3b467f8e8f38e2e78f0a7c : SUCCESS in 39m 46s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/42cfae1745994466af605222c9fabb11 : SUCCESS in 42m 50s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/5a6fffa9b1f042d396bdf2221b2263c9 : FAILURE in 23m 35s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/40b38f694b08499ab9f172a994c787dc : SUCCESS in 1h 04m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d0dc49ee2ce540c18b07c9cf7aa382b0 : SUCCESS in 37m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d357f04f387348cf8fc6550b50232465 : SUCCESS in 38m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e606fb652d834ac38375c8b45b3496f0 : FAILURE in 4m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd671536084948f282b572d8b07b67d3 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca62c64975d9450da71fa96828adfcbf : SUCCESS in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fa8f6742d79841cda9f304f9d950718a : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0acc3f68c8e482b83c0644f682dea67 : FAILURE in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aecee8f965fd4c7fa2b42befaa988a8d : FAILURE in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83796df81cb04af2b4eeba8e4acc96ff : SUCCESS in 8m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ee710d5d51424c758ec31802f12c1d7f : FAILURE in 3m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/08d25248aac14708b2b04ae62af2daa9 : SUCCESS in 39m 54s","accounts_in_message":[],"_revision_number":5},{"id":"f0b2258c5c0044aa179169ed89da9cf4e40c79e0","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-30 17:39:42.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"389c1e585cc3471a0ac04d376429de63639eeefc","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2019-10-30 19:06:22.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"68c5d46a3312ca42956e1ac08950386a2bd50924","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 20:02:20.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f2343c6c920d487e8df2b21092e32f36 : SUCCESS in 2m 21s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/db3e0e7b17a748a5b1bf7e3bc538a00e : SUCCESS in 44m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ae7f8d302f944a2087c4660efe9dff0e : SUCCESS in 42m 30s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/63bcd3750d314d47b1982ffe18155541 : SUCCESS in 41m 55s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/b980305cf9b14d2d80352235cf8d38c2 : SUCCESS in 35m 14s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/fb43792cccca47ab9417481a02df725c : FAILURE in 22m 12s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/de254b06982e46fea2c92038472afe65 : SUCCESS in 54m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/22bfc8ae685940a79b56c8ece52d1af8 : SUCCESS in 34m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/777aeb26a3de477283b7ce2060b770ce : SUCCESS in 36m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e27bc7db3d784bfe8f50b24b63178070 : SUCCESS in 4m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19a2d5297dba460b8d14539bed120fd2 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cec8d38e3cb4171bab4054db58c66be : FAILURE in 3m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/63d3d44f0d6241b7b863d0b923abd06c : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be31f6164c7549b1b3f52e60b8b57944 : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3055af63083e4f16bf06f9d9f38a1985 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ab892cc2c124dc7b29b898828d701f2 : SUCCESS in 6m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a0a256a725af4a099931a48cef6705fa : FAILURE in 2m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7931ef86879145ac8ae4efe939eea661 : SUCCESS in 40m 31s","accounts_in_message":[],"_revision_number":6},{"id":"2d949548e6af8189a71fe6f5bc3bbe17fef016bd","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2019-10-30 20:08:19.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0992e41934a717e58c615bd48ed9c5e9d62b2d92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 21:16:40.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4a7bd55e1577409098f4fe09f2aade4f : SUCCESS in 2m 00s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9896314837de44a4ace520bc2f6aef1a : SUCCESS in 34m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c9c1378bb4b9494196ac4c06307d3e8a : SUCCESS in 49m 13s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/4838a4e2e55044bab3f057ee02cf4171 : SUCCESS in 42m 33s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/f23f861b6e28485785874d7a186f3841 : FAILURE in 39m 49s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/7342ccd8ad84412587e34ffb1df7e71c : FAILURE in 29m 20s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2c2acd13338141a1a2c3a787837d5029 : SUCCESS in 1h 05m 57s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8d051353e15e4c7a8ec9c544cd7b0d3a : SUCCESS in 42m 20s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/29f1707255ba47b3ab99104523b63d95 : SUCCESS in 47m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f8c2dbf780e45fb9c774f8988c91d5a : SUCCESS in 5m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c84510c6777a480a948b0b0b5b3cab18 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9031d8f19f584448b17e72659c848762 : SUCCESS in 3m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d531df31b1e040cca014b3dac57b6888 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5762f5ba62d4a2c8f631549ea30bedd : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f639ab99df2648e98abce79fb1f1defd : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6881dda95561403fbdb48f35b20dc915 : SUCCESS in 7m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a8ce383323f844a6ad539822e13c2403 : FAILURE in 2m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6e7b8909355a4c6a8983ea8a58c51692 : SUCCESS in 34m 25s","accounts_in_message":[],"_revision_number":7},{"id":"77ba9c5634a1f7ee7edd07b8d5a646ca5c3fbe68","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-10-31 04:06:57.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"56a34f0211f47e5c62887ef74ba54f7510003a96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 05:05:59.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7468e5a8a38344818a7257860deb71a4 : SUCCESS in 3m 34s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/32c58a8e6cea4c2b8b7a4159927b78d7 : SUCCESS in 39m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c8a966f89eae44859f49ac1642312b69 : SUCCESS in 34m 16s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/63f256f35869483f9c64553b0ffadfae : SUCCESS in 38m 37s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/d8aed2b06d994987846d6cf75e9e80a2 : SUCCESS in 34m 41s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/2a8a5b9c65c94233a2698bba791ba201 : FAILURE in 26m 31s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/99578f62162848658afb5b5e4ee014f2 : SUCCESS in 55m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/dcb33a26caf04ff6994af92c70a79213 : SUCCESS in 38m 54s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5e4c8b26e14f4385b7d97bb4342277f2 : SUCCESS in 38m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d665ab0de5db4b9f97179625de480ef6 : SUCCESS in 4m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0319ddc9ddc47159428bda430c62bae : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b25a8d110d0444cc88e86603dff98a89 : SUCCESS in 3m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/033358ff2a854d6c9ad9876ba824c77b : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61c71862bbc14398a8851197f8abd41e : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac55b803310f4be6b26f21fedc5ad067 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2504976265d64d16816b7f4f0d3c394b : SUCCESS in 9m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e5a531ed125b449d9ec7ae8f8d72e62b : FAILURE in 2m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cda654c48d33444bae9fc40dce6d8fc0 : SUCCESS in 34m 51s","accounts_in_message":[],"_revision_number":7},{"id":"9341b137c0b06246d505ef1e301c1f865e782153","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-01 22:50:03.000000000","message":"Patch Set 7:\n\nI guess we need it in global reqs if we want it here\n\u003e ERROR: Requirement \u0027{Requirement(package\u003d\u0027edgegrid-python\u0027, location\u003d\u0027\u0027, specifiers\u003d\u0027\u0027, markers\u003d\u0027\u0027, comment\u003d\u0027# Apache-2.0\u0027, extras\u003dfrozenset())}\u0027 not in openstack/requirements\n\nAn alternative is to add it to test-reqs and have users install the library only when using this specific driver.","accounts_in_message":[],"_revision_number":7},{"id":"354b767c797afcad625cf0a28042fa53a7ad38d2","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2019-11-03 13:51:00.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"918fa33813be652f003b6e430ab31cae513c4082","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-03 14:48:56.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2553220b5f284c14a47d5b8cde89f480 : SUCCESS in 1m 34s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a7271f471675444ea9a5bb4121f397fb : SUCCESS in 39m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/98aa030d770a4bea9d595b1088de677f : SUCCESS in 31m 22s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/e043f21d7600465b9fb0af679cf3839c : SUCCESS in 39m 48s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/937e1d52b9374e88bc88b8b33ab689d4 : SUCCESS in 38m 57s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/8c3c193bb80a4c2e897f41393ca026e8 : FAILURE in 24m 23s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/52ade20c4df543c6bc10b1baaaa4c214 : SUCCESS in 57m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d53a84e315aa4e4c85e7585b2d506d0c : SUCCESS in 44m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d18b5431aa0140719570e84ff028fd48 : SUCCESS in 31m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1c476000be14b7a85377e48f1ca0447 : SUCCESS in 5m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5242e65bea1044879d460ab17ffcf82b : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3749998cf51f454db7395d6b27a38ff6 : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/52013b3f0f604e379036800643ff4871 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea6e384fb1e3463b824d5d6931612ef2 : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b69c342f9e53472dbadf849e36899786 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df12cfb34ce543d68157258c2cf8f370 : SUCCESS in 6m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d4082f4ea4a74e398ee37e6a7705e658 : FAILURE in 2m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0427c9f19c38444b92a166f592676b6f : SUCCESS in 29m 30s","accounts_in_message":[],"_revision_number":8},{"id":"a51cada19fd413c15674edb62e273abc02682ad8","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2019-11-03 15:51:54.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d028d1a88a6ccf68f472f9fcd334433756a4fbf4","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2019-11-03 16:15:39.000000000","message":"Patch Set 9: Workflow-1\n\n@Erik, I will try to add this dependency to global-requirements and give you know about result. For now mark PR as In_progress.\n\nAlso could you please give advice about obtaining TsigKey during zone creation. I could not fund any clear way to get it in Backend implementation, because context has not any tsigkey_id and more over it has nt method to call DB directly as it\u0027s done in TsigInfoMiddleware. Is it mean, that there is only one way for it is to use pools.yaml like in akamai v1 ?","accounts_in_message":[],"_revision_number":9},{"id":"71de2770db4295fa72913b647f92f5f86111b87e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-03 16:56:49.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a8c66a94290c413cbd43a395aecffc91 : SUCCESS in 1m 45s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e775515cc1504aaea95cd783e845974c : SUCCESS in 36m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a33a0226dfb0471ba194c378ea9ecd3c : SUCCESS in 41m 38s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/608d7baea6a74ae4bf1773205357bb43 : SUCCESS in 41m 56s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/dbcc24576a624807bfb1477214dd835d : SUCCESS in 37m 58s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/dabdfbcd64a84100a06fd284b2463f8a : FAILURE in 28m 10s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b97a33643f2a4f2182e5d5f50ab1271a : SUCCESS in 58m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e73ede5e7f6b4f0faf3e02fe055c10b6 : SUCCESS in 39m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e23fd6fc5cd646f99b9c26678233938f : SUCCESS in 38m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1cb742e3f1784f4d9d133f2cc9996b6a : SUCCESS in 4m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a09e97496f1d467791df4350cce17f70 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c7a1e792ebf4f42a57c2cfa6f7d09d5 : SUCCESS in 3m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1789299178ba44ceb11d1ba949330aab : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d305ae9ad91d488d8e6837a857e0ba6e : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c746108f16a8448490c22a737208a8d8 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59b06b76f62f42f68aa8a08e847b1c73 : SUCCESS in 6m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0d5dd8e5d81549e995759704600a5d95 : FAILURE in 2m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dbe02060cb28422bbf6951a9e00ac2e4 : SUCCESS in 33m 47s","accounts_in_message":[],"_revision_number":9},{"id":"eb6508d46f22eb28c8e4943aceceedc56e6ad5d6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-03 20:18:58.000000000","message":"Patch Set 9:\n\n(1 comment)\n\n\u003e @Erik, I will try to add this dependency to global-requirements and\n \u003e give you know about result. For now mark PR as In_progress.\n \u003e \n \u003e Also could you please give advice about obtaining TsigKey during\n \u003e zone creation. I could not fund any clear way to get it in Backend\n \u003e implementation, because context has not any tsigkey_id and more\n \u003e over it has nt method to call DB directly as it\u0027s done in\n \u003e TsigInfoMiddleware. Is it mean, that there is only one way for it\n \u003e is to use pools.yaml like in akamai v1 ?\n\nI think so yea. Is there a use-case you are thinking of? Maybe something I can help implement if it makes sense.","accounts_in_message":[],"_revision_number":9},{"id":"d8d4abfd49df1ba12bbf755780c42dd3e7005999","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-03 20:19:42.000000000","message":"Patch Set 9:\n\nbtw would be nice to add a devstack script for this as well.\nhttps://github.com/openstack/designate/blob/master/devstack/designate_plugins/backend-pdns4","accounts_in_message":[],"_revision_number":9},{"id":"82a9211928475952da51a22eca2bd197fc1741c7","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2019-11-04 14:03:26.000000000","message":"Patch Set 9:\n\n\u003e btw would be nice to add a devstack script for this as well.\nhttps://github.com/openstack/designate/blob/master/devstack/designate_plugins/backend-pdns4\n\nI am not sure, that it\u0027s possible to install akamai locally with devstack. The plugin is used for work with API of external service.\n\n\u003e I think so yea. Is there a use-case you are thinking of? Maybe something I can help implement if it makes sense.\n\nI thought about some option with implicit using TsigKey for creation zone. Like if user have an TsigKey configured for current Pool, we take first available TsigKey from DB and specify it for the current zone (except the case, when TSIG explicitly specified in pools.yaml)","accounts_in_message":[],"_revision_number":9},{"id":"44a6dfce3d495d081553e0d3fdefb6f3fd4c5e98","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2019-11-04 14:12:23.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"0947e1bb184ef4257e653475abd3c22d01483347","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 15:21:05.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/28fd3b98af884d3f82c54a6e2d7ca74c : SUCCESS in 2m 26s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dd58748608d646beb2dc93202d1f592f : SUCCESS in 36m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e760981625134f19a4ee1a89561e56b6 : SUCCESS in 40m 18s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/f254e02ae5d945acb1c062a02470e4b7 : SUCCESS in 43m 18s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/d5e913a458d64239a314c90e4a51c43b : SUCCESS in 35m 07s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/5a17bd3e6aec49a3b7c8740b442f4fda : FAILURE in 34m 40s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5153678c7cb54197b8acc4aa009eb9e4 : SUCCESS in 1h 00m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0fcbafb8e7274ac1acefdfac7892e8d5 : SUCCESS in 39m 51s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/db0ca94684bc4b6ea9e25c516aa296da : SUCCESS in 39m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db8cc78602694313af2f4ee47074f2b5 : SUCCESS in 6m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e220ea54be054c8f8e57f264c62eab92 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9eb346b5592c45ab93da7955e897d2ac : SUCCESS in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aee25a31ef8a4b19a2450c858ada0b29 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0976f7c70ba64b51ba3d279087d497f6 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b332cb4068d45ddb0d0f9bd46390cee : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d783d747a4149a0b294e960060dfbf5 : SUCCESS in 8m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9c1f055d77cc4901bf0da6e2dc121b38 : FAILURE in 2m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2d73e6bbb59645d28cd53bf339162728 : SUCCESS in 36m 19s","accounts_in_message":[],"_revision_number":10},{"id":"c8e77eab2baa645763427a4150b9d1c6e95596a9","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-05 03:24:25.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"160d79cce368ab30cdb4caf86f689ea6ec3c13ad","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2019-11-17 11:09:06.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"878adc3e352ce2f65c16c498431339af2cf1efac","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2019-11-17 11:09:26.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"b7c485550221c3b8d7e782bdbd05b02e31bd2116","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2019-11-17 11:19:44.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"365125ce0542c825bcb1e7d4ad1a7bd2b551e603","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-17 12:13:25.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/962036a86520494d8bd2b6c63bae0565 : SUCCESS in 2m 53s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9874f760c91a4508afa4e3a4ae02afc2 : SUCCESS in 40m 29s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/34f7f02f41b04ce980a68646ce9c6f6e : SUCCESS in 39m 44s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/c575e43cea1a447fb9a48d5c6fa414b9 : SUCCESS in 37m 00s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/2264cc50137f428998635c21717a2434 : SUCCESS in 39m 08s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/da8a77443c6849b697b2ffa13aa94fdb : FAILURE in 26m 00s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/32b648a2fd164593b6449e5d99be34ec : FAILURE in 50m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7d9d9184c8fe4c48bece048dcb4f9bd9 : SUCCESS in 34m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/47934da6348b404cab043c7085404220 : SUCCESS in 39m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5bb0343c84d84304b9bd22c95b841c15 : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c40a85feeec340d4ab5c948ec8368f92 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/484cc415fa8c41dfbe7449a97832ea88 : FAILURE in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/98a8a4f81d74454f90055b0615883712 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7110fb37169540a7901e7fbd5f479613 : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/560f54efeebe4c49bb66d7509de186d7 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c0fd353cd574bf8a53d702595921641 : SUCCESS in 6m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ae63d084ee51414ab3165bee5c3f4525 : FAILURE in 2m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/896ac673b4404726a1c53a1bfcc4a84d : SUCCESS in 31m 13s","accounts_in_message":[],"_revision_number":12},{"id":"54d00058fb5d8a2cee749844c1903b3c33676932","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2019-11-17 17:14:09.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"ee8525ff427cc90e457aaffd2d745fa727f6221a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-17 18:14:30.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/207ba19cbeb84d0dbe192eaa64301a9a : SUCCESS in 2m 02s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/383661097acf4010bbce5e59c00f9718 : SUCCESS in 49m 45s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3f98b868caa94e50add8b9ca4acba394 : SUCCESS in 31m 40s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/eaa756f4d29a472db12b5869125fb8a2 : SUCCESS in 51m 16s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/7e2cda05bfe54ec19c85bc2390ecf60f : SUCCESS in 33m 25s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/5faa0e3aab0f4da79f9be4a8cf5b0929 : FAILURE in 32m 23s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1e33069dc50c44aa97cdac20f443590f : FAILURE in 57m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8766a19dc9b549d483d7ebce7b2fe51b : SUCCESS in 44m 33s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cc285f3fbb2f47849d51981588451cb1 : SUCCESS in 44m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dfe87baa6a0f4c48872e62fe4750dda7 : SUCCESS in 4m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0178b8294d364102be11ef470b775b03 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e87611806db641e9a25ab3b0351d35e4 : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/71fe736320214434a2ff981ead06617b : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3cc1b69f00764e568b9bade27987bbc7 : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fbe26fb203004d52ad5575fb6d455a73 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af0d1614a47c4ff6bd7be24d7929ad8b : SUCCESS in 6m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2fb9bc208ae24c4ba4c6ad275251b8a3 : SUCCESS in 4m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5763cca86ddc437fadd8d774545dcf06 : SUCCESS in 40m 58s","accounts_in_message":[],"_revision_number":13},{"id":"1a86b14cd73dca88ec5c9c93964b82f6b889092a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-17 21:08:34.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"ecb768cdbcb76b6dd2b3cc515e39a408952fc0a6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-18 18:04:41.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"308142c2aed451cc7975cc821e4d0b269545f7d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 19:07:06.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7047a7b48b4b467c9d46ca2ed6c4fb6d : SUCCESS in 2m 52s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5a3c1612898a4b9d8e15069a92acada5 : SUCCESS in 39m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e71fe084b71c4d029cd86c16f6ec1025 : SUCCESS in 44m 45s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/cc367a2de57a4eab89d695294bdca5b0 : SUCCESS in 46m 06s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/fc2bae80385a46f2ae3be57142465faa : SUCCESS in 40m 12s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/deed849de944443181c49036b19fb4d1 : FAILURE in 28m 19s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7f34f27215cc494aa558b50c8b3f24a0 : SUCCESS in 56m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/dd31450940974a408d00fed635dd9018 : SUCCESS in 38m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6d5946846a6348b295077bc56dc69dd4 : SUCCESS in 40m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a1f3a27ac144c349af6189041f5839e : SUCCESS in 5m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2cae94257e5b4904b5d2a4651a078b57 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/530c6d4239e449e4aff6356abe35638f : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f0bbb5b8d9fe41aa8817ab2f98e6e180 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29f126b9f1f941718d2233386d8f5f40 : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da274e2c43f84484a571bbdf1efd8279 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dd756e15a14479e83b4f189c37f05af : SUCCESS in 7m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/211d1c5101174c8da28bfa1760ce8817 : SUCCESS in 2m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/601b3cc021c844c5b4ee97cecf6e5915 : SUCCESS in 40m 40s","accounts_in_message":[],"_revision_number":13},{"id":"f2d4c19715f1287c89593def1a073d224db5d0a7","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-28 01:39:11.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"79095eb59dbbd83023bc585d351f28a1a82cbcc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 02:39:46.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d646a66b680a43b798ab76eec61d4bf1 : SUCCESS in 2m 04s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8a15df068d0142c48f4d541ee33a4227 : SUCCESS in 38m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/41715f1bb08247bf917a9932be6404f7 : SUCCESS in 40m 14s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/4db6710b4ed7443eb7568f2871d017e8 : SUCCESS in 40m 22s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/d223221b123942f3a6642baa2d9da5d8 : SUCCESS in 33m 57s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/3fc4bb8a8efb4de9bb36a7b4a0b958fb : FAILURE in 29m 24s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/19aa0e8606954b6c8d0f18291e13aa2b : SUCCESS in 56m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/747cfe62e58f49c5bc340befed3019bb : SUCCESS in 38m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/09d210e3e4ee4846bd7b4eb16596b7e4 : SUCCESS in 39m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01b61c424b3d4ec39cefb2542112bdad : SUCCESS in 4m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6c506f21e374abcbc839977dbc92bcf : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d596eadc13b4f83a70cc5dba20ffc6c : SUCCESS in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e9b8589255234f2585bc3f23638cf386 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f56d576e325a4973ac2ab91d1bbb5745 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff0bc2870d0e47b7b3ec4441d8fa886a : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad3804e7a915411e83b5b67f707365df : SUCCESS in 6m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7dcda72e039342a7874ccc8dd09ef6fa : SUCCESS in 2m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4ba4845bc39d4a3a9f378dc421aeb102 : SUCCESS in 35m 29s","accounts_in_message":[],"_revision_number":14},{"id":"c5165fa09871acbf9b4d2145f96206142c2d6c31","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2019-12-01 10:46:37.000000000","message":"Patch Set 14:\n\nPatch with new dependency was merged, so this patch could be reviewed and merged too.","accounts_in_message":[],"_revision_number":14},{"id":"bb077526df7cbd100270fe68215ed7631616e1bb","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-12-28 06:22:59.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"c8a4005bf22dfcf5fcfdcd403f262429892c546e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-12-28 06:23:11.000000000","message":"Patch Set 15: Review-Priority+1","accounts_in_message":[],"_revision_number":15},{"id":"fb1413dc7926d87991cd4a742e87af7bbc09e7ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-28 07:43:13.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e7a9085a7cba40dd90a7fdcb4863a13b : SUCCESS in 2m 08s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fd3e0e16a2104ed492754e9482c594f8 : SUCCESS in 45m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/497268c2f6484f74a31b842dd73c7049 : SUCCESS in 1h 00m 26s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/4859f10699114c7c86b1bf86d302bb82 : FAILURE in 50m 08s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/91ce8b5e28664dd88a75b0d017147d92 : SUCCESS in 1h 14m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5bbc636a36574e1cb55858bbf42d0761 : SUCCESS in 57m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4073662b2ab44a87b1b15a60186dc89c : SUCCESS in 59m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c622ed3c0904434fa8925e056de7b725 : SUCCESS in 5m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f34b9fcbea3f421e9824bd4ad16261b6 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6076401fe3cd4b6784fdb6c80d178370 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ef5943dc64a42a593392ee16bedd59d : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3afa798fe8b74d1ea7eb170d8e1b27ea : SUCCESS in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42aee1af4db0409fa78e98406fb326ae : SUCCESS in 8m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e91b7a10210f4e5e946b5b59385aeacd : SUCCESS in 2m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e994d9c128a14053a575589f30b2165c : SUCCESS in 58m 06s","accounts_in_message":[],"_revision_number":15},{"id":"377ced2939a12aa8cc8c253cd6fa1b54f9efa817","author":{"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},"date":"2020-01-06 07:06:37.000000000","message":"Patch Set 15: Code-Review+1\n\nLooks good.","accounts_in_message":[],"_revision_number":15},{"id":"56e7308505305d83f8d10796a355c91e1927e151","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-01-16 18:54:30.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"65c45f91565e8535200494a9372fe3c3c936e27f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 20:02:17.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6df753b395274b868e32fd84eb0d3761 : SUCCESS in 2m 45s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1ff6c18ee2e44771bc883fcdbe57921f : SUCCESS in 41m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/99d10dbab8d243f8a45fa8d19447a21a : SUCCESS in 39m 42s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/6831b97cfd2d405ba66796b62674d869 : FAILURE in 30m 12s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f034e447b00e421aa6062e9e382d6c9a : SUCCESS in 59m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9327512ef29b4002a43c41cfa24f5d8a : SUCCESS in 44m 29s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e709b23d0c1044af9126d7116ee2b2c4 : SUCCESS in 39m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b3da94c191448e9a8068dcd8693512e : SUCCESS in 6m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f16da3af9da43068219728d179d9c37 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/061a2bbc2fca4a21963d0bb417377f36 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/400c7b8070714d1e9555c6d10e12cbf1 : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/76bd58aefa6f4713a5647db26276aa1c : SUCCESS in 8m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e38b245dd2e949bcb0dd1718fc31a4a7 : SUCCESS in 9m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/849bb194908442eb9578b30aea6707d5 : SUCCESS in 2m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1dfaa707959a493fb1c647b8d25968c7 : SUCCESS in 34m 44s","accounts_in_message":[],"_revision_number":16},{"id":"18cb4b5759b9ae8ae01f179704f3d524a4fe2041","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2020-02-12 17:13:00.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1\n\nas a follow up we need to add the backend here: https://opendev.org/openstack/designate/src/branch/master/doc/source/admin/support-matrix.ini","accounts_in_message":[],"_revision_number":16},{"id":"7affc81c85752f3bae6b4464c5f4ed97468cfc1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 17:13:13.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"1a527f1c235268ae1099583239e166e266b64623","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 19:18:15.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3b6c2a55b8f6453eb185387c1fdb927b : SUCCESS in 5m 41s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2c9de46882bc4c1ba624103608d66622 : SUCCESS in 37m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/09f6f78481da4d529fb4f3ff477e786b : SUCCESS in 44m 53s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/940efe143c50425392be10d5765f84af : SUCCESS in 1h 09m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1719ef0f63fd466aa6d306ba463abd5c : SUCCESS in 36m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3f5274a257bb4fd391c38348ff2441ab : SUCCESS in 48m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c52b23f514654e8c9cf87b8662323054 : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86bc6912a71b4b52b5ef79c7c031c5a1 : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0a28823a27145828b0f8fa9dd2971c5 : SUCCESS in 5m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46a5b008f3404b1eacce63a400659ab5 : SUCCESS in 14m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77e29941cecc4d458422a4ab6416a13c : SUCCESS in 11m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ab532e0ffa8243edb02555438aa63227 : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":16},{"id":"c78833a8ec0c29a0ca26666b01e401a583270ca8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 19:18:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":16},{"id":"b2e7a7a05df97ee765ace2efbba21b99cb0cccf4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 19:20:59.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/474ef583c6ac4339b848c4063b2d1290 : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"318b8d03193d3d9c26842528e9eaf25814df3ba9","revisions":{"c6c4f03631e7d699a1a7f1120221f199b3a42301":{"kind":"REWORK","_number":1,"created":"2019-10-27 14:41:38.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/23/691523/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/691523/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/691523/1"}}},"commit":{"parents":[{"commit":"1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11","subject":"Merge \"Fix api version not providing the full url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-10-03 19:38:10.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-10-27 14:41:24.000000000","tz":240},"subject":"Implement create/delete zone for Akamai v2 API","message":"Implement create/delete zone for Akamai v2 API\n\n- Process PRIMARY and SECONDARY zones separately\n- Ignore duplicate Zone error\n- Handle error when contractId or gid is missed\n- Ignore port for masters servers, because Akamai uses only 53 port and\ndoes not allow to specify any port in list of masters servers.\n- Add timeout and retries for soft Zone Delete\n- Added handling errors on the delete zone action\n- Add Log info message with RequestId on soft zone delete\n\nChange-Id: Ib221f4cf0371e70fc6900582d826ffc1bdfc12b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c6c4f03631e7d699a1a7f1120221f199b3a42301"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c6c4f03631e7d699a1a7f1120221f199b3a42301"}]},"branch":"refs/heads/master"},"deea5816e2f4163fedead82e466400be0f0bfedb":{"kind":"REWORK","_number":2,"created":"2019-10-27 18:52:43.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/23/691523/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/691523/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/691523/2"}}},"commit":{"parents":[{"commit":"1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11","subject":"Merge \"Fix api version not providing the full url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-10-03 19:38:10.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-10-27 18:51:57.000000000","tz":240},"subject":"Implement create/delete zone for Akamai v2 API","message":"Implement create/delete zone for Akamai v2 API\n\n- Process PRIMARY and SECONDARY zones separately\n- Ignore duplicate Zone error\n- Handle error when contractId or gid is missed\n- Ignore port for masters servers, because Akamai uses only 53 port and\ndoes not allow to specify any port in list of masters servers.\n- Add timeout and retries for soft Zone Delete\n- Added handling errors on the delete zone action\n- Add Log info message with RequestId on soft zone delete\n\nChange-Id: Ib221f4cf0371e70fc6900582d826ffc1bdfc12b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/deea5816e2f4163fedead82e466400be0f0bfedb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/deea5816e2f4163fedead82e466400be0f0bfedb"}]},"branch":"refs/heads/master"},"d306285b5bfb61344dc5fb698c18041221bfdce0":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-10-27 18:52:55.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/23/691523/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/691523/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/691523/3"}}},"commit":{"parents":[{"commit":"1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11","subject":"Merge \"Fix api version not providing the full url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-10-03 19:38:10.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-10-27 18:52:45.000000000","tz":240},"subject":"Implement create/delete zone for Akamai v2 API","message":"Implement create/delete zone for Akamai v2 API\n\n- Ignore duplicate Zone error\n- Handle error when contractId or gid is missed\n- Ignore port for masters servers, because Akamai uses only 53 port and\ndoes not allow to specify any port in list of masters servers.\n- Add timeout and retries for soft Zone Delete\n- Added handling errors on the delete zone action\n- Add Log info message with RequestId on soft zone delete\n\nChange-Id: Ib221f4cf0371e70fc6900582d826ffc1bdfc12b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d306285b5bfb61344dc5fb698c18041221bfdce0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d306285b5bfb61344dc5fb698c18041221bfdce0"}]},"branch":"refs/heads/master"},"adf0a9e44950016d180683e01436bfb7844993d2":{"kind":"REWORK","_number":4,"created":"2019-10-27 19:59:49.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/23/691523/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/691523/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/691523/4"}}},"commit":{"parents":[{"commit":"1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11","subject":"Merge \"Fix api version not providing the full url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-10-03 19:38:10.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-10-27 19:59:35.000000000","tz":240},"subject":"Implement create/delete zone for Akamai v2 API","message":"Implement create/delete zone for Akamai v2 API\n\n- Ignore duplicate Zone error\n- Handle error when contractId or gid is missed\n- Ignore port for masters servers, because Akamai uses only 53 port and\ndoes not allow to specify any port in list of masters servers.\n- Add timeout and retries for soft Zone Delete\n- Added handling errors on the delete zone action\n- Add Log info message with RequestId on soft zone delete\n\nChange-Id: Ib221f4cf0371e70fc6900582d826ffc1bdfc12b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/adf0a9e44950016d180683e01436bfb7844993d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/adf0a9e44950016d180683e01436bfb7844993d2"}]},"branch":"refs/heads/master"},"1e0ec40ae0a1d27d0868e365d66808c8f27db783":{"kind":"REWORK","_number":5,"created":"2019-10-30 16:05:57.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/23/691523/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/691523/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/691523/5"}}},"commit":{"parents":[{"commit":"1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11","subject":"Merge \"Fix api version not providing the full url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-10-03 19:38:10.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-10-30 16:04:21.000000000","tz":240},"subject":"Implement create/delete zone for Akamai v2 API","message":"Implement create/delete zone for Akamai v2 API\n\n- Ignore duplicate Zone error\n- Handle error when contractId or gid is missed\n- Ignore port for masters servers, because Akamai uses only 53 port and\ndoes not allow to specify any port in list of masters servers.\n- Add timeout and retries for soft Zone Delete\n- Added handling errors on the delete zone action\n- Add Log info message with RequestId on soft zone delete\n\nChange-Id: Ib221f4cf0371e70fc6900582d826ffc1bdfc12b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1e0ec40ae0a1d27d0868e365d66808c8f27db783"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1e0ec40ae0a1d27d0868e365d66808c8f27db783"}]},"branch":"refs/heads/master"},"afd5e41110610b091d528212e454807a0adce93c":{"kind":"REWORK","_number":6,"created":"2019-10-30 19:06:22.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/23/691523/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/691523/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/691523/6"}}},"commit":{"parents":[{"commit":"1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11","subject":"Merge \"Fix api version not providing the full url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-10-03 19:38:10.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-10-30 19:06:09.000000000","tz":240},"subject":"Implement create/delete zone for Akamai v2 API","message":"Implement create/delete zone for Akamai v2 API\n\n- Ignore duplicate Zone error\n- Handle error when contractId or gid is missed\n- Ignore port for masters servers, because Akamai uses only 53 port and\ndoes not allow to specify any port in list of masters servers.\n- Add timeout and retries for soft Zone Delete\n- Added handling errors on the delete zone action\n- Add Log info message with RequestId on soft zone delete\n\nChange-Id: Ib221f4cf0371e70fc6900582d826ffc1bdfc12b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/afd5e41110610b091d528212e454807a0adce93c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/afd5e41110610b091d528212e454807a0adce93c"}]},"branch":"refs/heads/master"},"0d510dbb20e531625bfb363cbea1b09237dc53e0":{"kind":"REWORK","_number":7,"created":"2019-10-30 20:08:19.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/23/691523/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/691523/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/691523/7"}}},"commit":{"parents":[{"commit":"1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11","subject":"Merge \"Fix api version not providing the full url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-10-03 19:38:10.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-10-30 20:07:43.000000000","tz":240},"subject":"Implement create/delete zone for Akamai v2 API","message":"Implement create/delete zone for Akamai v2 API\n\n- Ignore duplicate Zone error\n- Handle error when contractId or gid is missed\n- Ignore port for masters servers, because Akamai uses only 53 port and\ndoes not allow to specify any port in list of masters servers.\n- Add timeout and retries for soft Zone Delete\n- Added handling errors on the delete zone action\n- Add Log info message with RequestId on soft zone delete\n\nChange-Id: Ib221f4cf0371e70fc6900582d826ffc1bdfc12b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0d510dbb20e531625bfb363cbea1b09237dc53e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0d510dbb20e531625bfb363cbea1b09237dc53e0"}]},"branch":"refs/heads/master"},"04a7addeb4e50dde0511ed71735c2d23da5b738e":{"kind":"REWORK","_number":8,"created":"2019-11-03 13:51:00.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/23/691523/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/691523/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/691523/8"}}},"commit":{"parents":[{"commit":"1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11","subject":"Merge \"Fix api version not providing the full url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-10-03 19:38:10.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-11-03 13:50:39.000000000","tz":240},"subject":"Implement create/delete zone for Akamai v2 API","message":"Implement create/delete zone for Akamai v2 API\n\n- Ignore duplicate Zone error\n- Handle error when contractId or gid is missed\n- Ignore port for masters servers, because Akamai uses only 53 port and\ndoes not allow to specify any port in list of masters servers.\n- Add timeout and retries for soft Zone Delete\n- Added handling errors on the delete zone action\n- Add Log info message with RequestId on soft zone delete\n\nChange-Id: Ib221f4cf0371e70fc6900582d826ffc1bdfc12b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/04a7addeb4e50dde0511ed71735c2d23da5b738e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/04a7addeb4e50dde0511ed71735c2d23da5b738e"}]},"branch":"refs/heads/master"},"884eefb1bf8de07d0e371949522ad948d0e81dbc":{"kind":"REWORK","_number":9,"created":"2019-11-03 15:51:54.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/23/691523/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/691523/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/691523/9"}}},"commit":{"parents":[{"commit":"1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11","subject":"Merge \"Fix api version not providing the full url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-10-03 19:38:10.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-11-03 15:51:14.000000000","tz":240},"subject":"Implement create/delete zone for Akamai v2 API","message":"Implement create/delete zone for Akamai v2 API\n\n- Ignore duplicate Zone error\n- Handle error when contractId or gid is missed\n- Ignore port for masters servers, because Akamai uses only 53 port and\ndoes not allow to specify any port in list of masters servers.\n- Add timeout and retries for soft Zone Delete\n- Added handling errors on the delete zone action\n- Add Log info message with RequestId on soft zone delete\n- Add processing for TsigKey during creation zone\n\nChange-Id: Ib221f4cf0371e70fc6900582d826ffc1bdfc12b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/884eefb1bf8de07d0e371949522ad948d0e81dbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/884eefb1bf8de07d0e371949522ad948d0e81dbc"}]},"branch":"refs/heads/master"},"fb0a4549b67392fd5f4931e270a16ee4286bbd2f":{"kind":"REWORK","_number":10,"created":"2019-11-04 14:12:23.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/23/691523/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/691523/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/691523/10"}}},"commit":{"parents":[{"commit":"1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11","subject":"Merge \"Fix api version not providing the full url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-10-03 19:38:10.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-11-04 14:12:01.000000000","tz":240},"subject":"Implement create/delete zone for Akamai v2 API","message":"Implement create/delete zone for Akamai v2 API\n\n- Ignore duplicate Zone error\n- Handle error when contractId or gid is missed\n- Ignore port for masters servers, because Akamai uses only 53 port and\ndoes not allow to specify any port in list of masters servers.\n- Add timeout and retries for soft Zone Delete\n- Added handling errors on the delete zone action\n- Add Log info message with RequestId on soft zone delete\n- Add processing for TsigKey during creation zone\n\nChange-Id: Ib221f4cf0371e70fc6900582d826ffc1bdfc12b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fb0a4549b67392fd5f4931e270a16ee4286bbd2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fb0a4549b67392fd5f4931e270a16ee4286bbd2f"}]},"branch":"refs/heads/master"},"b02f88f71e1b5ad6142af6ae37a1a5197e581faa":{"kind":"REWORK","_number":11,"created":"2019-11-17 11:09:06.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/23/691523/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/691523/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/691523/11"}}},"commit":{"parents":[{"commit":"1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11","subject":"Merge \"Fix api version not providing the full url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-10-03 19:38:10.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-11-17 11:08:08.000000000","tz":240},"subject":"Implement create/delete zone for Akamai v2 API","message":"Implement create/delete zone for Akamai v2 API\n\n- Ignore duplicate Zone error\n- Handle error when contractId or gid is missed\n- Ignore port for masters servers, because Akamai uses only 53 port and\ndoes not allow to specify any port in list of masters servers.\n- Added timeout and retries for soft Zone Delete\n- Added handling errors on the delete zone action\n- Added Log info message with RequestId on soft zone delete\n- Added processing for TsigKey during creation zone\n- Added devsatck_plugin for akamai_v2 backend\n\nDepends-On: https://review.opendev.org/#/c/692819/4\n\nChange-Id: Ib221f4cf0371e70fc6900582d826ffc1bdfc12b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b02f88f71e1b5ad6142af6ae37a1a5197e581faa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b02f88f71e1b5ad6142af6ae37a1a5197e581faa"}]},"branch":"refs/heads/master"},"35aaf8729d98dfa9a0ffa4fb8e66d163850b64e5":{"kind":"REWORK","_number":12,"created":"2019-11-17 11:19:44.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/23/691523/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/691523/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/691523/12"}}},"commit":{"parents":[{"commit":"1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11","subject":"Merge \"Fix api version not providing the full url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-10-03 19:38:10.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-11-17 11:19:34.000000000","tz":240},"subject":"Implement create/delete zone for Akamai v2 API","message":"Implement create/delete zone for Akamai v2 API\n\n- Ignore duplicate Zone error\n- Handle error when contractId or gid is missed\n- Ignore port for masters servers, because Akamai uses only 53 port and\ndoes not allow to specify any port in list of masters servers.\n- Added timeout and retries for soft Zone Delete\n- Added handling errors on the delete zone action\n- Added Log info message with RequestId on soft zone delete\n- Added processing for TsigKey during creation zone\n- Added devsatck_plugin for akamai_v2 backend\n\nDepends-On: https://review.opendev.org/#/c/692819/4\n\nChange-Id: Ib221f4cf0371e70fc6900582d826ffc1bdfc12b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/35aaf8729d98dfa9a0ffa4fb8e66d163850b64e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/35aaf8729d98dfa9a0ffa4fb8e66d163850b64e5"}]},"branch":"refs/heads/master"},"b200a8830e470407d8818816f03bdb31ed313616":{"kind":"REWORK","_number":13,"created":"2019-11-17 17:14:09.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/23/691523/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/691523/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/691523/13"}}},"commit":{"parents":[{"commit":"1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11","subject":"Merge \"Fix api version not providing the full url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-10-03 19:38:10.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-11-17 17:13:46.000000000","tz":240},"subject":"Implement create/delete zone for Akamai v2 API","message":"Implement create/delete zone for Akamai v2 API\n\n- Ignore duplicate Zone error\n- Handle error when contractId or gid is missed\n- Ignore port for masters servers, because Akamai uses only 53 port and\ndoes not allow to specify any port in list of masters servers.\n- Added timeout and retries for soft Zone Delete\n- Added handling errors on the delete zone action\n- Added Log info message with RequestId on soft zone delete\n- Added processing for TsigKey during creation zone\n- Added devsatck_plugin for akamai_v2 backend\n\nDepends-On: https://review.opendev.org/#/c/692819/4\n\nChange-Id: Ib221f4cf0371e70fc6900582d826ffc1bdfc12b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b200a8830e470407d8818816f03bdb31ed313616"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b200a8830e470407d8818816f03bdb31ed313616"}]},"branch":"refs/heads/master"},"09929eb1613eadb159b15b9fd3e9ddb54797d0f3":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2019-11-28 01:39:11.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/23/691523/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/691523/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/691523/14"}}},"commit":{"parents":[{"commit":"d9fb34767c5bd9f8f822acc2b91715db45ab68db","subject":"Run grenade old side with python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d9fb34767c5bd9f8f822acc2b91715db45ab68db"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-10-03 19:38:10.000000000","tz":240},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-11-28 01:39:11.000000000","tz":0},"subject":"Implement create/delete zone for Akamai v2 API","message":"Implement create/delete zone for Akamai v2 API\n\n- Ignore duplicate Zone error\n- Handle error when contractId or gid is missed\n- Ignore port for masters servers, because Akamai uses only 53 port and\ndoes not allow to specify any port in list of masters servers.\n- Added timeout and retries for soft Zone Delete\n- Added handling errors on the delete zone action\n- Added Log info message with RequestId on soft zone delete\n- Added processing for TsigKey during creation zone\n- Added devsatck_plugin for akamai_v2 backend\n\nDepends-On: https://review.opendev.org/#/c/692819/4\n\nChange-Id: Ib221f4cf0371e70fc6900582d826ffc1bdfc12b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/09929eb1613eadb159b15b9fd3e9ddb54797d0f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/09929eb1613eadb159b15b9fd3e9ddb54797d0f3"}]},"branch":"refs/heads/master"},"b08a75232cd26fa3512116b648606930c4672a17":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2019-12-28 06:22:59.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/23/691523/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/691523/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/691523/15"}}},"commit":{"parents":[{"commit":"0050549441a1889a40b85accc4733d3dca8f067f","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0050549441a1889a40b85accc4733d3dca8f067f"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-10-03 19:38:10.000000000","tz":240},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2019-12-28 06:22:59.000000000","tz":0},"subject":"Implement create/delete zone for Akamai v2 API","message":"Implement create/delete zone for Akamai v2 API\n\n- Ignore duplicate Zone error\n- Handle error when contractId or gid is missed\n- Ignore port for masters servers, because Akamai uses only 53 port and\ndoes not allow to specify any port in list of masters servers.\n- Added timeout and retries for soft Zone Delete\n- Added handling errors on the delete zone action\n- Added Log info message with RequestId on soft zone delete\n- Added processing for TsigKey during creation zone\n- Added devsatck_plugin for akamai_v2 backend\n\nDepends-On: https://review.opendev.org/#/c/692819/4\n\nChange-Id: Ib221f4cf0371e70fc6900582d826ffc1bdfc12b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b08a75232cd26fa3512116b648606930c4672a17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b08a75232cd26fa3512116b648606930c4672a17"}]},"branch":"refs/heads/master"},"318b8d03193d3d9c26842528e9eaf25814df3ba9":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2020-01-16 18:54:30.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/23/691523/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/691523/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/691523/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/691523/16"}}},"commit":{"parents":[{"commit":"f355aae93955b76df9ea9ef5554b451df0ac91b4","subject":"Merge \"Improve sink recordset creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f355aae93955b76df9ea9ef5554b451df0ac91b4"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2019-10-03 19:38:10.000000000","tz":240},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-01-16 18:54:30.000000000","tz":0},"subject":"Implement create/delete zone for Akamai v2 API","message":"Implement create/delete zone for Akamai v2 API\n\n- Ignore duplicate Zone error\n- Handle error when contractId or gid is missed\n- Ignore port for masters servers, because Akamai uses only 53 port and\ndoes not allow to specify any port in list of masters servers.\n- Added timeout and retries for soft Zone Delete\n- Added handling errors on the delete zone action\n- Added Log info message with RequestId on soft zone delete\n- Added processing for TsigKey during creation zone\n- Added devsatck_plugin for akamai_v2 backend\n\nDepends-On: https://review.opendev.org/#/c/692819/4\n\nChange-Id: Ib221f4cf0371e70fc6900582d826ffc1bdfc12b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/318b8d03193d3d9c26842528e9eaf25814df3ba9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/318b8d03193d3d9c26842528e9eaf25814df3ba9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
