)]}'
{"id":"openstack%2Fdesignate~693429","triplet_id":"openstack%2Fdesignate~master~I605bbdfe7d0b734b31cdf85691f8c6d3850fc8c7","project":"openstack/designate","branch":"master","hashtags":[],"change_id":"I605bbdfe7d0b734b31cdf85691f8c6d3850fc8c7","subject":"Make designate reproducible","status":"MERGED","created":"2019-11-07 21:08:09.000000000","updated":"2019-11-20 00:27:27.000000000","submitted":"2019-11-20 00:25:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693429-1574209555162-a496186d","meta_rev_id":"9a4aa870cc1e013d337bb0922bcf47fabde31416","_number":693429,"virtual_id_number":693429,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},{"value":0,"date":"2019-11-11 09:05:55.000000000","_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-20 00:25:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2019-11-08 09:14:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2019-11-12 18:44:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-11 15:30:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"value":2,"date":"2019-11-18 23:48:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"value":1,"date":"2019-11-18 23:48:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2019-11-08 09:14:53.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-08 09:14:53.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-11-11 15:30:11.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-11-12 18:44:15.000000000","updated_by":{"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},"reviewer":{"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},"state":"REVIEWER"},{"updated":"2019-11-18 23:48:26.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-11-20 00:25:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a6bd94ca494e1912c4c544c77847998e3ded49b8","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2019-11-07 21:08:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5cfc32bf92913df2f8e3dcc5c4202d0d364f9515","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 22:15:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/70de400ebee34144a06100425bca1662 : SUCCESS in 35m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/49876143d8e445348381e1d73ea65d8c : SUCCESS in 41m 57s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/8143f150ef6d425eaad48a4fa5b10bb1 : SUCCESS in 39m 13s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/922463ebedbf46b48db5cc89c3152b0e : SUCCESS in 48m 57s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/050e29d28065431eb8dd9d9df77557da : FAILURE in 31m 22s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1395c384ea2f4abd85d7844245452ba3 : FAILURE in 1h 04m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cc14bd06378e4ff39374a4658341c616 : SUCCESS in 58m 11s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/04720043b11b45e6b5e1fa28e25a9be5 : SUCCESS in 36m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1aa670c219394f50812b9a06a6cef3fb : SUCCESS in 5m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed32af3d9c0b488abea3644514c57ec1 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/967b328742274127b1ac59f7843373d2 : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b6293733feca4110833945defc181ef7 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/142e5798748b422d9337fcb28b9b47ff : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/77a478fd36784e95b3aa3b4014e9cccd : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d55ea8a54a86411ab1cb730883c5ad50 : SUCCESS in 6m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ab28f9acd3194548a4aa4ca6497d964f : SUCCESS in 32m 31s","accounts_in_message":[],"_revision_number":1},{"id":"edf0425d16a0cd802cda41f3165c0c5c2233374d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-11-08 09:14:53.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1\n\nrecheck grenade failure seems unrelated","accounts_in_message":[],"_revision_number":1},{"id":"fee51df50ad3e48d584ca37129d5cb0ffc9b47fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 10:08:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cfcf1fa3e19c4ff99db47ced9f5763f6 : SUCCESS in 38m 43s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/37cc67970cf342a4804caf4c74dca239 : SUCCESS in 53m 01s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/cea8a0461d484d7a900cbf1d2a061d6d : SUCCESS in 35m 27s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/7f59a1cc371948099a0af2e2b3281217 : SUCCESS in 35m 11s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/0abc6fb2e6ce482db241f969bda0cb15 : FAILURE in 27m 49s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/83ec6d4aa66a4c6aa5a9c9d1d456c849 : FAILURE in 48m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/dc4fdaddc02d4052a83fdd6a9569deab : SUCCESS in 36m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/518d2657cc0848828d3133396e9852c0 : SUCCESS in 36m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b318cba6f8994f269000c553acb00b26 : SUCCESS in 4m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c40518a2660646a1bb684382cff5841c : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af32f4a9f3bf4f21aadc02df6dd11ac3 : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/756fa5ec37c949089491e63e3b1320d4 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67be133792f84382a45bfd206f40710c : SUCCESS in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ad009d54fa9a40cba13502e609fcc60f : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86f3d34b99314f829acde5f3625553a5 : SUCCESS in 12m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c9f95ab759dc4fb2af6f875fe527408a : SUCCESS in 33m 31s","accounts_in_message":[],"_revision_number":1},{"id":"5ab04ab3aba2a5840271362a023f2064c3ce9799","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-11 00:08:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"425a83667b0b7f7d12113e42a0dd7e60d3eecb0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 00:53:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/777bc702c59e4731a3fef8ea4fecdf66 : SUCCESS in 33m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5334f4469f7e485d91133d0cf5e41b16 : SUCCESS in 30m 57s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/6a0d8b77d99a49c9a8150cc5d3cbcbeb : SUCCESS in 34m 24s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/7ecd06b208a34060a03334bf16c559ec : SUCCESS in 37m 13s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/789f711b2b5f4640add13cee70d3b76f : FAILURE in 22m 38s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/38bf0b53ef2c41de8c94541d0d70b407 : FAILURE in 43m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c2d944b87f884c2f8785c660cd2e5de5 : SUCCESS in 33m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bf2ea61065cb41f6ad05b26b56a89733 : SUCCESS in 32m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cbf70fa03cb4a69a1ac0188d7ad7007 : SUCCESS in 4m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44772bbfc86641c0a0be5099f93e0fb6 : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eeeb9ccadf84cdfbd8b772bdd908215 : SUCCESS in 3m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab6f457c9a5c42eca710227f3dc6689b : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/796277fd8c444cf9929372671b22c27d : SUCCESS in 3m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/623069d05f2445d993dbc5626ab10558 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e4251cc20fb4b76849142f719382c18 : SUCCESS in 6m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f76637ea65af48cea361e4084ba40eb4 : SUCCESS in 29m 38s","accounts_in_message":[],"_revision_number":1},{"id":"d3cddefb6277646a836c5aa7f456f432177f395d","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2019-11-11 09:05:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1d28b0d1efa9c59305acc96e24b8ae060141a9e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 09:56:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3c06166f8eec40b78dcfa272945cbc28 : SUCCESS in 37m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cf3c84e8c2eb4ecb9d172667938055f6 : SUCCESS in 31m 45s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/7a670105338e4b1cb85acff7504f963d : SUCCESS in 40m 01s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/709096d21b5f4cf8862efcd7db4d2987 : SUCCESS in 40m 55s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/0caf4124cff2441abdb5e63e0a24ab67 : FAILURE in 26m 30s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b2763db04b1e424f8b5fbbc2114474c7 : FAILURE in 44m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4b988d21b61346c69911db869b547ba3 : SUCCESS in 35m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bba695bb00ff4631acab28ec67710cd8 : SUCCESS in 33m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70f350b71b7d4247a7c4d3cff66e0997 : SUCCESS in 4m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1fa51ceacda740a8bf751b374eccd8cf : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9901c5ebdae942eeaef688a96485072d : SUCCESS in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eef32362362a40c29f50977429359a2d : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83873572e4b34ebb8c34b5bd145ecacd : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cea50b4cf6d84f88a9e20f77d40f452d : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aa91bb8f4434f46817e37dd3320ab14 : SUCCESS in 6m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7cb70ae09b124e509d351aade42d9b0b : SUCCESS in 32m 15s","accounts_in_message":[],"_revision_number":1},{"id":"dc38990333ef0de927088d9a65df83d883485c8e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-11-11 10:29:59.000000000","message":"Patch Set 1:\n\ngrenade job is actually broken, see https://review.opendev.org/693671 and maybe https://review.opendev.org/693670","accounts_in_message":[],"_revision_number":1},{"id":"5bca5bf4dca8160d222ac68e208fc026c37c139e","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-11-11 15:30:11.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c4a6dbae939c1148cf53d9ae039d2297a1ea4907","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 16:44:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0ac358c9e3f84e3e85fb20907a0f4ed5 : POST_FAILURE in 1h 09m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9b1462df53e74306915a087b970563a7 : SUCCESS in 35m 10s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/3660228c733341e8bfa0352432ee60d4 : SUCCESS in 46m 14s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/f95ec0c84f7c4121bc8bc2ea10c6b658 : SUCCESS in 36m 55s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/0157d0754c8a495b95aef00a22934e88 : FAILURE in 27m 48s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5d20d6e4a262449c80934a5c68466a4e : FAILURE in 58m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/118a5ee88dca4b78b41c82915c64d5f9 : SUCCESS in 37m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3c882598eeed4494bde1d2d02bccbb3d : SUCCESS in 40m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e7d75cc7abf4910af61e3b1b634480a : SUCCESS in 5m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1704a37f926543bea38a6f4c534fb6a0 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/328261e0014c411fa5a8a71d1e30d266 : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/97188a8f5814414f852dfbf07ef87229 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f8b5beed41245b984d1f3f4c7b09c60 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/59752abd37b547ad9c3a8bb4716236d0 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24f0474a95b14c3da379f24a0c538e3e : SUCCESS in 8m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ece5db3892c04cae88aa8f33ed2ce64f : SUCCESS in 42m 02s","accounts_in_message":[],"_revision_number":1},{"id":"0fc71e9fb6a1d840c32d4e900df4f0e74b757fde","author":{"_account_id":15736,"name":"Ritesh Anand","email":"tellrit@gmail.com","username":"ritesh"},"date":"2019-11-12 18:44:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"eec6d28213cc740a15af65f4d4326899ed88d07f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-11-15 13:41:22.000000000","message":"Patch Set 1:\n\nrecheck grenade job should work better now","accounts_in_message":[],"_revision_number":1},{"id":"5d4049804eded1290b590c291498a75ca53cb6a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 14:34:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/42620511ed8f4f6e80134e8c45432c57 : SUCCESS in 41m 19s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cbd2db8628ec4bfeb6c06c476a26543e : SUCCESS in 37m 42s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/6a832fc37960414e99bda2223dde007f : SUCCESS in 40m 54s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/de3136a944874459a9fa45c8f789d597 : SUCCESS in 49m 49s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/ab86d819e6f24d4cadc8c8d68c54e5e3 : FAILURE in 35m 19s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c0fa3ef71293429786d714265c6df4c0 : FAILURE in 51m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/211faca79969473083b1058558e084eb : SUCCESS in 48m 28s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ba25f6753f824b6abaa09191285cf398 : SUCCESS in 50m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b5df44a59944e0d9a3e93f357e6f8ab : SUCCESS in 6m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9013b0c9afdd4720842bcddc36c63443 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68b3bbd0da3c480b974220f8afe1b373 : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9e22929317fe46b693b0397dfff52fcc : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5efdf8f17724209ae04f627c9e70e7b : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e51b81fcbafb499cb73ac3f1d8d862f1 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d51e710c05004a4c8d5c83a1fb211fc1 : SUCCESS in 8m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9d6fd059e58d4f789b053bbf5790de9a : SUCCESS in 36m 25s","accounts_in_message":[],"_revision_number":1},{"id":"3bbdc53c4b87beaaf801e568a8104adc7ce5cf40","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-18 18:04:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"05bd0de251bd5cc4a2f99c7e669f68ed6a7bf69b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 19:12:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/41ea70951eda49209363810dd7d47cf1 : SUCCESS in 39m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bca6727b657247908670458c4a6e0935 : SUCCESS in 44m 18s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/4f5032fad3184a958b296935d575b25d : SUCCESS in 39m 42s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/6bef5017bba1408eb86a72310fe880eb : SUCCESS in 41m 03s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/9ce58ce0977f42178c5bbe7c2124d55c : FAILURE in 35m 14s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/dd5e2f470c7648abbb7387d63c537866 : SUCCESS in 59m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ad7b36eb21bf46c49bf2101e8153ccbc : SUCCESS in 43m 54s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e72e2b6927b7485ba77d1c280e57786a : SUCCESS in 43m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f23066c0d8e4c98b4f6a1efd0acdc95 : SUCCESS in 4m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/580ab514a8704846a3cf63830c1e9de2 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4fcd26cb60e4815ae2f761e2dd50703 : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/25019a9e7cbb4cf699c3f7e3311a9bf1 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54a6a3d2c28c466fa931720655f97320 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f428c337170845bfacd51bb35f5b8034 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62c238f4f2bb4a88b4b38628742f1265 : SUCCESS in 7m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c7efb24059ab46c5a5f2dc90b510f941 : SUCCESS in 36m 05s","accounts_in_message":[],"_revision_number":1},{"id":"005db8913fd3a02a758cb81d2669036b0ce30436","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-18 23:48:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d3ad9957b54d7f440c43cc28a95cc0298e24f696","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 23:48:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"94652f7efd10ee1a8768090a80a74c75e826746a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 00:42:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/af201a7d072340c5bc3958341d4ffc5c : SUCCESS in 32m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/10d8401867e84724a85cb4ed1cbba513 : SUCCESS in 45m 44s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/aec5b28bcb904bb485cb33eed3257220 : FAILURE in 22m 54s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/ee3841859b20421daf06f3be9c1911fb : FAILURE in 16m 02s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1ed2334f7fb14d18b334b6f77067cf15 : FAILURE in 49m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a4529be1b3094bbe9e4705b2bc231d04 : SUCCESS in 45m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/54e8562ce0b2447ab9260ea2eff91eb4 : SUCCESS in 47m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/322c2cdd2a06430688824cb6795501d6 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d06a1a72908448a98bdedd985d477c18 : SUCCESS in 7m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4bf3ef04bcea4e09b76ed6d1c9f8ae04 : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d140979ac624f41a5ba1b745c8a7a81 : SUCCESS in 7m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f78b6cd6ba3445cab1546323212de012 : SUCCESS in 10m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0ef3c6e03f14d62b1a9bc91dcdddd3f : SUCCESS in 15m 22s","accounts_in_message":[],"_revision_number":1},{"id":"8471334ede0ff8f7fdfc58383da39d45c86690bf","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-19 22:00:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6ed79a84afc8ae0849a2847d3adb908def73bfc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 23:04:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dc4e0eb44228466aa3da59a5d2605678 : SUCCESS in 33m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/aebec2173717401e886fd6d0337bcf8c : SUCCESS in 32m 50s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/65c22190a5024513b146d8151373f7d7 : SUCCESS in 39m 49s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/47f98c1399c74ee6a4522bf4d774e023 : SUCCESS in 41m 06s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/c343aa969b844a5690e6ea5a1b502097 : FAILURE in 27m 31s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a911b1ade31f4b20b8bb6c260cb65838 : SUCCESS in 57m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/585f6e2fa60949da8de1899476526cde : SUCCESS in 38m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b786c4ace75a46a7a37245ca009d8dec : SUCCESS in 36m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32bb932fe1bc4d0da2734b6ec646891e : SUCCESS in 5m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/945f15eb7e5148729e4d6f9bfd836568 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc29a8f745554b1ba49f8a2d4df3a7d9 : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c0334f7f461d40459cb952973c4c1100 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0a163fa5b1345b895c60712468fb39e : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1fc2c32b644f4a5491921328fd63d46f : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0680dac9d510421fad001c2882d1d004 : SUCCESS in 6m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c9cfb1a20d1a47dcba814df7f975cb2b : SUCCESS in 33m 24s","accounts_in_message":[],"_revision_number":1},{"id":"6eb5b6d1ba42cab31f4f1e7a7ffa220f79e55c45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 23:04:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"87736f5efac6d030162fc469239d6bf31f747d4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 00:25:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b26866199dd34ba99c60ee010f05b22f : SUCCESS in 36m 34s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6a82c78c90fb4436b089d0a39715c053 : SUCCESS in 35m 39s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/90d6d8b4fbfb455892b868aad7857f46 : SUCCESS in 40m 48s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/b9808f536017460586ed9294771c237a : SUCCESS in 37m 25s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b1bb841e1a0d417584c8c005a4d7353b : SUCCESS in 56m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0dc58f0a14d54b0186e532fb27f3d248 : SUCCESS in 41m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/37e19b62696844c1ad3990974b82ee96 : SUCCESS in 37m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd380dbbeae54190b9fe291870cfb4f8 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe5fcda63041482da640dd5c7c6cd9f2 : SUCCESS in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7e08b40a5d7b4b92a4bb757883a343bf : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/788444bb2f1f4827b1794b61377a8f3b : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/345907d424b040d9bc229125a75f0fa7 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b34994cda28c4567acf7739942670892 : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":1},{"id":"03eaa99b412db452ef210abd2d4264ad75e90643","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 00:25:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9a4aa870cc1e013d337bb0922bcf47fabde31416","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 00:27:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93bf727328c14fcdb26fa95b299eb93f : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2ef51c5ee2ad65411ff6e431557644ab4d73e28b","revisions":{"2ef51c5ee2ad65411ff6e431557644ab4d73e28b":{"kind":"REWORK","_number":1,"created":"2019-11-07 21:08:09.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/29/693429/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/29/693429/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/693429/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/693429/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/693429/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/29/693429/1"}}},"commit":{"parents":[{"commit":"1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11","subject":"Merge \"Fix api version not providing the full url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1cd2dddf3f181a792b9cb2e9fd02b11ea9896f11"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2019-11-07 21:06:53.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2019-11-07 21:06:53.000000000","tz":60},"subject":"Make designate reproducible","message":"Make designate reproducible\n\nI\u0027m forwarding the patch against the Debian package at:\nhttps://bugs.debian.org/942848\n\nWhilst working on the Reproducible Builds effort [0] we noticed\nthat designate could not be built reproducibly.\n\nThis is because it embeds the absolute build path in the sample\nconfiguration. Patch attached that uses oslo-config\u0027s sample_default\ninstead.\n\n [0] https://reproducible-builds.org/\n\nChange-Id: I605bbdfe7d0b734b31cdf85691f8c6d3850fc8c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2ef51c5ee2ad65411ff6e431557644ab4d73e28b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2ef51c5ee2ad65411ff6e431557644ab4d73e28b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
