)]}'
{"id":"openstack%2Fdesignate~694926","triplet_id":"openstack%2Fdesignate~master~Ifcca742fae45c031748224eb460ef83ecec6ad35","project":"openstack/designate","branch":"master","hashtags":[],"change_id":"Ifcca742fae45c031748224eb460ef83ecec6ad35","subject":"Stop testing Python 2","status":"ABANDONED","created":"2019-11-19 07:51:53.000000000","updated":"2019-11-21 20:51:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6d0f07aff3dd6b64a3df239ff0f9be3308ba57ae","_number":694926,"virtual_id_number":694926,"owner":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-11-19 08:49:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-11-19 08:00:54.000000000","_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-2,"date":"2019-11-20 20:35:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-19 08:00:54.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-11-19 08:49:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-20 20:35:33.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"d0b858ab28a02b5d2e5578aff581c169252b0c59","author":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"date":"2019-11-19 07:51:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b02f69c8dc375a64438d61a6add65f4e5c2a7dfc","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-19 08:00:54.000000000","message":"Patch Set 1:\n\nIsn\u0027t this a dup of https://review.opendev.org/#/c/694439/ ?","accounts_in_message":[],"_revision_number":1},{"id":"efd84d6b6b72deb57eb380676e820990604e9e6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 08:49:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/86e1d5eb280f4db08c0fc674586d4354 : SUCCESS in 4m 05s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4c39a8c8a7234a3188876e6667359d7d : SUCCESS in 1m 51s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/060ee88d805e493aaa2aea99b7872848 : SUCCESS in 41m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/282055b89aa54bbb8ad994a81ba9ae68 : SUCCESS in 40m 37s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/ee21819963464ba199d19a49246c7121 : FAILURE in 19m 19s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/252aab08c23749708556cc502b3d07ed : FAILURE in 20m 50s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/016549db7ad04722983475ec3233f74d : FAILURE in 32m 05s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/84dfbd2c79bf4e74a6087ea629aa8e66 : FAILURE in 50m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a2921378ada44fd0b37122555260c425 : SUCCESS in 41m 20s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7cd927b21b0b4626b453d6a907567995 : SUCCESS in 39m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d8c6c56739e4c0ba5170b17da55f7a5 : SUCCESS in 4m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b6a37e37e9d4b3cb8107bac489add00 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31ec2be1c3984d0383bc8ab118977d64 : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f04cc873cd84490c9f66f438bd5efbf8 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb2dc3f2bec24980bde13b225e45e816 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/328e38f36d244faca33af3a88a67cc89 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dee37090a3924a72a4d2f95ee781e964 : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8eaab8c6008d4e7588e52fda9906d1eb : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":1},{"id":"b54179db015547cbdf810fd0cd75ebd858506dc1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-11-20 20:35:33.000000000","message":"Patch Set 1: Code-Review-2\n\nIt is a duplicate and it doesn\u0027t even do what the commit message says.","accounts_in_message":[],"_revision_number":1},{"id":"6d0f07aff3dd6b64a3df239ff0f9be3308ba57ae","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-11-21 20:51:01.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f5959e7cfa37dcab60c784c2f8a4cdbf6b07ead6","revisions":{"f5959e7cfa37dcab60c784c2f8a4cdbf6b07ead6":{"kind":"REWORK","_number":1,"created":"2019-11-19 07:51:53.000000000","uploader":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"ref":"refs/changes/26/694926/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/694926/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/694926/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/694926/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/694926/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/694926/1"}}},"commit":{"parents":[{"commit":"306e231c4ba63fcfd2bbafc34718c1aba333c755","subject":"Don\u0027t install Horizon as part of the upgrade test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/306e231c4ba63fcfd2bbafc34718c1aba333c755"}]}],"author":{"name":"pangliye","email":"pangliye@inspur.com","date":"2019-11-19 07:46:56.000000000","tz":480},"committer":{"name":"pangliye","email":"pangliye@inspur.com","date":"2019-11-19 07:51:15.000000000","tz":480},"subject":"Stop testing Python 2","message":"Stop testing Python 2\n\nIt\u0027s Ussuri. We can stop testing Python 2\n\nChange-Id: Ifcca742fae45c031748224eb460ef83ecec6ad35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f5959e7cfa37dcab60c784c2f8a4cdbf6b07ead6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f5959e7cfa37dcab60c784c2f8a4cdbf6b07ead6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
