)]}'
{"id":"openstack%2Fdesignate~695137","triplet_id":"openstack%2Fdesignate~master~I143ddd4f197950b94114f0ddb62a06509c77ebc8","project":"openstack/designate","branch":"master","hashtags":[],"change_id":"I143ddd4f197950b94114f0ddb62a06509c77ebc8","subject":"Migrate grenade jobs to py3","status":"ABANDONED","created":"2019-11-20 07:24:37.000000000","updated":"2019-11-21 00:53:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"24af5bf866c4228aed61f4993356f785ddc3fcde","_number":695137,"virtual_id_number":695137,"owner":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"date":"2019-11-21 00:53:30.000000000","_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-11-20 08:28:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":-2,"date":"2019-11-20 21:15:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":-2,"date":"2019-11-20 20:33:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-20 08:28:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-20 20:33:55.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-11-20 21:15:49.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"}],"messages":[{"id":"47f9ba67867711a55979fe8784f03e7b02d55638","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2019-11-20 07:24:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb3a912b73be7818b2d869cc7f3a922dfd8f5776","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 08:28:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/efe94e1ad7714496aa928c550670f014 : SUCCESS in 38m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1e8c29fd10814602924f455fd19cb74e : SUCCESS in 45m 55s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/9487830f903c4a2882defed25a482c7e : SUCCESS in 42m 12s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/2aa196e227f84da7a45434cfdb9b6d5a : SUCCESS in 41m 11s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/19db07be2c0e46538399e97514b16b35 : FAILURE in 30m 58s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/51b4b2b7abad4c7882dd2bd361c420fe : SUCCESS in 57m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ff70da1a466f4644bff97fa6e2bf41d6 : SUCCESS in 37m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8ceea6bb40744afcb9562eb344299c5d : SUCCESS in 43m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c6b8d83ec12487e991becd7bcd6e0ac : SUCCESS in 4m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10f93a6ea46941638901a922c1963585 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/631189497aa24234a8bbdc3aedf27fff : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/46f615c49de34ab582fd307a146d2315 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d947bdd5cc14131b9a844692079d200 : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8d5abfb39826421daa512c240f6e9dc8 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3776d20f53b349f59366dfa5c58aee5d : SUCCESS in 6m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7a17e6b8db0242c08ff185a62b001188 : SUCCESS in 36m 09s","accounts_in_message":[],"_revision_number":1},{"id":"31a4f7d2390ae777549498047d051f76011b2cee","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-11-20 19:05:45.000000000","message":"Patch Set 1: Code-Review+1\n\nChange looks good to me. Thanks!","accounts_in_message":[],"_revision_number":1},{"id":"96d3d616a2228ec2a0d43df209fc3714a18ea2e2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-11-20 20:33:55.000000000","message":"Patch Set 1: Code-Review-2\n\nThis is already included in https://review.opendev.org/#/c/694439/4","accounts_in_message":[],"_revision_number":1},{"id":"c6fd7252d00f567dfd2a21a437e0131a81f74f90","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-11-20 21:15:49.000000000","message":"Patch Set 1: Code-Review-2\n\nSorry I missed that Jens.","accounts_in_message":[],"_revision_number":1},{"id":"3946e2ec72417cd1dab7e708835fcc7c9d5df028","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2019-11-21 00:53:30.000000000","message":"Patch Set 1:\n\n\u003e This is already included in https://review.opendev.org/#/c/694439/4\n\nOK, I got it, thanks","accounts_in_message":[],"_revision_number":1},{"id":"24af5bf866c4228aed61f4993356f785ddc3fcde","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2019-11-21 00:53:42.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"170f2780a8cdf10f6812a56096909deb4ee55ec4","revisions":{"170f2780a8cdf10f6812a56096909deb4ee55ec4":{"kind":"REWORK","_number":1,"created":"2019-11-20 07:24:37.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/37/695137/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/37/695137/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/37/695137/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/37/695137/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/37/695137/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/37/695137/1"}}},"commit":{"parents":[{"commit":"7951a7d7f5f870f5b97322e252a14eeb303413da","subject":"Merge \"Make designate reproducible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7951a7d7f5f870f5b97322e252a14eeb303413da"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-11-20 07:24:05.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-11-20 07:24:05.000000000","tz":480},"subject":"Migrate grenade jobs to py3","message":"Migrate grenade jobs to py3\n\nAs part of community goal of dropping py27 support[1], we are\nmoving the devstack to py3 by default[2]. That will make grenade job\nto perform upgrade from py2 to py3 which will not work (we have seen\nthe failure in neutron-grenade job).\n\nTo avoid existing grenade py2 job break, this commit moves grenade\njobs to py3 which is what we planned as part dropping the py2 support.\n\n[1] https://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n[2] https://review.opendev.org/#/c/649097/12\n    http://lists.openstack.org/pipermail/openstack-discuss/2019-November/010938.html\n\nChange-Id: I143ddd4f197950b94114f0ddb62a06509c77ebc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/170f2780a8cdf10f6812a56096909deb4ee55ec4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/170f2780a8cdf10f6812a56096909deb4ee55ec4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
