)]}'
{"id":"openstack%2Fdesignate~697422","triplet_id":"openstack%2Fdesignate~master~I9c7adff0f684146102fac89d1cee74a785ab70d9","project":"openstack/designate","branch":"master","hashtags":[],"change_id":"I9c7adff0f684146102fac89d1cee74a785ab70d9","subject":"Fix devstack plugin python3 support","status":"ABANDONED","created":"2019-12-05 08:57:12.000000000","updated":"2019-12-05 09:57:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"122fb28ab8c1943fd6ab8b2354b45a14f1d05c8e","_number":697422,"virtual_id_number":697422,"owner":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"date":"2019-12-05 09:57:11.000000000","_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-05 09:54:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-2,"date":"2019-12-05 09:41:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-05 09:41:10.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-12-05 09:54:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d46c4c64af001ee49f815812645349f80bcd4277","author":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"date":"2019-12-05 08:57:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8aa20e5efc447f061d99e824423fb384617b542","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-12-05 09:41:10.000000000","message":"Patch Set 1: Code-Review-2\n\nThis is to be fixed in the dashboard, see https://review.opendev.org/#/c/694440/6/manage.py","accounts_in_message":[],"_revision_number":1},{"id":"010b40d8a8ee7df8bd9d5d209c0c2157410598a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 09:54:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3ceb09e6a0f44e009535970546043303 : POST_FAILURE in 39m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/382633e4e7a74776ac05b76b3dc6f129 : SUCCESS in 36m 37s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/6ea14c964d124e2d93d54ea4ada40d22 : SUCCESS in 39m 16s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/587da2767e3743908b7e88b9710165b2 : SUCCESS in 35m 30s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/c61c3ab3097e43e2848383d167f64fba : FAILURE in 26m 46s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/379cb44097af4e429a15ca2029e4e72d : SUCCESS in 52m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ec2ea5a851cc4ad7bcc6f13e2112f0b4 : SUCCESS in 35m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/35230369036f4dc58b4f9d7452c5ed02 : SUCCESS in 37m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7be229246a8b4537b8a319d5eee729c8 : SUCCESS in 4m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83835730a9204aa98d5e8622bf35dfc0 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e0bed68b1c84eeba3102560f81909b9 : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/654ebbba22bd486b876a212b6e8e0a0d : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07cf345af04e4b02826afd8f7b2cf80a : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e3bd5485095843a3868ff00707b5b429 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f3a90750df64e549ba2dd64c3087efb : SUCCESS in 7m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/947f6a0392b04febab6b2790b1386cf0 : SUCCESS in 33m 59s","accounts_in_message":[],"_revision_number":1},{"id":"0b35e0d982ffadd55b7c7cc625b762df9493d050","author":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"date":"2019-12-05 09:57:11.000000000","message":"Patch Set 1:\n\nok. thanks.","accounts_in_message":[],"_revision_number":1},{"id":"122fb28ab8c1943fd6ab8b2354b45a14f1d05c8e","author":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"date":"2019-12-05 09:57:14.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c4a7aef824fd3dbca5affa27c4361c308fd9c7ff","revisions":{"c4a7aef824fd3dbca5affa27c4361c308fd9c7ff":{"kind":"REWORK","_number":1,"created":"2019-12-05 08:57:12.000000000","uploader":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"ref":"refs/changes/22/697422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/22/697422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/22/697422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/22/697422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/22/697422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/22/697422/1"}}},"commit":{"parents":[{"commit":"d9fb34767c5bd9f8f822acc2b91715db45ab68db","subject":"Run grenade old side with python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d9fb34767c5bd9f8f822acc2b91715db45ab68db"}]}],"author":{"name":"asarfaty","email":"asarfaty@vmware.com","date":"2019-12-05 08:56:13.000000000","tz":120},"committer":{"name":"asarfaty","email":"asarfaty@vmware.com","date":"2019-12-05 08:57:07.000000000","tz":120},"subject":"Fix devstack plugin python3 support","message":"Fix devstack plugin python3 support\n\nThe devstack plugin.sh script was calling explicitly\n\"python\". This causes the plugin to fail on python3 only hosts.\nThis patch corrects this by using the devstack $PYTHON variable.\n\nChange-Id: I9c7adff0f684146102fac89d1cee74a785ab70d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c4a7aef824fd3dbca5affa27c4361c308fd9c7ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c4a7aef824fd3dbca5affa27c4361c308fd9c7ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
