)]}'
{"id":"openstack%2Fdesignate~699106","triplet_id":"openstack%2Fdesignate~master~I4cdef86ef4412eb92668bb08a5de73249f23ca9a","project":"openstack/designate","branch":"master","topic":"bug/1856442","hashtags":[],"change_id":"I4cdef86ef4412eb92668bb08a5de73249f23ca9a","subject":"fix bug: secondary zone loop xfr","status":"ABANDONED","created":"2019-12-15 09:39:08.000000000","updated":"2022-09-24 05:52:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b87d35525a2090be1f8cd6adbd3706a2c3227579","_number":699106,"virtual_id_number":699106,"owner":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-23 02:13:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"date":"2020-10-23 18:11:20.000000000","_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-23 02:13:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-23 18:11:20.000000000","updated_by":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"reviewer":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"state":"REVIEWER"}],"messages":[{"id":"6650a1efdf034e214b75bf4096ead07f908b1ed2","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2019-12-15 09:39:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9de23ec3a15d55fa95089899ee3d5e32532a198","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 10:17:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1c0f5a31a92943b2b14b262f8d9e3288 : FAILURE in 10m 42s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3dd34db7775d46b4a1390e244535d31a : FAILURE in 9m 42s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/70b49dbb5e674107906524861bdc04c3 : FAILURE in 10m 45s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/114ea4a157b64a589a16a10d08b34101 : FAILURE in 11m 09s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/d6e6fc5e3b6b4287b8505a183da97a3a : FAILURE in 10m 09s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a5b34ae8c3e74653afe411a84a674201 : FAILURE in 36m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/02d11a5aa07142bb8f5df2c8bb1cf957 : FAILURE in 9m 28s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/25fc9ba719f74aa4a85c8aad287191d1 : FAILURE in 10m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d22964a9122142d79b97434156de0105 : SUCCESS in 4m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/592d0de8665a4fe0b710687615abff35 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1064d51e8d824eedbb3fb1f4307e8886 : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0d5defaf764244318a48eae1894a6966 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75ff6a04e4bd4761981ac69175be2808 : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3d6c92455aa644aeb7f5a3a86724901c : FAILURE in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/741a60595ec54eab9382d726b71716a4 : SUCCESS in 7m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ad386b2e4bda4d65b4b55dd9333cd746 : FAILURE in 11m 28s","accounts_in_message":[],"_revision_number":1},{"id":"7fffe46d170d3920a46f9ded37b10589d154f0a4","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2019-12-16 01:09:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4449a9e9e87d46e528425e2f8e68b2345524a7f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 02:16:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/340115f7260f4c289fc6bd85cdc9dd96 : SUCCESS in 45m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9992a923fcf74ecf8c6595ca62a236a1 : RETRY_LIMIT in 8m 33s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/6bbbff2c95664001a20a3dd6373bcc4d : FAILURE in 17m 00s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/ef962b0e37c84c438bbefb0fee20979a : FAILURE in 18m 29s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/ef8bcbc3880c42d384ce734b63ea6db4 : FAILURE in 33m 53s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/db4dbe7945b34e04b9d1f11c9029dc51 : FAILURE in 34m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/641bac2f29e649f08c0755883df90246 : SUCCESS in 36m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/00ba9307d7734a3c84304b3b1e2f32b0 : SUCCESS in 36m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/498d6573d2ee4eff9a9444faba90dff6 : FAILURE in 8m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f48b0bdafe7400c9ef03d822f8b8ecc : FAILURE in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd06f28cb05542e39b6369dbc8d0979f : FAILURE in 9m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/90021893471b4398a2cbbf4e27d914f2 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bfa2ebd0a2447d0812112cf1d50fefd : FAILURE in 8m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df8e54d1f42c42968d348876942fd6e9 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62f9759e074c4c01a0c9ee143d8f57bb : SUCCESS in 7m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/290e1e15bbc74f3e91a5d541bdbeb6d3 : SUCCESS in 36m 10s","accounts_in_message":[],"_revision_number":1},{"id":"8f38e4efa59c9a6db1cd279230417e3e20b61e8a","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2019-12-16 02:51:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2cf3d4041903f351e9ef2afeb74d6e329984d130","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 03:59:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d0b59201737047f0a7580396db48a280 : SUCCESS in 47m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/89bef0351c6e48ad98030c31e256e56e : SUCCESS in 51m 24s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/d946f9c49d694e488c6d6efadef308ad : FAILURE in 15m 38s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/0e4a0bf71a1b4d678944c7e9cd5179fb : FAILURE in 22m 39s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/7fe4c138541b476684c85c6907e6f0ad : FAILURE in 35m 20s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/02c40eca141540b28e4bd2547abfbce0 : SUCCESS in 1h 00m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a6642987ef0e4ada9ef77600ba7a871f : SUCCESS in 48m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/eb3bb56e62fb4737a82c7c0495a8ac77 : SUCCESS in 50m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78be620437894f7182430b50eeb75eba : SUCCESS in 7m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89174cc2c1ab4725abf014e4897c1c06 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5c6b9b1601c40eb8f25d721b4db9f2c : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8aa5e96d41094b8d9be3025d9e485124 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04b72d1a9cba46aebc4cae2fca132872 : SUCCESS in 7m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/de522589df68411d9ee5f2d493f8d7cd : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/126e28528939413a8e031d2e9994d7d9 : SUCCESS in 7m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/db08b67a10ad4811a00f35ca03f012b6 : SUCCESS in 33m 34s","accounts_in_message":[],"_revision_number":1},{"id":"6bd626498b84096f70ce3290897bfa7de52154a6","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2019-12-16 05:47:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c8fdd7cc01f3820ab74703b8eb6cc8e7436ce22b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 06:53:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9b2e393347cd455e8082655d23ff2846 : SUCCESS in 45m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7c9d958a11af456b859a77c4f11d1215 : SUCCESS in 45m 13s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/2b89f5990a3747bb928108329c740c44 : FAILURE in 23m 05s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/6a212812cd064a4db205992e8c167144 : FAILURE in 19m 28s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/74af3b6dc95942bc9ee704bd6a9fea32 : FAILURE in 30m 51s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2d6eda515d294f0e84d376d4b1bd7c3a : SUCCESS in 1h 03m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/dc82cdcc8d7a42558cd6c2814d54c6dc : SUCCESS in 38m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9e60e309cfee45b6930a1ea626076ddb : SUCCESS in 45m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac0119bdd6bf43ed974073fc9ac82999 : SUCCESS in 4m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9f7c4b1b9fc41d1bb15ddd74003ff79 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/637bc7709d1846c688374f6c61b9292d : SUCCESS in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c8d7e90518bb442cbe0fb7a9eac51c58 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd018a3386234ee7aeaef291257e1554 : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8eea50f8aec84569b44a24d245888ff4 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca742cc0976d48a2aecb9cd29604bd93 : SUCCESS in 7m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d5dbff90cae64e34991121e2c3ea69f8 : SUCCESS in 36m 33s","accounts_in_message":[],"_revision_number":1},{"id":"263a4b597c57be57298b8bcaf39f90abef7384e1","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2019-12-16 09:07:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"368b402a849b697d646d59d7e0b696ded0954806","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 10:44:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9a3eac9b29a24133ab63838667320d0b : SUCCESS in 40m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/eedbeb3450434eb88116fed6bd7dc498 : SUCCESS in 49m 41s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/8a650e2b7b74408f98be058773929881 : FAILURE in 32m 11s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0c04e0e216994b86a8ffad3df5597955 : SUCCESS in 1h 00m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c7588d395f274dc180b60510051fae27 : SUCCESS in 39m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5e401747a61946b7886ddd29753b08a9 : SUCCESS in 39m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d384307d7374f34a0ac57e5b0f43fe5 : SUCCESS in 6m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a82b80f7bbb4d7fb0e4b4278666af4b : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f6a1c7b70724d7aa00fbd3476f48f58 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b4ed20e64a646cf9f66de35d1067763 : SUCCESS in 7m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/caeb1e5d2e6c40adb88d4f019b7af7fe : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2908e9a855d74df58e675ce0a075250c : SUCCESS in 7m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/91a3e284c197464e959a0cdf88d6845c : SUCCESS in 39m 35s","accounts_in_message":[],"_revision_number":1},{"id":"eef1271012e50b4aa1163f95ea111729183957f3","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2020-10-23 01:12:30.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d76fb2bfb7d1b4194616e7f5457ab1617091dcea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 02:13:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f0cbce3137a9464f83c99d17f91f0f9b : SUCCESS in 39m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/564d27bb7d9848fb902c07fbe36c70d4 : SUCCESS in 44m 00s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0584f77a7bfc4a33842b93092ed937e0 : SUCCESS in 50m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0f9fb08802bc4de2bd9ae592026179e0 : SUCCESS in 44m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ffefb33916a74ff29cb711855a1a515a : SUCCESS in 42m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af2f1d0decc24278a71193a1c0c7e364 : SUCCESS in 6m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8b793613eec47eeb0094789613b0785 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e1ab35fe00d496d87dcf9c5f6a7dd2a : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/426230afc22d4a1dbceb97a44863a343 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbeee9ea6f1944caa2fee8b028ff7fb0 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d24c0bc0e774d17858271718dba74b1 : SUCCESS in 8m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d88425cccc2e428eb0979cf9130b7ac5 : SUCCESS in 37m 24s","accounts_in_message":[],"_revision_number":2},{"id":"938f45563101cc537ccc74b9ffedda59da5e25c1","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-10-23 18:11:20.000000000","message":"Patch Set 2:\n\ncan u please explain more about the bug, and what is the scenario you are trying to fix","accounts_in_message":[],"_revision_number":2},{"id":"b87d35525a2090be1f8cd6adbd3706a2c3227579","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-09-24 05:52:58.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"78ceca0e2b44635fab225341bbba68a9fb37240f","revisions":{"b185217c5f7290d3082f6a58567be8b5512aa9a5":{"kind":"REWORK","_number":1,"created":"2019-12-15 09:39:08.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/06/699106/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/699106/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/699106/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/699106/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/699106/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/699106/1"}}},"commit":{"parents":[{"commit":"d9fb34767c5bd9f8f822acc2b91715db45ab68db","subject":"Run grenade old side with python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d9fb34767c5bd9f8f822acc2b91715db45ab68db"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2019-12-15 09:35:28.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2019-12-15 09:35:37.000000000","tz":480},"subject":"fix bug: secondary zone loop xfr","message":"fix bug: secondary zone loop xfr\n\nwhen create a secondary zone.the zone will perform zone xfr from master.\nNormally, only one xfr should be performed. but not is loop xfr.\n\nthis patch fix it.\n\nPartial-bug: #1856442\n\nChange-Id: I4cdef86ef4412eb92668bb08a5de73249f23ca9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b185217c5f7290d3082f6a58567be8b5512aa9a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b185217c5f7290d3082f6a58567be8b5512aa9a5"}]},"branch":"refs/heads/master"},"78ceca0e2b44635fab225341bbba68a9fb37240f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-10-23 01:12:30.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/06/699106/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/699106/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/699106/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/699106/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/699106/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/699106/2"}}},"commit":{"parents":[{"commit":"769a55c2e523d749628d6ff4a813344801312d13","subject":"Merge \"Fix lower-constraints test not constraining properly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/769a55c2e523d749628d6ff4a813344801312d13"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2019-12-15 09:35:28.000000000","tz":480},"committer":{"name":"ZhouHeng","email":"zhouhenglc@inspur.com","date":"2020-10-23 01:12:30.000000000","tz":0},"subject":"fix bug: secondary zone loop xfr","message":"fix bug: secondary zone loop xfr\n\nwhen create a secondary zone.the zone will perform zone xfr from master.\nNormally, only one xfr should be performed. but not is loop xfr.\n\nthis patch fix it.\n\nPartial-bug: #1856442\n\nChange-Id: I4cdef86ef4412eb92668bb08a5de73249f23ca9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/78ceca0e2b44635fab225341bbba68a9fb37240f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/78ceca0e2b44635fab225341bbba68a9fb37240f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
