)]}'
{"id":"openstack%2Fdesignate~700320","triplet_id":"openstack%2Fdesignate~master~I3d8d4021b17b0522ab3a37a337f2a7a1936ca23b","project":"openstack/designate","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"I3d8d4021b17b0522ab3a37a337f2a7a1936ca23b","subject":"Imported Translations from Zanata","status":"MERGED","created":"2019-12-22 09:04:26.000000000","updated":"2019-12-26 00:37:52.000000000","submitted":"2019-12-26 00:36:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700320-1577320586987-a981c286","meta_rev_id":"48fe2f11253667a2182371af77c9e17c5d18ba26","_number":700320,"virtual_id_number":700320,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-26 00:36:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-25 23:32:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-25 23:32:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-25 23:32:13.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-12-26 00:36:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7d0f1f993541abb343ac1ebebf69d80135357be7","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-22 09:04:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f92094dda24140fe0235eeb0a5c5d58cc44fc5b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-22 09:54:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3e1ae765c58e469b8b8e4ee2e1a95ab2 : SUCCESS in 38m 43s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c23960aba57f4019a5810de87769ca56 : SUCCESS in 37m 08s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/39ca56cee2b948298700b4c0729aedb2 : FAILURE in 27m 36s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a8dc3d0bd2fa4f3b97f18c391195a3ca : SUCCESS in 48m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/23c97871bdf445eb91ce9c09983fd493 : SUCCESS in 39m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f389f26f6616424786a9f6a2ca3de1b1 : SUCCESS in 37m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc0a6c90a9db4f678422642a20c7d3e2 : SUCCESS in 4m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd25458678c34c9f8829fbc23f39848a : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3c04133c75e43679d6bc8e55ceb1ed0 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f8e5f45a68840fe83195e467db69a41 : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f13833d4d0c34f3da75ad9cebebd539e : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a43b408f5b8b4efabbca283719049755 : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99b2a5f9d16f43f0a31776e0bb318814 : SUCCESS in 5m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a696e62427b749f09fcbb660f5cdaab1 : SUCCESS in 33m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d810e74808feb83eaf87ba54ec3db0c07cbef843","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-25 09:11:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5b25cb065dc8996c70045e052eb1d9e9bc009803","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-25 10:20:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/be14320dbdf2487c9901173bde634a5d : SUCCESS in 39m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/48e9c4ee7ec14b179032bbe4af9ad6b8 : SUCCESS in 35m 27s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/6e06032970e64379bf4a6ae387110ec7 : FAILURE in 26m 46s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6600cfb249c34b7a8325dc5e73048446 : SUCCESS in 1h 03m 05s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/762f1421a16f4605a058ff45abd83fa0 : SUCCESS in 36m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f35a741097ec497ba198fadc4b11fd23 : SUCCESS in 38m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c3b7d3bc77f4364b5f6652bd91bce24 : SUCCESS in 5m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1cc648fb772c48a692d8d58b7d377b2d : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/813f55cdec124f0fbfe188530897c7b9 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc473ee30758410b8a40dc282e713c0c : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/75db207c0cd946b7b05db882d028901c : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/148ca1aba1ce4ff983a0a2cb26cbd3a0 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b99c6e03eb446318dfc371b3646b2b5 : SUCCESS in 5m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5928cf1ab1614220bb7147cc6aa5d2a2 : SUCCESS in 35m 01s","accounts_in_message":[],"_revision_number":2},{"id":"f21d326723bce513e97463eba2da786f3f87b7a2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-12-25 23:32:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1086df3e185962ca6b02c2ed98f8e00b0aa388c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-25 23:32:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7a413f06d98adf7daefabe3228cb8bd66f2aada4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-26 00:36:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b2e0e7e2e44c4cc3b1711ded8d13d360 : SUCCESS in 37m 43s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/597e2003220a418fb26a8bd907e1f967 : SUCCESS in 40m 07s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e89eca56644c4d379c582426b05bd72d : SUCCESS in 58m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e525b87722d948e8b57234d151ea0a77 : SUCCESS in 35m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/017172f16a9b46d1bb7acfa406cd6753 : SUCCESS in 41m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab641dc7fde94b08b8b2f1d57758e39c : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ad1e9d5d3db494fb1dbf3d40c0a6f33 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86c034e716ae40aba8d913cd42fe8a89 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a6b70920edbb4175b0d2c910bd904d31 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38715bc4d3af4386a49c91baa71681aa : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d380093481244aada4e8a5d92a1c8ba5 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":2},{"id":"760eb8b351fc754f95090574f9b063406264d106","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-26 00:36:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6e6411243092a036ad618977c44e3f052e544183","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-26 00:37:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0b5f93ca65744aaa92a404f507a5dde : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d18d07a6bf7542fe8e75110b8b85b481 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0050549441a1889a40b85accc4733d3dca8f067f","revisions":{"fa0194123bb8fd8bb78eed67a94646def5e9cc9f":{"kind":"REWORK","_number":1,"created":"2019-12-22 09:04:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/20/700320/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/20/700320/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/20/700320/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/20/700320/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/20/700320/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/20/700320/1"}}},"commit":{"parents":[{"commit":"8fee76623e113f8a76af14b9515b968367c0c1fc","subject":"Don\u0027t use hasattr to lazy-load properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8fee76623e113f8a76af14b9515b968367c0c1fc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-22 09:04:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-22 09:04:24.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I3d8d4021b17b0522ab3a37a337f2a7a1936ca23b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fa0194123bb8fd8bb78eed67a94646def5e9cc9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fa0194123bb8fd8bb78eed67a94646def5e9cc9f"}]},"branch":"refs/heads/master"},"0050549441a1889a40b85accc4733d3dca8f067f":{"kind":"REWORK","_number":2,"created":"2019-12-25 09:11:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/20/700320/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/20/700320/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/20/700320/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/20/700320/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/20/700320/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/20/700320/2"}}},"commit":{"parents":[{"commit":"4d0496ffdcfd830ca749bdec8c9ecdbeadeabf17","subject":"Merge \"Delete removed dashboard panel when upgrading\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4d0496ffdcfd830ca749bdec8c9ecdbeadeabf17"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-25 09:11:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-25 09:11:28.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I3d8d4021b17b0522ab3a37a337f2a7a1936ca23b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0050549441a1889a40b85accc4733d3dca8f067f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0050549441a1889a40b85accc4733d3dca8f067f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
