)]}'
{"id":"openstack%2Fdesignate~704327","triplet_id":"openstack%2Fdesignate~master~I0161bc82878ca936c3b0d9e9f9624be81fad21db","project":"openstack/designate","branch":"master","topic":"fix-ds","hashtags":[],"change_id":"I0161bc82878ca936c3b0d9e9f9624be81fad21db","subject":"Drop use of USE_SYSTEMD var in devstack plugin","status":"MERGED","created":"2020-01-27 14:17:33.000000000","updated":"2020-01-29 23:31:53.000000000","submitted":"2020-01-29 23:30:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704327-1580340601633-641ca554","meta_rev_id":"28ce03decae5092348fc04229269ebd2347e436e","_number":704327,"virtual_id_number":704327,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-29 23:30:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2020-01-29 17:03:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-29 22:12:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-29 22:12:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-29 17:03:30.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2020-01-29 22:12:24.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-01-29 23:30:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6bb4992f937fbc79cba1a326f48921d2f78ab550","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-01-27 14:17:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2cd091164b098bfc86ab793d8703c25ce1891c86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 15:25:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cad768049f8d49a99536fe063983844d : SUCCESS in 36m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ac19db096fd9468ebff76cb385a0a384 : SUCCESS in 35m 06s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/bb643406cccc4dc88c994bfd477824fb : FAILURE in 29m 02s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1cddb9a12d514f45b5226330e06aad29 : SUCCESS in 1h 00m 23s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d634929559bf4c7ab0de4299d0fab883 : SUCCESS in 37m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1a2790ced55f4b378ee8c1dfd0a84d59 : SUCCESS in 43m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e08e8681e624be293f9403cdd3c26c5 : SUCCESS in 6m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b60bca6d0624ad4871649c10edd9134 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32c3c65885de48ea8fcfbe1847d70a9c : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01117b29ba694e2da6f2a65429b989f9 : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/966bdcd067fc4863b9939e7817f66a86 : SUCCESS in 8m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c911d8eb1e54dd99d346f6e5a6d4ec8 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04b44ba2cf524aa1a6f04e7ec71f860b : SUCCESS in 7m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/47057c73ff5540f4810683fef128eb1f : SUCCESS in 37m 06s","accounts_in_message":[],"_revision_number":1},{"id":"58bc7b8237e9ddbbdd624a21764e90a7d8b1ac4c","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2020-01-29 17:03:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5c1533c036990c15077c08b446da419a42b451ca","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-01-29 22:12:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"37330a88c41ecb0d7710bb91b3a1ec30ef295762","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 22:12:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"99d9967d8efcc8d5c50c561cda00a29784afc7e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 23:30:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2bc654533d895251fc6ef3921a40850524eef772","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 23:30:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4b6c2e63c787473e9458e0d0d411f068 : SUCCESS in 42m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/44fee103f4b44662a201dc852e1baee2 : SUCCESS in 39m 53s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f65035369e954e12a0d6eadbf1f0896c : SUCCESS in 1h 10m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b346b30865b54682a6c7326e13416263 : SUCCESS in 44m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ac59d8bfd83241f4b945a30614552dc4 : SUCCESS in 43m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24d06d8a84584c6a808efb4e90d54a4e : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b8e11c095bd4a6cbec86eb35bac626b : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d235b321b2a418ab2611ac71a5e6de1 : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/49ec15cb6b8c48e78571c1f87385d8dd : SUCCESS in 9m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59132f2e87ee4169a7c0372220e146a7 : SUCCESS in 7m 29s","accounts_in_message":[],"_revision_number":1},{"id":"1540c14fa44efad2f6ca6643672f0eef02f69f93","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 23:31:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8be8ffe678b545a99e3af3d528cdf7cb : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d0731c03f5d996717daca149c4b0e3bf0aa8214e","revisions":{"d0731c03f5d996717daca149c4b0e3bf0aa8214e":{"kind":"REWORK","_number":1,"created":"2020-01-27 14:17:33.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/27/704327/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/704327/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/704327/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/704327/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/704327/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/704327/1"}}},"commit":{"parents":[{"commit":"f355aae93955b76df9ea9ef5554b451df0ac91b4","subject":"Merge \"Improve sink recordset creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f355aae93955b76df9ea9ef5554b451df0ac91b4"}]}],"author":{"name":"Dr. Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-01-27 14:15:44.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-01-27 14:17:20.000000000","tz":0},"subject":"Drop use of USE_SYSTEMD var in devstack plugin","message":"Drop use of USE_SYSTEMD var in devstack plugin\n\nThe USE_SYSTEMD variable is to be dropped from devstack[0], as running\nwith systemd is the only option now. Drop it\u0027s usage in our plugin.\n\n[0] https://review.opendev.org/703734\n\nChange-Id: I0161bc82878ca936c3b0d9e9f9624be81fad21db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d0731c03f5d996717daca149c4b0e3bf0aa8214e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d0731c03f5d996717daca149c4b0e3bf0aa8214e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
