)]}'
{"id":"openstack%2Fdesignate~704827","triplet_id":"openstack%2Fdesignate~master~I7265c525f1091012011ba12d65e9669a826ab979","project":"openstack/designate","branch":"master","topic":"fix-for-py3","hashtags":[],"change_id":"I7265c525f1091012011ba12d65e9669a826ab979","subject":"Drop setcap on python binary","status":"MERGED","created":"2020-01-29 15:37:29.000000000","updated":"2020-01-29 23:24:10.000000000","submitted":"2020-01-29 23:22:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704827-1580340157177-159bbb0d","meta_rev_id":"30250cd98037ece906b1fe06781d94959287afa9","_number":704827,"virtual_id_number":704827,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-29 23:22:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2020-01-29 17:03:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-29 22:12:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-29 22:12:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-29 17:03:15.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2020-01-29 22:12:09.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-01-29 23:22:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f9b876271ded705b89bf3b3b6439380b8d0d0110","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-01-29 15:37:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b900411f04f05f70547c2adc70b52b8b996f8ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 16:53:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/64afde57624844b5a8b838ee426f31b6 : SUCCESS in 37m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fb920c266ae043ec9085e407acc0be4e : SUCCESS in 39m 02s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/3a52f29110f04e13b7cc6955b53aa6af : FAILURE in 29m 06s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b264865a72054028a65db76eb566e749 : SUCCESS in 1h 05m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/11a52ad39e9440bc912d4f91b46184db : SUCCESS in 44m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/609df85aa0e64223b62b996f32cb9dd5 : SUCCESS in 44m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5427e1120224e49b5206073e4643ad2 : SUCCESS in 6m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd684f74844a4c2babbfdcc858fc1aa1 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/490b667eb8eb4191bd84529a3db91b38 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef1a257912bd4f9ca7e49a3386df784b : SUCCESS in 7m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dde9e946cb284d2c9efff0ee3c4bc40d : SUCCESS in 8m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c69235667e3f46768d5643191026c0fd : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/077c9a52fac64ea591b30e0fb00298db : SUCCESS in 7m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0b3f01f2db6b4624824687f4ae47f2d2 : SUCCESS in 34m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e23470bf4cd474d514eda6d95da340bdde63991f","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2020-01-29 17:03:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"67d557fabefc5e65dbec9ea18ae600c55a506f4e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-01-29 22:12:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9edca17b8851f85d73455b1062d3ee40e85e8906","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 22:12:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d86db32720273f2e5961a5bdce6a62b74a5a76df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 23:22:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0f15ee0bdb90456dbdabcb0aa088dc71 : SUCCESS in 49m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b482d1a2c0db4cf19d3513a51c78f53b : SUCCESS in 54m 10s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/331189e04f2a47538104018f1095acc8 : SUCCESS in 1h 06m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0fd8cdf93c4f49d9826a9a8df3e44d42 : SUCCESS in 53m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d986b88d1fab4c0ba0411e844fc88c2c : SUCCESS in 43m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab9d3ec438a347e5970bf49cf9140609 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2c47ce8b3f641708eb6b908a402211a : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ef93498458d4781af6c346ba22fcc2f : SUCCESS in 7m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2841fed99aa94f939db7a2d72eb97e54 : SUCCESS in 8m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47cfbf2077d2470686ecd64fb54e2c7f : SUCCESS in 10m 25s","accounts_in_message":[],"_revision_number":1},{"id":"0783f40a425433863c90b3ad898ede28da593713","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 23:22:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"825f7130536976773ea3585253c9201e7425c9a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 23:24:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0857bd78c7804de28e1f672e38580991 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3770b006d0c442461a23008d382bd7bd103584ae","revisions":{"3770b006d0c442461a23008d382bd7bd103584ae":{"kind":"REWORK","_number":1,"created":"2020-01-29 15:37:29.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/27/704827/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/704827/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/704827/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/704827/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/704827/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/704827/1"}}},"commit":{"parents":[{"commit":"f355aae93955b76df9ea9ef5554b451df0ac91b4","subject":"Merge \"Improve sink recordset creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f355aae93955b76df9ea9ef5554b451df0ac91b4"}]}],"author":{"name":"Dr. Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-01-29 15:35:25.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-01-29 15:35:25.000000000","tz":0},"subject":"Drop setcap on python binary","message":"Drop setcap on python binary\n\nThis is a bit of a security issue, also we don\u0027t seem to need it anymore\nsince we have long switched to running under python3 and the setcap\ndoesn\u0027t have any effect anyway. It will also break deployments once we\ndrop installing python2 completely, so just remove it now.\n\nChange-Id: I7265c525f1091012011ba12d65e9669a826ab979\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3770b006d0c442461a23008d382bd7bd103584ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3770b006d0c442461a23008d382bd7bd103584ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
