)]}'
{"id":"openstack%2Fdesignate~706415","triplet_id":"openstack%2Fdesignate~stable%2Fstein~I03b59b3f227dcbe64461650e2491d1bc4f1eef46","project":"openstack/designate","branch":"stable/stein","topic":"bug/1862240","hashtags":[],"change_id":"I03b59b3f227dcbe64461650e2491d1bc4f1eef46","subject":"Use Tempest \u0027all\u0027 tox env","status":"ABANDONED","created":"2020-02-07 00:14:05.000000000","updated":"2020-02-09 18:51:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b9aefa4e53b1958468a921bf7acb59b2d246f528","_number":706415,"virtual_id_number":706415,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"date":"2020-02-09 00:40:55.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-02-07 10:40:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-02-09 08:06:59.000000000","_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"date":"2020-02-09 09:49:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-07 10:40:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-09 08:06:59.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-02-09 09:49:16.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"e4b55f85b0c9c476d137f21472e6b7b3d097eb89","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-07 00:14:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79eeeb08b519bfd7753901b7541dbdae1532fa55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 01:12:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/36ac42ae43fc4fd9b88fd47a006c7912 : SUCCESS in 40m 50s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/7f72babc141b46dfa21677bda76ef536 : SUCCESS in 41m 11s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/897c2a698f22433cbe853e6bbe10c0e7 : SUCCESS in 35m 54s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/8a3549f5709b4f9bab31b7962effc38d : FAILURE in 31m 58s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/55b942a4c80748bd9bbd672faabf0174 : FAILURE in 30m 59s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e59d7f4e12434de4850992088aa8f3d0 : FAILURE in 51m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/248ae271ffea46e1b3e4e624192a4f55 : SUCCESS in 5m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f899751ff5b4b0e996ff35b8472bec7 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/424a664400234eb0bd356635010d4923 : SUCCESS in 6m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eba0d5a233114ba2948d56204987c362 : SUCCESS in 7m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4b41407e1e0944beae6a62a819071a0d : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/297e9a40ca7c4108803bbe5119633389 : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/96af517fe19244fbb96f45254d1539b1 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f7d6a06a3b14d6099316d0f81f3fda2 : SUCCESS in 4m 59s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/f3c0735bc5c24707abfa25a0470aa945 : SUCCESS in 36m 05s","accounts_in_message":[],"_revision_number":1},{"id":"c9465eb46e8651a21662f1a48e7dd06d109193d7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-07 08:59:58.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7ee976fb5b8225f8adc059a608f8201a47b49a06","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-07 09:01:04.000000000","message":"Patch Set 2:\n\nfixed change-id to be consistent","accounts_in_message":[],"_revision_number":2},{"id":"8ca1bbeba2d9f812b228dbdafb99b25ac7be9dab","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-07 09:03:23.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"250fbf0cce347f7e4e16ac5134c41c8be3143767","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-07 09:19:57.000000000","message":"Patch Set 2: -Code-Review\n\ndsvm to fix","accounts_in_message":[],"_revision_number":2},{"id":"6c829f74d049e031bc6b10fc65a9d761bf4e516d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-07 09:27:40.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"9a02b1e82e65ff12ba9b265155afe250fcded305","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-07 09:27:57.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6706a6b37fcb84a78eb6079df0b76df5c88f55f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 10:40:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/479050f54f164f0ca3cf88b13183f028 : SUCCESS in 42m 27s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/0e1d53fc6fc0466194671ff8ae5d9f88 : SUCCESS in 44m 12s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/38972a61bc0242cbaba9fe0c9465edcf : SUCCESS in 45m 07s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/1806c8b001a5462f9fffe1460fc25bb8 : FAILURE in 37m 00s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/813eca6cb94f498ea5ef5a6e220f723e : FAILURE in 31m 54s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/59417e704b9b4c048eb6f047e826d7ec : FAILURE in 59m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d26999fcc63466f80cc90dca48a4f4b : SUCCESS in 8m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68a015adc859414fb6e44f8c38b3fb8d : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11fb61b350044a35b9653acaec05252a : SUCCESS in 7m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2342688a433c4dcca5493f9b4c217b41 : SUCCESS in 6m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9aa9df58f4b84c40aba644f2a7a508c1 : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aed440a849114f9ca209165d3216b199 : SUCCESS in 6m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8936c684f6ec4aeb920e30b2f87b9205 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c28a77b794694e00a7b716a732151da5 : SUCCESS in 7m 02s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/33c310e55f9e4773a16ad56c880fee86 : SUCCESS in 44m 59s","accounts_in_message":[],"_revision_number":3},{"id":"90e63c8e5fc611794b8b24b9bdf12ca0d67bfb12","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-07 13:27:03.000000000","message":"Patch Set 3:\n\nwaiting for below series to merge first and then check grenade issues\n - https://review.opendev.org/#/q/topic:fix-stable-gate+(status:open+OR+status:merged)","accounts_in_message":[],"_revision_number":3},{"id":"8e74f1c18cc66b5efe3a35cb89d8d201ade1ba6e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-09 00:40:51.000000000","message":"Patch Set 3:\n\nstable/rocky fixes are merged now. this should be green too.","accounts_in_message":[],"_revision_number":3},{"id":"6d837ca6c43332da292d325e680176bb417f48c2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-09 00:40:55.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c61823807481c79946d8365174e000938d948033","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 01:33:36.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2ea10d2622874b82bbd36df910df9415 : SUCCESS in 36m 49s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/9b9b243f2f24493e877c08e4cab59d45 : SUCCESS in 39m 14s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/219931880b384f118d42bd79806fd7ee : SUCCESS in 37m 20s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/29016ca02e164655adc25147eefe5bb5 : FAILURE in 29m 28s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/d939c29c6326484a847772550a92b8ad : FAILURE in 29m 25s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9bc468cd9a3d47b8bd30e3bd822d641f : FAILURE in 45m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b1e8d8cc8e84f2ea9a0e1004970f979 : SUCCESS in 5m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d839ca5680d4a048fe0e53a9e6a830c : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fa350d0f1da4b8a8cac5de5579f35d6 : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f373dae4a28e49fbacca06e8db02ba54 : SUCCESS in 5m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/17226543db2e4a7f8887e8c9d86d7cd3 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45e3c5cef26a45fcb0a871e1e560c799 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cfc7015d482043beb46e28b449645af7 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3aac3400305a4b06a0c113022a8578b6 : SUCCESS in 5m 06s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/c58a9d783f3746b992db8e954da122f7 : FAILURE in 50m 06s","accounts_in_message":[],"_revision_number":3},{"id":"d6b3134a9e71b8536ec093e785e7dad4c710c162","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-02-09 08:06:59.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d44670b9fe0f01c5104205719a06a59a3761c45b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 09:04:55.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/147bc302546c40f5ae1ae9b9550b38dd : SUCCESS in 40m 01s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/39f880000da7472cacb38003f4c477f3 : FAILURE in 45m 53s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/096b735d544a48178a6a4a7ad5b2db91 : SUCCESS in 35m 33s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/d26105a4e11048e7928d9365fa4e08af : FAILURE in 28m 36s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/2a03e5ffee8a4527bb38b187ddb870d6 : FAILURE in 26m 08s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/02b4d4928854452da1d3ba8f4b25df13 : FAILURE in 52m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aba29de5ade148c08465ef6466faffdd : SUCCESS in 5m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a32abafa59ab4e86a71418e7ac2dcaaa : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f32163f947444789e14a5781a15413a : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/73e4dc58a6c04a3096ac14d58fd759ac : SUCCESS in 4m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9bc1a0ae70484e37a32aea702ba34eff : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c0d11cec92a440f9c358c1aafe2b771 : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c57d95971a2f4e6f963ce8152ccb6495 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fe176778cbe47df9633d28631d942ec : SUCCESS in 4m 30s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/f23fe526731f45699425ae7ced062218 : SUCCESS in 32m 31s","accounts_in_message":[],"_revision_number":3},{"id":"b2e47273feb2b2d605e7d421a522766444250226","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-09 09:49:16.000000000","message":"Patch Set 3: -Code-Review\n\nhmm, same story here","accounts_in_message":[],"_revision_number":3},{"id":"c847f603421237e06d2dffd5e3a819c4ebd8a965","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-09 18:51:19.000000000","message":"Abandoned\n\nright backport- https://review.opendev.org/#/c/706699/","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8aca87aaf6272fb26c5472acc2d7f38fda909382","revisions":{"cabdf3c3f2ec86b38d6047a1d19080acc372c069":{"kind":"REWORK","_number":1,"created":"2020-02-07 00:14:05.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/15/706415/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/15/706415/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/15/706415/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/15/706415/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/15/706415/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/15/706415/1"}}},"commit":{"parents":[{"commit":"9a2bac0db2f5a3b3d5f39a6747d545e5586a9c31","subject":"Use stable branch scenario job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9a2bac0db2f5a3b3d5f39a6747d545e5586a9c31"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-02-07 00:13:13.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-02-07 00:13:13.000000000","tz":-360},"subject":"Use Tempest \u0027all\u0027 tox env","message":"Use Tempest \u0027all\u0027 tox env\n\nTempest recently moved tox default\nto py3 including all-plugin env.\n-https://review.opendev.org/#/c/703011/\n\nThis will fail the py2 jobs which are using\nall-plugin tox env. Because tox all-plugin\nenv var enable the sitepackages\u003dTrue. Py2 jobs\ninstall the tempest plugins on py2 and all-plugin\ntox try to find the installed plugins on py3 which\ncause no test to run and fail the job.\n\ntox all-plugin has been deprecated since long and not\nsupposed to use for plugins who already in migrated to\nseparate repo in queens cycle.\n\nUsing tox \u0027all\u0027 env to fix the issue.\n\nChange-Id: I03b59b3f227dcbe64461650e2491d1bc4f1eef46\nCloses-Bug: 1862240\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cabdf3c3f2ec86b38d6047a1d19080acc372c069"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cabdf3c3f2ec86b38d6047a1d19080acc372c069"}]},"branch":"refs/heads/stable/stein"},"79ce00c2432082f7220d03d8b06fdc3f72c5383c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-02-07 08:59:58.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/15/706415/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/15/706415/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/15/706415/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/15/706415/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/15/706415/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/15/706415/2"}}},"commit":{"parents":[{"commit":"9a2bac0db2f5a3b3d5f39a6747d545e5586a9c31","subject":"Use stable branch scenario job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9a2bac0db2f5a3b3d5f39a6747d545e5586a9c31"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-02-07 00:13:13.000000000","tz":-360},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-07 08:59:55.000000000","tz":0},"subject":"Use Tempest \u0027all\u0027 tox env","message":"Use Tempest \u0027all\u0027 tox env\n\nTempest recently moved tox default\nto py3 including all-plugin env.\n-https://review.opendev.org/#/c/703011/\n\nThis will fail the py2 jobs which are using\nall-plugin tox env. Because tox all-plugin\nenv var enable the sitepackages\u003dTrue. Py2 jobs\ninstall the tempest plugins on py2 and all-plugin\ntox try to find the installed plugins on py3 which\ncause no test to run and fail the job.\n\ntox all-plugin has been deprecated since long and not\nsupposed to use for plugins who already in migrated to\nseparate repo in queens cycle.\n\nUsing tox \u0027all\u0027 env to fix the issue.\n\nChange-Id: Ib1a2be90c0b3dbacc6daf52f0fc6f326ab47fc98\nCloses-Bug: 1862240\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/79ce00c2432082f7220d03d8b06fdc3f72c5383c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/79ce00c2432082f7220d03d8b06fdc3f72c5383c"}]},"branch":"refs/heads/stable/stein"},"8aca87aaf6272fb26c5472acc2d7f38fda909382":{"kind":"REWORK","_number":3,"created":"2020-02-07 09:27:40.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/15/706415/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/15/706415/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/15/706415/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/15/706415/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/15/706415/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/15/706415/3"}}},"commit":{"parents":[{"commit":"9a2bac0db2f5a3b3d5f39a6747d545e5586a9c31","subject":"Use stable branch scenario job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9a2bac0db2f5a3b3d5f39a6747d545e5586a9c31"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-02-07 00:13:13.000000000","tz":-360},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-07 09:27:37.000000000","tz":0},"subject":"Use Tempest \u0027all\u0027 tox env","message":"Use Tempest \u0027all\u0027 tox env\n\nTempest recently moved tox default\nto py3 including all-plugin env.\n-https://review.opendev.org/#/c/703011/\n\nThis will fail the py2 jobs which are using\nall-plugin tox env. Because tox all-plugin\nenv var enable the sitepackages\u003dTrue. Py2 jobs\ninstall the tempest plugins on py2 and all-plugin\ntox try to find the installed plugins on py3 which\ncause no test to run and fail the job.\n\ntox all-plugin has been deprecated since long and not\nsupposed to use for plugins who already in migrated to\nseparate repo in queens cycle.\n\nUsing tox \u0027all\u0027 env to fix the issue.\n\nChange-Id: Ib1a2be90c0b3dbacc6daf52f0fc6f326ab47fc98\nCloses-Bug: 1862240\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8aca87aaf6272fb26c5472acc2d7f38fda909382"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8aca87aaf6272fb26c5472acc2d7f38fda909382"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
