)]}'
{"id":"openstack%2Fdesignate~706699","triplet_id":"openstack%2Fdesignate~stable%2Fstein~Ib1a2be90c0b3dbacc6daf52f0fc6f326ab47fc98","project":"openstack/designate","branch":"stable/stein","topic":"bug/1862240","hashtags":[],"change_id":"Ib1a2be90c0b3dbacc6daf52f0fc6f326ab47fc98","subject":"Use Tempest \u0027all\u0027 tox env","status":"MERGED","created":"2020-02-09 10:24:08.000000000","updated":"2020-02-10 10:46:43.000000000","submitted":"2020-02-10 10:40:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706699-1581331249859-85323f19","meta_rev_id":"1a30ac11595099933654f2b9058f9db5a93987a2","_number":706699,"virtual_id_number":706699,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-10 10:40:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2020-02-10 08:18:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2020-02-10 01:22:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2020-02-10 08:18:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-09 10:24:08.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-02-10 01:22:06.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-02-10 10:40:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"85faccd5b45a74943821e20b5585136ca30a8c95","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-02-09 10:24:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed9eb873bf6c0fff14377161f4eb2583bb5f2930","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 11:23:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fe6c5737632140ffbe14984e7a876a28 : SUCCESS in 40m 17s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/478f3ded3db54593a340587addc5cf8e : SUCCESS in 39m 24s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/1fb26c1c933546a69d6de9e7cfc57244 : SUCCESS in 37m 07s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/4f2680f51b3442729659881f5f58a074 : FAILURE in 31m 00s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/aa98a2a170a54169b522fcb9ebbe89b9 : FAILURE in 27m 46s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/61c28281ad6c4a55917bf0743c004930 : FAILURE in 53m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5dd72e537d84e94a990e155576f68b6 : SUCCESS in 5m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29e29e2fcf714c0e98a4b527242c61d6 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e2543fa8b5e477d9b189c4c93625737 : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/471da442e7564f26916d67cab7e3286b : SUCCESS in 5m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/581cd18a2edf4134b8ba3b5799adf3c6 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70db8e8ae8574833b555bc964eb62bfc : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/694843f58a5b42ee8dd9634ef9e223a6 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/353199834a9b41129b92d090294ee9ca : SUCCESS in 4m 32s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/9ff36c05109242549e197c2859e35c78 : SUCCESS in 32m 48s","accounts_in_message":[],"_revision_number":1},{"id":"e74212a654748854717ab26e6ae4054e10d53e65","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-09 18:49:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"963bd5ebd093f94d0293fae34bb777d9c9581cc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 19:34:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/139f58b0a3284b7bbea371795334bb8f : SUCCESS in 33m 09s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/e560250c15f84291a1ebca35996e3d3a : SUCCESS in 34m 02s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/c54b13ac0df547ca94df5d5ea2cdaa16 : SUCCESS in 37m 11s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/4d6b341d28ab4631a5258d85ab95462d : SUCCESS in 30m 15s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/d4fb9ab65771438385bf5a65262651ca : SUCCESS in 42m 35s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/673de8bded424850bfa1b11b9c975867 : FAILURE in 36m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afbe0ff8f9d64697889bb519f6a9ab20 : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/955328bafde34021a970b1680a8beeb1 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72751a79c7c943eb90dfc5282237ed64 : SUCCESS in 3m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/870ae34646b54bf49b918b62f61eb6f1 : SUCCESS in 4m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/166a81f7c8724f4880deb60a54d94fa6 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a56460f4e23b4f9d9e6563400911de0e : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a6ddda60e40e4b27adcae014377ef310 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff753e8a8f084268b884a6460b65f844 : SUCCESS in 4m 16s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/e541265465fe43a588ba02aa30743ad3 : SUCCESS in 35m 33s","accounts_in_message":[],"_revision_number":2},{"id":"489d7ad19e79888e36f4f81f2c944c78ca2aac06","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-09 22:20:56.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"a56a5a4e70fb337a186ecb79f2372ec5b59f3313","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 23:23:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/82d3c1b0d53743c59b942e70d9ab953e : SUCCESS in 36m 34s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/32bc60c02d8641eb977ada35a8f3b9b9 : SUCCESS in 38m 29s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/75ca8c90a7b4487bb190855b64faf8e1 : SUCCESS in 38m 18s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/c7ae25169a454ad890bb25e5f553db70 : FAILURE in 28m 49s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/21e3c376a433461ea267f43b0e9ddf2a : FAILURE in 30m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/df028413467f44989b062ef4cfe32f1e : SUCCESS in 57m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3cf702f445c84cb8b2e305e4db003699 : SUCCESS in 5m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9061580ce69a4497bb8ef4b27e7fcd58 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0bd9fedc4f940afb62e9eb3c44a6134 : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f69bf7b651454960a11773ff4e5587f1 : SUCCESS in 4m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2334b378abe24621b035418dec98caa9 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa93f2a8ace34fdbaad50c72e435443f : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ad3de80884054fd3a9b15deb5fb1c68c : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/569a3d3721f34c35973a584552bcce77 : SUCCESS in 4m 22s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/4bf6a631f27047f68667019e1416f844 : SUCCESS in 34m 14s","accounts_in_message":[],"_revision_number":3},{"id":"ff30923df96abf480293ce5fb79382e1eccbf96e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-09 23:59:50.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"acf7549b776571108f0bf5ee479dce4136007e64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 00:59:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8ac301caf7ac4a3bb97aa9ca12b4e281 : SUCCESS in 36m 46s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/59a44d64fb22406996d2f1f9f56f4f06 : SUCCESS in 37m 27s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/bbe2ed411eb5428fa99c72812ba055d3 : SUCCESS in 36m 12s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/3e70de53403449879a2a21cc42016ed0 : SUCCESS in 33m 55s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/882d7cefeea344158434d5a10ca361ab : SUCCESS in 38m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c261dba96b0e490ba4642e46060448f6 : SUCCESS in 58m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9cf133625604e44b59cea941bd43c9a : SUCCESS in 5m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b640853456a24ffab20e9cf8a433f6e1 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f4545e7e09d44c3a3cf039c262f4e2c : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dffcbe73ad944163a8d42e9bed1a2490 : SUCCESS in 4m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f837e510d43f49c3ab6176c9b4fa6c6a : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fb4ba07aecd44f7a36586c2d94009b5 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/019a11dcadff406480a3c898ebaedfe6 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b91f4124d67449bacaf0acb10389dcf : SUCCESS in 4m 55s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/fdece83b132849598852b1350e78e16e : SUCCESS in 33m 11s","accounts_in_message":[],"_revision_number":4},{"id":"2094c7cb1b2f12b5fe159de4ab8f9ba127e7e1a2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-02-10 01:22:06.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"bfe7b5613fc916eb97e17d3469dcf7fe9002a143","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-02-10 08:18:01.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e693b99b1867f6f2740e63555c30a189c9d00322","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 08:18:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"301bfbd9afb417f601a6cccac25fab3572358cb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 10:40:49.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5cc16597df0b4167a02b80d606dc7f4d : SUCCESS in 42m 07s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/3b82db3ee8534a34871d253f99f2affd : SUCCESS in 40m 56s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/0d7210fbaaa34aa586903888e8e76dcf : SUCCESS in 43m 48s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/6504e6b630e645e393ae6d3962c6756b : SUCCESS in 38m 21s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/303b311face44a7c9e3510410cd13924 : SUCCESS in 38m 25s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b3585ca11b7f441084c6809dd80e484c : SUCCESS in 1h 08m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e83cdffdef54eef87e7c590c1b0718a : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47354371d39544ceb62ed2b601f39778 : SUCCESS in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/38f07dd7dad0420c8dcb5a2868e24d16 : SUCCESS in 6m 04s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/74ba75e97c674835a502263150038d9d : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87457bddb58147cb8126dc7ca091ba31 : SUCCESS in 5m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ada9e1b37fae41efb8c48ca9828ab236 : SUCCESS in 9m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/995a5e7ec2274cdfb53544389c668a31 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":4},{"id":"e7c2ed3d58b7baaed679db714c85f006761b357b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 10:40:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"d84fd18d7e1d2423148df0bc5f5132dca99b989e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 10:46:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33bc927843614faebf752e5172d1250e : SUCCESS in 2m 01s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"36086c73df5d1552e531bc0729207bb2bd875ac8","revisions":{"20d9f5947f5a4a4a9ee3389baced1746cf7a13f1":{"kind":"REWORK","_number":1,"created":"2020-02-09 10:24:08.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/99/706699/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/99/706699/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/99/706699/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/99/706699/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/99/706699/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/99/706699/1"}}},"commit":{"parents":[{"commit":"9a2bac0db2f5a3b3d5f39a6747d545e5586a9c31","subject":"Use stable branch scenario job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9a2bac0db2f5a3b3d5f39a6747d545e5586a9c31"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-02-07 00:13:13.000000000","tz":-360},"committer":{"name":"Dr. Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-02-09 10:23:37.000000000","tz":0},"subject":"Use Tempest \u0027all\u0027 tox env","message":"Use Tempest \u0027all\u0027 tox env\n\nTempest recently moved tox default\nto py3 including all-plugin env.\n-https://review.opendev.org/#/c/703011/\n\nThis will fail the py2 jobs which are using\nall-plugin tox env. Because tox all-plugin\nenv var enable the sitepackages\u003dTrue. Py2 jobs\ninstall the tempest plugins on py2 and all-plugin\ntox try to find the installed plugins on py3 which\ncause no test to run and fail the job.\n\ntox all-plugin has been deprecated since long and not\nsupposed to use for plugins who already in migrated to\nseparate repo in queens cycle.\n\nUsing tox \u0027all\u0027 env to fix the issue.\n\nChange-Id: Ib1a2be90c0b3dbacc6daf52f0fc6f326ab47fc98\nCloses-Bug: 1862240\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/20d9f5947f5a4a4a9ee3389baced1746cf7a13f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/20d9f5947f5a4a4a9ee3389baced1746cf7a13f1"}]},"branch":"refs/heads/stable/stein"},"33ab6e9aeef05acdbf4b284ac8261054cc41b4c7":{"kind":"REWORK","_number":2,"created":"2020-02-09 18:49:43.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/99/706699/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/99/706699/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/99/706699/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/99/706699/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/99/706699/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/99/706699/2"}}},"commit":{"parents":[{"commit":"9a2bac0db2f5a3b3d5f39a6747d545e5586a9c31","subject":"Use stable branch scenario job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9a2bac0db2f5a3b3d5f39a6747d545e5586a9c31"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-02-07 00:13:13.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-09 18:48:58.000000000","tz":-360},"subject":"Use Tempest \u0027all\u0027 tox env","message":"Use Tempest \u0027all\u0027 tox env\n\nTempest recently moved tox default\nto py3 including all-plugin env.\n-https://review.opendev.org/#/c/703011/\n\nThis will fail the py2 jobs which are using\nall-plugin tox env. Because tox all-plugin\nenv var enable the sitepackages\u003dTrue. Py2 jobs\ninstall the tempest plugins on py2 and all-plugin\ntox try to find the installed plugins on py3 which\ncause no test to run and fail the job.\n\ntox all-plugin has been deprecated since long and not\nsupposed to use for plugins who already in migrated to\nseparate repo in queens cycle.\n\nEnable designate-tempest-plugin and using tox \u0027all\u0027 env\nto fix the issue.\n\nChange-Id: Ib1a2be90c0b3dbacc6daf52f0fc6f326ab47fc98\nCloses-Bug: 1862240\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/33ab6e9aeef05acdbf4b284ac8261054cc41b4c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/33ab6e9aeef05acdbf4b284ac8261054cc41b4c7"}]},"branch":"refs/heads/stable/stein"},"f4810e87f6eec240b6554b6fb57f18423eb3ea9c":{"kind":"REWORK","_number":3,"created":"2020-02-09 22:20:56.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/99/706699/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/99/706699/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/99/706699/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/99/706699/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/99/706699/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/99/706699/3"}}},"commit":{"parents":[{"commit":"9a2bac0db2f5a3b3d5f39a6747d545e5586a9c31","subject":"Use stable branch scenario job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9a2bac0db2f5a3b3d5f39a6747d545e5586a9c31"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-02-07 00:13:13.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-09 22:20:54.000000000","tz":0},"subject":"Use Tempest \u0027all\u0027 tox env","message":"Use Tempest \u0027all\u0027 tox env\n\nTempest recently moved tox default\nto py3 including all-plugin env.\n-https://review.opendev.org/#/c/703011/\n\nThis will fail the py2 jobs which are using\nall-plugin tox env. Because tox all-plugin\nenv var enable the sitepackages\u003dTrue. Py2 jobs\ninstall the tempest plugins on py2 and all-plugin\ntox try to find the installed plugins on py3 which\ncause no test to run and fail the job.\n\ntox all-plugin has been deprecated since long and not\nsupposed to use for plugins who already in migrated to\nseparate repo in queens cycle.\n\nEnable designate-tempest-plugin and using tox \u0027all\u0027 env\nto fix the issue.\n\nChange-Id: Ib1a2be90c0b3dbacc6daf52f0fc6f326ab47fc98\nCloses-Bug: 1862240\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f4810e87f6eec240b6554b6fb57f18423eb3ea9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f4810e87f6eec240b6554b6fb57f18423eb3ea9c"}]},"branch":"refs/heads/stable/stein"},"36086c73df5d1552e531bc0729207bb2bd875ac8":{"kind":"REWORK","_number":4,"created":"2020-02-09 23:59:50.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/99/706699/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/99/706699/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/99/706699/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/99/706699/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/99/706699/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/99/706699/4"}}},"commit":{"parents":[{"commit":"9a2bac0db2f5a3b3d5f39a6747d545e5586a9c31","subject":"Use stable branch scenario job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9a2bac0db2f5a3b3d5f39a6747d545e5586a9c31"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-02-07 00:13:13.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-09 23:59:48.000000000","tz":0},"subject":"Use Tempest \u0027all\u0027 tox env","message":"Use Tempest \u0027all\u0027 tox env\n\nTempest recently moved tox default\nto py3 including all-plugin env.\n-https://review.opendev.org/#/c/703011/\n\nThis will fail the py2 jobs which are using\nall-plugin tox env. Because tox all-plugin\nenv var enable the sitepackages\u003dTrue. Py2 jobs\ninstall the tempest plugins on py2 and all-plugin\ntox try to find the installed plugins on py3 which\ncause no test to run and fail the job.\n\ntox all-plugin has been deprecated since long and not\nsupposed to use for plugins who already in migrated to\nseparate repo in queens cycle.\n\nEnable designate-tempest-plugin and using tox \u0027all\u0027 env\nto fix the issue.\n\nChange-Id: Ib1a2be90c0b3dbacc6daf52f0fc6f326ab47fc98\nCloses-Bug: 1862240\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/36086c73df5d1552e531bc0729207bb2bd875ac8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/36086c73df5d1552e531bc0729207bb2bd875ac8"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
