)]}'
{"id":"openstack%2Fdesignate~706700","triplet_id":"openstack%2Fdesignate~stable%2Frocky~Ib1a2be90c0b3dbacc6daf52f0fc6f326ab47fc98","project":"openstack/designate","branch":"stable/rocky","topic":"bug/1862240","hashtags":[],"change_id":"Ib1a2be90c0b3dbacc6daf52f0fc6f326ab47fc98","subject":"Use Tempest \u0027all\u0027 tox env","status":"MERGED","created":"2020-02-09 10:25:16.000000000","updated":"2020-02-12 12:01:33.000000000","submitted":"2020-02-12 12:01:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706700-1581508893154-4f5f8f67","meta_rev_id":"b8d9aab48e5840b93aa4ea1c00332876808041f1","_number":706700,"virtual_id_number":706700,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"date":"2020-02-10 01:11:11.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-12 12:01:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2020-02-10 08:14:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-10 07:56:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2020-02-10 08:14:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-10 01:11:11.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-02-10 07:56:37.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-02-12 12:01:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"14999fa6d11f59425daf60f195d757e6e4c48bc9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-02-09 10:25:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e047ca659f719d4c7ad2964c8a8b85b76184357","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 11:24:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/361c5321a32d450e9b7b550ff9f8500e : SUCCESS in 38m 35s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/d79d9a2ca83845a197b3de5bc6468a92 : SUCCESS in 40m 34s\n- designate-bind9-py35 https://zuul.opendev.org/t/openstack/build/cc5b49332c074d9b8340dd47da98233b : SUCCESS in 34m 07s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/5721af44477246888c813676d625d509 : SUCCESS in 38m 54s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/9c4e1ce0908e41a19312db7149f47d61 : FAILURE in 31m 02s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/a58816c7f4fa477a96ee65644dda89aa : FAILURE in 29m 08s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d3b5e910acf8402ab457d901c019f676 : FAILURE in 45m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61991e39176f4666ab6f446eb783c703 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17e9be708a9240328a3d5b74f1905aa9 : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f8636983e4ca4bab9d959114cfab842f : SUCCESS in 5m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8a7d525a2b674c5ba9d512bec94fb958 : SUCCESS in 4m 58s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/31a44330fded483b9fe58c3a02060326 : SUCCESS in 5m 07s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/d247828d005e49f58fc0fec779099a50 : SUCCESS in 56m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87f516b017114290a1321916f554d3e4 : SUCCESS in 4m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21e1711345fb44309346e9ab2ecb7db4 : SUCCESS in 6m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b2c8af47e598bec57c926132202d6b1a56df4c70","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-09 18:50:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d3d81f559d3874f3b0a18fac76823908abef8c1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 19:34:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/55fc39715115406798030754e3d47f39 : SUCCESS in 37m 47s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/b4ef15dcae6e44cebfd0cd5082ee7b64 : SUCCESS in 41m 29s\n- designate-bind9-py35 https://zuul.opendev.org/t/openstack/build/9304e4828b1c41728551bac606fa744c : SUCCESS in 38m 32s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/acaffe795f5b433798502be8479f19a5 : SUCCESS in 37m 13s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/fca7b59721d1416e8d5ad41bf304c2ff : SUCCESS in 34m 02s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/4d57441d3f7b4be19094ff70312f407b : SUCCESS in 33m 44s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d7c6ac0691144d849b4b01ec759049c7 : FAILURE in 34m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ef783d011ec4d2a9a07de32a3dca1d1 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0e81645ac1b4d848dc7049d349795c1 : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a11c00ffc6a64c138a4e57bffd6b4745 : SUCCESS in 4m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e475eb9b22f644b38b228352dee3f8ac : SUCCESS in 4m 55s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/80fa20feca2e408d90e3c3dc225e6c4d : SUCCESS in 4m 51s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/53197fc0d3fe4462a574907880f875b2 : SUCCESS in 32m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28f3975f6ad54e349ad347fbaee7cacc : SUCCESS in 3m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/045a55af3c4446e2bb07ff94f086b15f : SUCCESS in 5m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c4a4fca2ede95ab0809efb5248c754186cbe741c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-09 22:21:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e3c4dd11b2233eb1b98d321daba78d71479c7b61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 23:22:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c38372b5f63a4a81b3e30cc145665278 : SUCCESS in 39m 33s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/4dac8960754443f886e8c4a7338fb20f : SUCCESS in 38m 40s\n- designate-bind9-py35 https://zuul.opendev.org/t/openstack/build/aaedb30e74cb4388a85aa72efc632fd6 : SUCCESS in 40m 27s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/4f1fd1a4b7c349c6914d235d54ac34aa : SUCCESS in 33m 02s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/4c13381898954d6898f473cc377afc49 : FAILURE in 29m 43s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/0da33a1b4112468faa6996d96e146ad9 : FAILURE in 27m 00s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6b3bc38e7df241fc8013f4ad455b542d : SUCCESS in 56m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/172d989a0da4427c9489d37c6eb2b299 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11dd855e95d44f9c8b70e89e6c62cb0f : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bd8948464bba4fef8da9639002f6ad1f : SUCCESS in 5m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4f4eed4b8e984ba5bb9c5bb562cb6d1b : SUCCESS in 4m 28s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4bcae10f9ae44f32aeb2c208e2163455 : SUCCESS in 5m 25s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/ef869007e6904cc993fee43a0d973afe : SUCCESS in 33m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4db4d0bbfbff4ee1a0fa0165f8ba6703 : SUCCESS in 4m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b38373a2d4e24c3ea6272cc385822ad0 : SUCCESS in 5m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2868e36be3cd88c11f023ed9c09e5e26e5d32d9e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-09 23:59:20.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"28542a077c42c8e3beb017f2fd8587330b501668","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 01:08:58.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0c286d74290e454fbc1443c51d30035e : FAILURE in 45m 21s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/cc372ab36a034cec9fe22a748dfc56d9 : SUCCESS in 38m 53s\n- designate-bind9-py35 https://zuul.opendev.org/t/openstack/build/785ec82c44ff4955967d5acbf83fba11 : SUCCESS in 50m 20s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/3fd89a8cf8d44ac192dde5b546301de3 : SUCCESS in 33m 11s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/34e4949727304dd7a6fb3a740888580e : SUCCESS in 33m 27s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/4537b787e4b949a08ce8818a3985b838 : SUCCESS in 33m 57s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c5a0610bd2da4d7eb7252bd8b841775a : SUCCESS in 1h 03m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a025b4b994c1415dabbdc1fb0bd79fed : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1500ba2e8f09457ab3e27e499c8a2e0e : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/66a53d5b19004ba3817b80a2e60256cd : SUCCESS in 4m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/84ec73626f7647f5a493192fd0e282fe : SUCCESS in 5m 36s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/3a01bce1a7e44501b85ccf402bd0600c : SUCCESS in 4m 14s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/10f99996a33843b5a16d539a962f265c : SUCCESS in 33m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c01c3f16e2c47409327b8a3560cab54 : SUCCESS in 3m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2449da896dc54eb9b64bda1a38c8c390 : SUCCESS in 10m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d5ab98fa0444baa33fb213942d318703823f45fe","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-10 01:11:11.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"da7e8f9d7a60738344e782cc498faf29f0c8960a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 02:13:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a621d53d1eea42ebaed44e13064794a5 : SUCCESS in 33m 47s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/6f085661898c4f20a6bc77871598034e : SUCCESS in 33m 01s\n- designate-bind9-py35 https://zuul.opendev.org/t/openstack/build/b32c8cb04a0e4f91a06fc3dd37e850f5 : SUCCESS in 40m 53s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/1cc5aadf20164d2eb8a8083b55a249d5 : SUCCESS in 37m 58s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/dab35f0e732f4b67b05a333dc6489656 : SUCCESS in 34m 34s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/270f250e94fc497f8b1aba9decec6a83 : SUCCESS in 35m 23s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8c3a2878158949b5a0579c702d813029 : SUCCESS in 57m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01f855825f104f1d92ddfd10241de5b3 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/428d4a9563db4152bb2bb52090b72790 : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/978dd8d4a27e496b8607ea8a5c9d8d71 : SUCCESS in 4m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6e4d22f5e7944ebbb72db71836683891 : SUCCESS in 4m 56s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e7c12976da4045e09898bfa8f316008c : SUCCESS in 5m 12s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/146851981eeb4837a26950a8c8661413 : SUCCESS in 33m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7fcfe42b4d3b47afbda49bbbcc5f36b9 : SUCCESS in 4m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abde93821795458cbb7c531873c9bf30 : SUCCESS in 5m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"312da0c675a66203d6748618f6005a4c8c0299ef","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-02-10 07:56:37.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"fb108c222bb3bba7c693c57bffc481b717ec2cbb","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-02-10 08:14:21.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f4a920aed4975d5d07ba7d4d4440f36a4c8c082d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 08:14:36.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a63640c786af5c03d7b5d48fc99fcf2e3a98724f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 09:20:32.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8ceb06422b554c96893ee9fc694f5466 : SUCCESS in 51m 54s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/6d6ecb35a09d47178cc5c69b993c9e3f : SUCCESS in 46m 09s\n- designate-bind9-py35 https://zuul.opendev.org/t/openstack/build/9240f9a9c9c94d2ca911fe8e83b6ed06 : SUCCESS in 42m 49s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/c4773f36878d4349826a6a41626b6fda : SUCCESS in 48m 36s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/76e96b611676429a92d3f5838cdd65d5 : SUCCESS in 37m 28s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/d0efd3c29460452c83f0312cd92607de : FAILURE in 55m 51s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3c506200ffd54bce9f1f04dc02da8ddc : SUCCESS in 1h 04m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d08b6c841ed436181f9d34df5a6bd53 : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfea792479a342b4be9eea4287c81e3c : SUCCESS in 6m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d0be5b4a49584be284cccbe6f4d0b90e : SUCCESS in 7m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b6e5082ef0cb4e49acc2af59cbd20669 : SUCCESS in 8m 54s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/712790ac2c9b431a9d0fe4a4c4b14d7c : SUCCESS in 7m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81b4a2aeb1584cb6bbb3fdf259a520cd : SUCCESS in 6m 47s","accounts_in_message":[],"_revision_number":4},{"id":"9a15c40e980ae0fb7fd9ed1a3d7338b26e53f435","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-02-10 18:05:11.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"0fd947badda8a9e967b9accefd474cd11a325bb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 19:09:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/925d8cf0848744a68ac9f2a21d90404b : SUCCESS in 48m 32s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/bdec273836e74410bc3b696d530a0309 : SUCCESS in 56m 04s\n- designate-bind9-py35 https://zuul.opendev.org/t/openstack/build/e24a4526baf842e0bb1b804bfb30b569 : SUCCESS in 41m 54s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/a1fa138bfd094cd484ea94dfd6c80492 : SUCCESS in 50m 45s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/8ca675cba01c4d269b8365b34f8aef8a : SUCCESS in 45m 15s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/806be859695e4a37a0eb848c1b0db1d3 : SUCCESS in 51m 34s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e396c3cd4c1c47b49f2196d86b81280e : FAILURE in 57m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/554095c63f0541c392ecb85c778e1599 : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfc3eebe400f40a9af391e125d7806f4 : SUCCESS in 8m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d106e131e41b493192d2a98f99b378a8 : SUCCESS in 8m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9d10069fc85b44679583ac74aaca581f : SUCCESS in 8m 26s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/08315dc63d0c49cc83d0faf46665a963 : SUCCESS in 8m 00s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/50330c56036540f69579695b84fa7323 : SUCCESS in 44m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06259a411d114ffab427157229898b8a : SUCCESS in 7m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32928315799b46ca9f212126442e7d77 : SUCCESS in 9m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"077959f87ec405e02ee6a995fb32244874f83fed","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-02-11 09:34:25.000000000","message":"Patch Set 4:\n\nrecheck yet another sporadic single test failure","accounts_in_message":[],"_revision_number":4},{"id":"0498d03cd355418c8e63857191cf61fbff1766df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 10:24:25.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f0f0f127a1c6410b8258c6d976265cc0 : FAILURE in 10m 50s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/e5b76443cb1b41e592a28b8903ec4d6e : FAILURE in 14m 08s\n- designate-bind9-py35 https://zuul.opendev.org/t/openstack/build/91969a842321405db374ba66bc3a27a7 : SUCCESS in 41m 37s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/1a567541d593472ebed861e7b4055f84 : SUCCESS in 43m 51s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/6c17f3d0c8914b06bccca4f2c5562304 : FAILURE in 13m 20s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/6d450aca21af4faabc4e753e5092456e : FAILURE in 12m 40s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/51c591e58ac34f73967af0031f799d79 : FAILURE in 15m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8c3159afe6f44498fc9ec89364b666c : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0816bd09e76a43a5a610f21ccc71ceb5 : SUCCESS in 7m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4091f897c97b46c7909614177bfccba0 : SUCCESS in 7m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/24caa05469f34e5db877e34fd34169df : SUCCESS in 6m 37s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b305991b67414be5aecb18885896484b : SUCCESS in 6m 10s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/e66d15601f0a42f494de6c9de4ee9fb7 : FAILURE in 11m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0c384ad2015492db9d51fece4a75702 : SUCCESS in 5m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/334530e3ce5e4447b5363360e3dcd270 : SUCCESS in 7m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"da986429fdb98a7162dfcee03fccd768d47d8df6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-02-11 10:47:39.000000000","message":"Patch Set 4:\n\nrecheck holding a node for virtualenv debugging","accounts_in_message":[],"_revision_number":4},{"id":"abf1734acc3189bf869419d0d929e233c3f30546","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 11:34:47.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3c44d87d870c4bf7be46efd715f1498e : FAILURE in 11m 29s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/a0d8d245346545ab90627e2b9adbbfd2 : FAILURE in 12m 16s\n- designate-bind9-py35 https://zuul.opendev.org/t/openstack/build/f4795269ea134eb8aeaf4f16f16cd387 : SUCCESS in 42m 06s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/3089e36759dc42b3bc071e346f8215ca : SUCCESS in 37m 49s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/c05a2fab11e54ffe96d234bb6d5a5780 : FAILURE in 13m 25s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/6aec52ede9b04e99b347c70bd45cb963 : FAILURE in 12m 27s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a4162807876a42498974924b1b975336 : FAILURE in 12m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6080d25bb5634bd1af2af0333a1d9447 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f9155cd8a02453ca6d8dc57a3102139 : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/547266114f7d4838b2a9a721ce4ede1b : SUCCESS in 5m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9c6e6875907b4017ab24a9d028a30d96 : SUCCESS in 4m 58s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a6b511576c53475f85f98b7f72e48255 : SUCCESS in 5m 01s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/e5351a5224374e84aaa81ad40e4f8c5f : FAILURE in 7m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/325899b38ecd4d4c9ec55651b5b4405a : SUCCESS in 4m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/251bf65f5d80474d92037087d222c3da : SUCCESS in 7m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"db5a201a9148a7d4862521682ba216480660ca40","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-02-12 09:47:42.000000000","message":"Patch Set 4:\n\nrecheck hopefully devstack got fixed once more","accounts_in_message":[],"_revision_number":4},{"id":"ed1311dda8e0458a2a628b58e38d3ecc71bfe7e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 10:52:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3284aef81a3241bf906488b6f2e6f6b5 : SUCCESS in 38m 45s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/c8d4ffe65e6245808be9e63f34e41e31 : SUCCESS in 44m 03s\n- designate-bind9-py35 https://zuul.opendev.org/t/openstack/build/1823649c40774e8ea7b7d394b112a0d5 : SUCCESS in 47m 08s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/1fc26e3951ed4bb9a4d51f1f8c20f6d5 : SUCCESS in 43m 31s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/06030f263eef4dc88826feec4dacea15 : SUCCESS in 39m 25s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/bded9b472160413bb46b856e60c4df92 : SUCCESS in 39m 43s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e2c6d2134faf43b78c4905280ef2d5c5 : SUCCESS in 1h 02m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac19f2e4041f49b89250bb9f4f7c6a07 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b79dd3a4ad8413cb440dda7a3c4016e : SUCCESS in 7m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/69f2fd851fea46319f7ded1333ac8b81 : SUCCESS in 6m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/46b3799bd63b43b6afc1cf0342c8176f : SUCCESS in 7m 40s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d0be9ef8fe4248c69c2b09327af5b36e : SUCCESS in 7m 28s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/88fc203386034b37978195cb8832e422 : SUCCESS in 35m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2292c30171eb42bda6cb357c3eacde00 : SUCCESS in 5m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/514b3107833e4e5db1e014a45832be80 : SUCCESS in 6m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4fb0aaebde9ac5c0c72037690bb3534388d7cbbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 10:52:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9a697cb74905fd4ba40a86b3f94c7352b6248f60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 12:01:32.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f2b56ee8619d4a40a0309341980b6987 : SUCCESS in 41m 45s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/190a2b42e4414ba08baf8522853cad21 : SUCCESS in 45m 28s\n- designate-bind9-py35 https://zuul.opendev.org/t/openstack/build/f669a70533a749ce8b15c6a5b50fb930 : SUCCESS in 45m 18s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/64c832f8e40c4daeb28e38c9900e17aa : SUCCESS in 47m 20s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/016d83b76d9f4fe3b38f6457898fa300 : SUCCESS in 36m 29s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/f6edf203887f4c6db0602da1059bd77a : SUCCESS in 37m 46s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/73a85dccd86a425082e5546cb2d8deef : SUCCESS in 1h 04m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89e9f26c1f3d45f1993e9d3541ebd77c : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce09e4e5cfb44222b3fdc2ddbd42a9cb : SUCCESS in 6m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/52496a45b89f463da1d1c84b2e6aac36 : SUCCESS in 6m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a646dd73fd874155a83db30bdcdb7deb : SUCCESS in 7m 39s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b366a0e314694d95a5d69457aa460eff : SUCCESS in 5m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e543c2e3cb940d3b7b7f68e6ffbe6cc : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":4},{"id":"be91571deb37e41ac42da63260ea53503e5e69fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 12:01:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0566f104ccf1b30a7f95d05fe20a238d01888dc5","revisions":{"2470462e22310c5f0bb2e5d291ddd625bb05bd1e":{"kind":"REWORK","_number":1,"created":"2020-02-09 10:25:16.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/00/706700/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/00/706700/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/00/706700/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/00/706700/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/00/706700/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/00/706700/1"}}},"commit":{"parents":[{"commit":"1e20d09272b4dbb082e72a663e931a61c81c6b37","subject":"Fixing 404 link","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1e20d09272b4dbb082e72a663e931a61c81c6b37"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-02-07 00:13:13.000000000","tz":-360},"committer":{"name":"Dr. Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-02-09 10:24:52.000000000","tz":0},"subject":"Use Tempest \u0027all\u0027 tox env","message":"Use Tempest \u0027all\u0027 tox env\n\nTempest recently moved tox default\nto py3 including all-plugin env.\n-https://review.opendev.org/#/c/703011/\n\nThis will fail the py2 jobs which are using\nall-plugin tox env. Because tox all-plugin\nenv var enable the sitepackages\u003dTrue. Py2 jobs\ninstall the tempest plugins on py2 and all-plugin\ntox try to find the installed plugins on py3 which\ncause no test to run and fail the job.\n\ntox all-plugin has been deprecated since long and not\nsupposed to use for plugins who already in migrated to\nseparate repo in queens cycle.\n\nUsing tox \u0027all\u0027 env to fix the issue.\n\nChange-Id: Ib1a2be90c0b3dbacc6daf52f0fc6f326ab47fc98\nCloses-Bug: 1862240\n(cherry picked from commit 20d9f5947f5a4a4a9ee3389baced1746cf7a13f1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2470462e22310c5f0bb2e5d291ddd625bb05bd1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2470462e22310c5f0bb2e5d291ddd625bb05bd1e"}]},"branch":"refs/heads/stable/rocky"},"bf1aa198ede8740d887147dd7feb3efb47a47149":{"kind":"REWORK","_number":2,"created":"2020-02-09 18:50:47.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/00/706700/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/00/706700/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/00/706700/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/00/706700/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/00/706700/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/00/706700/2"}}},"commit":{"parents":[{"commit":"1e20d09272b4dbb082e72a663e931a61c81c6b37","subject":"Fixing 404 link","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1e20d09272b4dbb082e72a663e931a61c81c6b37"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-02-07 00:13:13.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-09 18:50:47.000000000","tz":0},"subject":"Use Tempest \u0027all\u0027 tox env","message":"Use Tempest \u0027all\u0027 tox env\n\nTempest recently moved tox default\nto py3 including all-plugin env.\n-https://review.opendev.org/#/c/703011/\n\nThis will fail the py2 jobs which are using\nall-plugin tox env. Because tox all-plugin\nenv var enable the sitepackages\u003dTrue. Py2 jobs\ninstall the tempest plugins on py2 and all-plugin\ntox try to find the installed plugins on py3 which\ncause no test to run and fail the job.\n\ntox all-plugin has been deprecated since long and not\nsupposed to use for plugins who already in migrated to\nseparate repo in queens cycle.\n\nEnable designate-tempest-plugin and using tox \u0027all\u0027 env\nto fix the issue.\n\nChange-Id: Ib1a2be90c0b3dbacc6daf52f0fc6f326ab47fc98\nCloses-Bug: 1862240\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bf1aa198ede8740d887147dd7feb3efb47a47149"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bf1aa198ede8740d887147dd7feb3efb47a47149"}]},"branch":"refs/heads/stable/rocky"},"81d2500ada527f6c13446b40a59ac0ab6c3d81d1":{"kind":"REWORK","_number":3,"created":"2020-02-09 22:21:04.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/00/706700/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/00/706700/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/00/706700/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/00/706700/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/00/706700/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/00/706700/3"}}},"commit":{"parents":[{"commit":"1e20d09272b4dbb082e72a663e931a61c81c6b37","subject":"Fixing 404 link","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1e20d09272b4dbb082e72a663e931a61c81c6b37"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-02-07 00:13:13.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-09 22:21:04.000000000","tz":0},"subject":"Use Tempest \u0027all\u0027 tox env","message":"Use Tempest \u0027all\u0027 tox env\n\nTempest recently moved tox default\nto py3 including all-plugin env.\n-https://review.opendev.org/#/c/703011/\n\nThis will fail the py2 jobs which are using\nall-plugin tox env. Because tox all-plugin\nenv var enable the sitepackages\u003dTrue. Py2 jobs\ninstall the tempest plugins on py2 and all-plugin\ntox try to find the installed plugins on py3 which\ncause no test to run and fail the job.\n\ntox all-plugin has been deprecated since long and not\nsupposed to use for plugins who already in migrated to\nseparate repo in queens cycle.\n\nEnable designate-tempest-plugin and using tox \u0027all\u0027 env\nto fix the issue.\n\nChange-Id: Ib1a2be90c0b3dbacc6daf52f0fc6f326ab47fc98\nCloses-Bug: 1862240\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/81d2500ada527f6c13446b40a59ac0ab6c3d81d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/81d2500ada527f6c13446b40a59ac0ab6c3d81d1"}]},"branch":"refs/heads/stable/rocky"},"0566f104ccf1b30a7f95d05fe20a238d01888dc5":{"kind":"REWORK","_number":4,"created":"2020-02-09 23:59:20.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/00/706700/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/00/706700/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/00/706700/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/00/706700/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/00/706700/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/00/706700/4"}}},"commit":{"parents":[{"commit":"1e20d09272b4dbb082e72a663e931a61c81c6b37","subject":"Fixing 404 link","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1e20d09272b4dbb082e72a663e931a61c81c6b37"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-02-07 00:13:13.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-09 23:59:13.000000000","tz":0},"subject":"Use Tempest \u0027all\u0027 tox env","message":"Use Tempest \u0027all\u0027 tox env\n\nTempest recently moved tox default\nto py3 including all-plugin env.\n-https://review.opendev.org/#/c/703011/\n\nThis will fail the py2 jobs which are using\nall-plugin tox env. Because tox all-plugin\nenv var enable the sitepackages\u003dTrue. Py2 jobs\ninstall the tempest plugins on py2 and all-plugin\ntox try to find the installed plugins on py3 which\ncause no test to run and fail the job.\n\ntox all-plugin has been deprecated since long and not\nsupposed to use for plugins who already in migrated to\nseparate repo in queens cycle.\n\nEnable designate-tempest-plugin and using tox \u0027all\u0027 env\nto fix the issue.\n\nChange-Id: Ib1a2be90c0b3dbacc6daf52f0fc6f326ab47fc98\nCloses-Bug: 1862240\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0566f104ccf1b30a7f95d05fe20a238d01888dc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0566f104ccf1b30a7f95d05fe20a238d01888dc5"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
