)]}'
{"id":"openstack%2Fdesignate~707451","triplet_id":"openstack%2Fdesignate~master~I4f6c6c1c96251bfc4f7ad3bb6b0bd2025397fa03","project":"openstack/designate","branch":"master","topic":"691523","hashtags":[],"change_id":"I4f6c6c1c96251bfc4f7ad3bb6b0bd2025397fa03","subject":"Update the Akamai DNS Driver matrix information","status":"MERGED","created":"2020-02-12 18:49:43.000000000","updated":"2020-02-14 03:00:39.000000000","submitted":"2020-02-14 02:59:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707451-1581649154982-4edda447","meta_rev_id":"5fb98562482a1de427b35caebcf1f797f047c451","_number":707451,"virtual_id_number":707451,"owner":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-14 02:59:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2020-02-13 18:55:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-14 02:46:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-14 02:45:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-13 18:55:49.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-02-14 02:45:56.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-02-14 02:59:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"161624f3c2eec703c885e324a8abd3742983769f","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2020-02-12 18:49:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d616a02aef73ee1cf2289228f9617686d64094b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 19:22:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05b9cf221c7144a3bd7e05c46b4b5a33 : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9daa1e1586f4c6c917d32c09922db71 : FAILURE in 8m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d31e263e042412685e365176714d1e1 : SUCCESS in 10m 42s","accounts_in_message":[],"_revision_number":1},{"id":"9c2f1f7e5a1ce4e9e17662fd2f435764d2439cb9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-02-13 06:12:52.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5abf4150f534d3e42f62d5d92f1bb6542e696ba3","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2020-02-13 15:12:06.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7821da5c95ee226771ea6b12228e94bdcec10222","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2020-02-13 15:12:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ee65aec42e68dbccdea57aeff878f9b1a23c3f91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 15:24:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a023a8e634a4e8d902a0a8a17c961ca : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/021591c9c84745feb08ab79a6676c08c : SUCCESS in 11m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47212f31141e407cbc36c83169589c3d : SUCCESS in 8m 07s","accounts_in_message":[],"_revision_number":2},{"id":"965e3a2960680bdebcf3d75620a3792c29270007","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-02-13 18:55:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fd8d69d6c2763993dd9cebaf22d97f7db0f5281e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-02-14 02:45:56.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ffbe3e8bb512fcd92dcf88162f2fe6b9ca6d148f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-02-14 02:46:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8c153a36ea9d68d6677bb82f4b691386d6101ad3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 02:46:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d4a784dcc5169646ac5e965155ad14f074db0308","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 02:59:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9262f28abe64b3785700c5b8f0f145c : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3b48b2d2a7347618c4cfb0349335841 : SUCCESS in 8m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af7c94b95e4147f78e68f90c998e5ec8 : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":2},{"id":"ddf1fc17d52d9b58e99f881545d7f8a134429b31","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 02:59:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b74b1d7a5c84532a4b5bab261061e1a31eba2f35","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 03:00:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/042b67ba00484d88b6f14dfdc73dd07d : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee17471f2fe944e7a3a1b6a836b029e1 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ce7a4a8886422d56f96785835f170d84fa9333bf","revisions":{"cf1a9d8cbc65d8043bd974e1ed6522782bef6ca7":{"kind":"REWORK","_number":1,"created":"2020-02-12 18:49:43.000000000","uploader":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"ref":"refs/changes/51/707451/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/51/707451/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/707451/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/707451/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/707451/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/51/707451/1"}}},"commit":{"parents":[{"commit":"318b8d03193d3d9c26842528e9eaf25814df3ba9","subject":"Implement create/delete zone for Akamai v2 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/318b8d03193d3d9c26842528e9eaf25814df3ba9"}]}],"author":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2020-02-12 18:41:54.000000000","tz":0},"committer":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2020-02-12 18:47:26.000000000","tz":0},"subject":"Update the Akamai DNS Driver matrix information","message":"Update the Akamai DNS Driver matrix information\n\n* Mark Akamai v2 as untested\n* Mark Akamai eDNS as known broken due to the API being removed\n* Add basic docs for Akamai v2\n\nChange-Id: I4f6c6c1c96251bfc4f7ad3bb6b0bd2025397fa03\nSigned-off-by: Graham Hayes \u003cgr@ham.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cf1a9d8cbc65d8043bd974e1ed6522782bef6ca7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cf1a9d8cbc65d8043bd974e1ed6522782bef6ca7"}]},"branch":"refs/heads/master"},"ce7a4a8886422d56f96785835f170d84fa9333bf":{"kind":"REWORK","_number":2,"created":"2020-02-13 15:12:06.000000000","uploader":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"ref":"refs/changes/51/707451/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/51/707451/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/707451/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/707451/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/51/707451/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/51/707451/2"}}},"commit":{"parents":[{"commit":"318b8d03193d3d9c26842528e9eaf25814df3ba9","subject":"Implement create/delete zone for Akamai v2 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/318b8d03193d3d9c26842528e9eaf25814df3ba9"}]}],"author":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2020-02-12 18:41:54.000000000","tz":0},"committer":{"name":"Graham Hayes","email":"gr@ham.ie","date":"2020-02-13 15:11:54.000000000","tz":0},"subject":"Update the Akamai DNS Driver matrix information","message":"Update the Akamai DNS Driver matrix information\n\n* Mark Akamai v2 as untested\n* Mark Akamai eDNS as known broken due to the API being removed\n* Add basic docs for Akamai v2\n\nChange-Id: I4f6c6c1c96251bfc4f7ad3bb6b0bd2025397fa03\nSigned-off-by: Graham Hayes \u003cgr@ham.ie\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ce7a4a8886422d56f96785835f170d84fa9333bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ce7a4a8886422d56f96785835f170d84fa9333bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
