)]}'
{"id":"openstack%2Fdesignate~709350","triplet_id":"openstack%2Fdesignate~master~Iec8791077aa7f1cb8edfa64235fb718a98aaba90","project":"openstack/designate","branch":"master","topic":"doc-cleanup","hashtags":[],"change_id":"Iec8791077aa7f1cb8edfa64235fb718a98aaba90","subject":"Cleanup docs building","status":"MERGED","created":"2020-02-23 13:49:06.000000000","updated":"2020-03-08 23:28:12.000000000","submitted":"2020-03-08 23:26:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709350-1583710013556-cb471cbf","meta_rev_id":"6f7d756a00546795fd352fbb6cd07e4f4bf84855","_number":709350,"virtual_id_number":709350,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-08 23:26:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"recommended":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":1,"date":"2020-02-24 13:02:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-02-24 13:04:45.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2020-02-24 12:24:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-08 22:24:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-08 22:24:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2020-02-24 12:24:55.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-24 12:24:55.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2020-02-24 13:02:44.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-24 13:04:45.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-03-08 22:24:08.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-03-08 23:26:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d2a13b1f60af4d831f8d0eb805ae5a1a938570cd","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-23 13:49:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c583606f69806df24c76e9220b0cd83b9c4979b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 14:48:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8d4e8ecdaa2a49dc89514caba32e0df1 : SUCCESS in 4m 40s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9f709e4d3bad40feac3d902c33565b9f : SUCCESS in 38m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2314cdc445914bf7827e1c77fd9465f7 : SUCCESS in 40m 30s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/111cc0ee01c2426bb220d249648a7617 : FAILURE in 26m 44s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/655b4de4edfe470ab5b134ace6d821bf : SUCCESS in 58m 05s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0b57b24ed57049d893c21e9922a2ffb7 : SUCCESS in 34m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/533b7cd3e6d3457cab0a0b199a0a34dc : SUCCESS in 38m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7bc72df025d41e4b7672cefb542b4a0 : SUCCESS in 6m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2921f576de14072b428904b11b08d00 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5874201a00954ae6a98ab9b224405a68 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9b35f70b3b646078fa45756d5690571 : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0e83aa01cd6a470b9f0a4cb27da325ae : SUCCESS in 9m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7064b98a4b11405bbc5e24aef401d884 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b2bf83434d5474da216edcaf4b5dd2d : FAILURE in 6m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/68453cbf164e4ab69624d29af080a624 : SUCCESS in 3m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c69d13a9c074385bee2bbbc9a6ac93a : SUCCESS in 6m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6113029b61514d048d6fd36c1b214306 : SUCCESS in 35m 04s","accounts_in_message":[],"_revision_number":1},{"id":"bd38aa8051ad4fa520d5b2978c3770281ad771b3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-23 15:18:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"65a21e40219a2339e3ebf996066395274cb5c886","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 16:23:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b7ee19d8b0bb41d2a95ceb68d762d3bf : SUCCESS in 4m 54s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/900973c866d141dd8e1c19344d8dc2fd : SUCCESS in 35m 43s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d9a1a9cd4e954560a28c2a577ae35928 : SUCCESS in 38m 42s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/4e09ddc5da414e1fb425c83d48fa7b2c : FAILURE in 32m 04s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6bac1d2cf5cd4cf5a511fca2542f84bc : SUCCESS in 1h 00m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4a83bca7096c43079da933b98eb70f48 : SUCCESS in 39m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9fd0ef9017124d67a2951b0ebff543f3 : SUCCESS in 36m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36b784d23bea46adacf418efd521a01d : SUCCESS in 6m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebfedb209375446caf2918da0270645e : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c16ca93e83fc41a1a351a0c59cf1212b : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87f574e99717486c901532dbcbd8b9d1 : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/365c540bb5e04762ad9222229187da40 : SUCCESS in 8m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/edbd337b6efd4b4387b35e54acd35906 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24ac4f4920924364806bf0645a132d87 : FAILURE in 5m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ae0cbe04e25c408f9a79042a900246cb : SUCCESS in 3m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2b746b22e7b4d3a928dc32a0528b590 : SUCCESS in 7m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ca7e764daa6a4f158b6646476d2c50e1 : SUCCESS in 33m 49s","accounts_in_message":[],"_revision_number":2},{"id":"705f7bdf917f1c0d3525bceb5d26dbac261f8987","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-23 16:33:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e6144348aca98a0ae59d9431737cfaaf29a7dca9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 17:35:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/109d78073d1a47a3b3dc599bb7daae4c : SUCCESS in 6m 06s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d346d1f1203e492d873282491d1c3b3b : SUCCESS in 43m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ade04726f69e433187a06d361a60c6db : SUCCESS in 37m 46s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/d1d6fefa705442b6b8d62674b11c0fde : FAILURE in 32m 15s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9146b3fdca314c6e90d186e83cd24cf5 : SUCCESS in 59m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/31783e472d9744ca8d815b07d6aef2ae : SUCCESS in 38m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/42e0470fb2904ac1a0b621befd625473 : SUCCESS in 38m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c0f9d56b477492c85d042a76da1e944 : SUCCESS in 5m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57685b06323543e8b6464ab88067e0d6 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ee8dcf132b64557ac772c94b722b648 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/580370e482fa4e5e84038afc68ad0235 : SUCCESS in 5m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/296d0f123d4d444dafd1387098a46955 : SUCCESS in 8m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c68274606a84a689f9bca5123aa0b5f : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/631e2fb22d40402dbbe1be3b305ce919 : SUCCESS in 7m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/18e24dba1d4948d0ad1f3fcda68618a3 : SUCCESS in 3m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b93647b9dbd464b85bac929576fba51 : SUCCESS in 6m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d7ff3cae35d24932a3f86897e140d2a5 : SUCCESS in 36m 11s","accounts_in_message":[],"_revision_number":3},{"id":"547d77443e2e4fec8b0553e85cc4b32812f092f8","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2020-02-24 12:24:55.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"c0b838cb415b1268bc250ab3efae1c81065e10c0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-24 13:02:44.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"cb48bfbc11b923c13b0d32ed758473e71e948c7a","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-02-24 13:04:45.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5347cdfa8aa3abbe27a67d221090282af34c7902","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-03-08 22:24:08.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6c552b31678a3cf169b9db23435be448b550f39d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-08 22:24:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0b2fb0e7ecba6bbbdc6348e16b5b4d269566669c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-08 23:26:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c5b341ccc5f840c6b594f04ae864c822 : SUCCESS in 3m 50s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/369980c565bd4378bd1b50a4b65e107f : SUCCESS in 47m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/330554289cf346f78af355ff13001d96 : SUCCESS in 45m 06s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5795a54fd2c4428c841d203dc4a3b7c1 : SUCCESS in 1h 00m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e3d8f62376d1452a9073857d0f0ea301 : SUCCESS in 42m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b17be85d3c644fa8bdae18ea3ccc03a8 : SUCCESS in 42m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4bf2e5094d94fdbb6ff36445306c4bc : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3667fd6280b4dffba382f06d3adfa20 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39007b6614464a2fbf02e5d7f3902e28 : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd534944f2d44da8945f28f053503269 : SUCCESS in 12m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df24d18f74f14511b7157ce879bddfd1 : SUCCESS in 9m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a77c2c8c2f63423b8ba2cc2bdfbc408e : SUCCESS in 3m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46929d211d2b40f2b754e6c1e76b44e6 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":3},{"id":"d2f88c46ae1d20227d9c8567e8b4f28fd5235763","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-08 23:26:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"59203b50977570bb6ba0bed97a078643f94303a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-08 23:28:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/97280f2169414fc0884e10e9ff07ad2e : SUCCESS in 59s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c429e7b71324b50863abb04835aaecf : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/136d1ff1339a4417829bca6b89f1ae71 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"21f94dea86abecd0f238f01c4d39d2fafb84d832","revisions":{"368f963a404883342da472f758a071f687d12cfc":{"kind":"REWORK","_number":1,"created":"2020-02-23 13:49:06.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/50/709350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/50/709350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/709350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/709350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/709350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/50/709350/1"}}},"commit":{"parents":[{"commit":"8b5ec12818c11be667195a954aa1f15c8971e3c3","subject":"Merge \"Update the Akamai DNS Driver matrix information\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8b5ec12818c11be667195a954aa1f15c8971e3c3"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-23 13:48:14.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-23 13:48:14.000000000","tz":60},"subject":"Cleanup docs building","message":"Cleanup docs building\n\n* Switch to newer openstackdocstheme version and remove settings\n  that are default now\n* Remove unneeded requirements from tox.ini\n\nChange-Id: Iec8791077aa7f1cb8edfa64235fb718a98aaba90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/368f963a404883342da472f758a071f687d12cfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/368f963a404883342da472f758a071f687d12cfc"}]},"branch":"refs/heads/master"},"fc49885a2fe5e597f5f148a5a45a4d456f2eb4e7":{"kind":"REWORK","_number":2,"created":"2020-02-23 15:18:25.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/50/709350/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/50/709350/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/709350/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/709350/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/709350/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/50/709350/2"}}},"commit":{"parents":[{"commit":"8b5ec12818c11be667195a954aa1f15c8971e3c3","subject":"Merge \"Update the Akamai DNS Driver matrix information\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8b5ec12818c11be667195a954aa1f15c8971e3c3"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-23 13:48:14.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-23 15:18:08.000000000","tz":60},"subject":"Cleanup docs building","message":"Cleanup docs building\n\n* Switch to newer openstackdocstheme version and remove settings\n  that are default now\n\nChange-Id: Iec8791077aa7f1cb8edfa64235fb718a98aaba90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc49885a2fe5e597f5f148a5a45a4d456f2eb4e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc49885a2fe5e597f5f148a5a45a4d456f2eb4e7"}]},"branch":"refs/heads/master"},"21f94dea86abecd0f238f01c4d39d2fafb84d832":{"kind":"REWORK","_number":3,"created":"2020-02-23 16:33:35.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/50/709350/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/50/709350/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/709350/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/709350/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/50/709350/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/50/709350/3"}}},"commit":{"parents":[{"commit":"8b5ec12818c11be667195a954aa1f15c8971e3c3","subject":"Merge \"Update the Akamai DNS Driver matrix information\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8b5ec12818c11be667195a954aa1f15c8971e3c3"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-23 13:48:14.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-23 16:33:27.000000000","tz":60},"subject":"Cleanup docs building","message":"Cleanup docs building\n\n* Switch to newer openstackdocstheme version and remove settings\n  that are default now\n\nChange-Id: Iec8791077aa7f1cb8edfa64235fb718a98aaba90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/21f94dea86abecd0f238f01c4d39d2fafb84d832"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/21f94dea86abecd0f238f01c4d39d2fafb84d832"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
