)]}'
{"id":"openstack%2Fdesignate~712830","triplet_id":"openstack%2Fdesignate~stable%2Frocky~I0a11fc62a6574ef733074a7e5292054119812ac2","project":"openstack/designate","branch":"stable/rocky","hashtags":[],"change_id":"I0a11fc62a6574ef733074a7e5292054119812ac2","subject":"Eventlet bump + properly decode strings in py3x","status":"ABANDONED","created":"2020-03-13 03:46:01.000000000","updated":"2021-12-24 04:30:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9855403eac4ef5ef80f3df10f42802c549723b14","_number":712830,"virtual_id_number":712830,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-03-19 21:08:01.000000000","_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-19 22:16:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-19 22:16:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1eee1e16cd6bb287a5f47ba236f2e1a4639d10cd","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-03-13 03:46:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58965c2123ceb2e61685e3f6fb8ca16955209789","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 04:53:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bc55e0f6a3794d2a8e051f094ef63f37 : SUCCESS in 40m 01s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/59f8ecae49424a36814b28262ac70260 : SUCCESS in 40m 18s\n- designate-bind9-py35 https://zuul.opendev.org/t/openstack/build/42166b2d52d34dd6b3337997e0adf6e5 : SUCCESS in 39m 40s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/32992cc944bc42808a823ee7d5c2b6ca : SUCCESS in 36m 43s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/502c615b2ab24fffbc5104c2111e5bfb : SUCCESS in 35m 11s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/12f3ad7524af4efeb416a23418acf670 : SUCCESS in 37m 16s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/919795ede90e490c9ef90280140bd20b : FAILURE in 1h 01m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9c435af770b4e69876ebcab622489b5 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33551493936e4ec8bd7a63a2d2714bcc : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/301f545c24a94e348ee5e040c1cafca3 : SUCCESS in 4m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ed46a341029d4057b1735ce77c070967 : SUCCESS in 4m 39s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b661a3c4a2064f05b6be33f95be3999a : SUCCESS in 4m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f19bcb02bfa94c97bfc5be4247724b56 : FAILURE in 1m 59s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/6d61ac70bf844c2494127c67a9824492 : SUCCESS in 32m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4052ecd169c94c859bc1cd1f1602c927 : SUCCESS in 3m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73982f6879ae417b93043e95a732472b : SUCCESS in 5m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"89c06d365895917169eab6da43e3177ba3e220da","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-03-13 06:06:43.000000000","message":"Patch Set 1:\n\nDoubt this will be able to merge, as requirements probably won\u0027t be willing to bump the upper-constraint for eventlet in an extended-maintenance branch (and it honestly wouldn\u0027t be a good call, I think, for all projects). Leaving this patch here for posterity, as other people may run into this.","accounts_in_message":[],"_revision_number":1},{"id":"1e1271d109530225d81150667ad2362688ad2c93","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-03-17 18:15:07.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"1de56c8ebb7c59449b1f5648e79f670f3ec4521a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-03-19 21:06:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"92d1f9bd1aeaa0f612392b8c1df0229b5b84a016","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-03-19 21:08:01.000000000","message":"Patch Set 2:\n\nWell, turns out I was wrong, and they totally did bump the upper-constraint! So this is mergeable, and will make designate ACTUALLY work on py3 instead of horribly break. :D","accounts_in_message":[],"_revision_number":2},{"id":"0a5ba76096459315f52afffcfcd3cd2f0d0b7ed9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 22:16:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1ad69eb280824da3828fc2fc0660d75e : SUCCESS in 44m 56s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/fcd1d3fb1be645afa341bad15b846673 : SUCCESS in 41m 35s\n- designate-bind9-py35 https://zuul.opendev.org/t/openstack/build/4e11c24d709f455f9fa5170f760dadbc : SUCCESS in 49m 09s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/6e1e27be46a14486aeef84b0a0035f12 : SUCCESS in 46m 57s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/795ce1e7933d4ac38f257c3e1fbccce5 : SUCCESS in 39m 02s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/814b521203434b268b7ee6c2767b8b80 : SUCCESS in 34m 39s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5c8c6bfb496b449f839966e299f22356 : SUCCESS in 1h 03m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6866bc71e90f4183ac6fc6a115fff9a6 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48a390402521447eaae7fc9aaf636850 : SUCCESS in 6m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/50e739e8f4c642f6bcf7998dd672cb1d : SUCCESS in 4m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4f19209133fe4a049ab119e744144140 : SUCCESS in 6m 40s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/41bdd128f880489cac4afe280ce7cea9 : SUCCESS in 4m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/61ce8a6e64d24464967d095101cdac18 : FAILURE in 2m 34s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/f82c25fe79ea4c7d9ab5c743d2be59e7 : FAILURE in 33m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abd6d0ca5e6f4784a9235b6087f2010c : SUCCESS in 3m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33d794a336174d17bad2251ea6ef3c6a : SUCCESS in 5m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9855403eac4ef5ef80f3df10f42802c549723b14","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-24 04:30:15.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8b7ff405826019a6371eb351e04fab78d82c65a5","revisions":{"7cee2addfef46c0bb719b24afbb1d7b6c11f7d6f":{"kind":"REWORK","_number":1,"created":"2020-03-13 03:46:01.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/30/712830/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/30/712830/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/712830/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/712830/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/712830/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/30/712830/1"}}},"commit":{"parents":[{"commit":"34160d0602abe7fbc566191126698d8bac23ea9d","subject":"Merge \"Clean up zone locking\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/34160d0602abe7fbc566191126698d8bac23ea9d"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-03-13 03:41:17.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-03-13 03:45:43.000000000","tz":-420},"subject":"Eventlet bump + properly decode strings in py3x","message":"Eventlet bump + properly decode strings in py3x\n\nEventlet 0.20.0 does not work in Python36, due to a SSLContext bug:\nhttps://bugs.launchpad.net/manila/+bug/1788253\n\nWith eventlet 0.21.0 (the fix), eventlet now returns str not bytes in\npy3, so we need to determine the type before blindly decoding.\n\nChange-Id: I0a11fc62a6574ef733074a7e5292054119812ac2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7cee2addfef46c0bb719b24afbb1d7b6c11f7d6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7cee2addfef46c0bb719b24afbb1d7b6c11f7d6f"}]},"branch":"refs/heads/stable/rocky"},"8b7ff405826019a6371eb351e04fab78d82c65a5":{"kind":"REWORK","_number":2,"created":"2020-03-19 21:06:58.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/30/712830/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/30/712830/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/712830/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/712830/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/30/712830/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/30/712830/2"}}},"commit":{"parents":[{"commit":"34160d0602abe7fbc566191126698d8bac23ea9d","subject":"Merge \"Clean up zone locking\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/34160d0602abe7fbc566191126698d8bac23ea9d"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-03-13 03:41:17.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-03-19 21:06:33.000000000","tz":-420},"subject":"Eventlet bump + properly decode strings in py3x","message":"Eventlet bump + properly decode strings in py3x\n\nEventlet 0.20.0 does not work in Python36, due to a SSLContext bug:\nhttps://bugs.launchpad.net/manila/+bug/1788253\n\nWith eventlet 0.21.0 (the fix), eventlet now returns str not bytes in\npy3, so we need to determine the type before blindly decoding.\n\nChange-Id: I0a11fc62a6574ef733074a7e5292054119812ac2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8b7ff405826019a6371eb351e04fab78d82c65a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8b7ff405826019a6371eb351e04fab78d82c65a5"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
