)]}'
{"id":"openstack%2Fdesignate~717955","triplet_id":"openstack%2Fdesignate~master~I98f7f80ce365cdee33528f9964c03274f62a795a","project":"openstack/designate","branch":"master","topic":"lock","hashtags":[],"change_id":"I98f7f80ce365cdee33528f9964c03274f62a795a","subject":"Adding distributed locking to central","status":"MERGED","created":"2020-04-07 05:40:30.000000000","updated":"2020-06-16 21:47:08.000000000","submitted":"2020-04-08 14:55:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717955-1586357713933-9e742fff","meta_rev_id":"7e413035bb7824f5e13d09ff87d69bf170f09c01","_number":717955,"virtual_id_number":717955,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-08 14:55:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"recommended":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":1,"date":"2020-04-08 11:50:45.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2020-04-08 13:41:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":2,"date":"2020-04-08 08:18:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2020-04-08 13:41:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":1,"date":"2020-04-08 08:18:49.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-08 08:18:49.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-04-08 11:50:45.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-04-08 13:41:16.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2020-04-08 14:55:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"61fb56ab227e208319c4e9e40dbaeafad422d723","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-04-07 05:40:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1506f86a0374aeb3f6a7b9956209d39808c10921","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-04-07 06:13:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f5492b421ca82255fc3c9b4e8760492ec1fc6a0e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-04-07 06:16:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4e90c115be6a8b7ea4c09464279d5484e7cd3982","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-04-07 06:28:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e2cf203dcdd1a30b1ada03f529df8ae3d965e627","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 07:38:34.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/72f8bc57b7094b25be8faacac39cc668 : SUCCESS in 6m 31s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f20035e17f034ce291589e6ccd9c7b1a : FAILURE in 41m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/54a77be200ed45f29a7cb77d607c514d : FAILURE in 57m 23s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/574c0e8292284b4da32c494c5d55cb7b : SUCCESS in 1h 17m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/afa1b2665554472faaf7d077e1ab5225 : FAILURE in 47m 10s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a54eeb5afa4144bbb8fcbd25539ae34a : FAILURE in 47m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20df5eaecb20419da7ff202c618b6231 : SUCCESS in 10m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/750c181c267c48e5aa6d487efb853ede : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0b0ed6ea5364e2fad7d25eb377cc8fa : FAILURE in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d71eda6800fb4ada96c58f362c6b0cbc : SUCCESS in 6m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0c19b7ec1c964cb697e2ce709be81196 : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ca14e4615474fe4a2c582f7b048d03e : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c11979730684d2bbe60aa20895fd4ce : SUCCESS in 12m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6c753e3e640b4373be9ea2267e6d572a : SUCCESS in 3m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd3c6de5fb314950b7e934221125f08d : SUCCESS in 8m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0cbb68da8a4048faaf0f1a662f9b1192 : FAILURE in 42m 30s\n\nWarning:\n  Comments left for invalid file designate/tests/unit/test_context.py\n  Comments left for invalid file doc/source/admin/config.rst\n  Comments left for invalid file designate/context.py\n  Comments left for invalid file designate/tests/__init__.py\n  Comments left for invalid file doc/ext/support_matrix.py\n  Comments left for invalid file designate/utils.py\n  Comments left for invalid file designate/sqlalchemy/session.py\n  Comments left for invalid file tools/pretty_flake8.py","accounts_in_message":[],"_revision_number":3},{"id":"824c8f8a409154237efd91f687dcb34f4a067de1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-04-07 08:06:14.000000000","message":"Patch Set 3:\n\n\u003e (1 comment)\n\nYea that was my first idea, but the nested implementation makes that real hard.","accounts_in_message":[],"_revision_number":3},{"id":"412df64d273e3603f9335885a9ad8188349868b9","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-04-07 08:09:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2c42f28f7dcd366470c51a4ada95e0f94a094401","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-04-07 08:52:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"77e97560e03b1cb42be5dafdfadf06d910a3d809","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-04-07 08:53:22.000000000","message":"Patch Set 5:\n\n\u003e (1 comment)\n\nYea - you were right. Moved the lock to the top.","accounts_in_message":[],"_revision_number":5},{"id":"8ddeab52107dc95d063f1ab65ee6c71b0dc2f15e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-04-07 09:20:28.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1652e25c07a77ec8cf8ab99f35b0aa6b06002071","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-04-07 09:20:35.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"73985fa5daedabbb5aecddd51ab2ccc299248a84","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2020-04-07 09:50:40.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"aa8fea524ff3cf0be295e66ee86ffb685434b6e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 10:41:47.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b2439febefe14151943546160b5c1303 : SUCCESS in 8m 28s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1afd6452934b43779914837c6f38aa87 : FAILURE in 1h 02m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1d3e3486b21c40008b49725fc22dc4c6 : FAILURE in 1h 03m 54s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/aa937def1578412bb87516abef0eb7bb : SUCCESS in 1h 13m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/830aec4b24684c3188b1908aaa94ce1e : FAILURE in 1h 06m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/80956acad7b14b9d8375571285ab71af : FAILURE in 1h 07m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7d4eda85c2a44e3b7092c75fdd7c79f : SUCCESS in 6m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec2f4905984f48ec8fee6be3af7cf5bd : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5fa4e11d39e4e4dac6298d6df1151b0 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69775b3f54c7400498a471b2af473037 : SUCCESS in 6m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6958c3d76502460e96c555a78b19fd6c : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/690b5b91112e4802b9686b971de4ef4a : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb7b3efe4a3c4ff4bbcbf243937f30c4 : SUCCESS in 9m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/013d778c17904a408ac4d2367d1de045 : SUCCESS in 3m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22c4ab57f46a48e09be3995568a30ed0 : SUCCESS in 7m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/72650f758fc44bc2b6730fb88b8b1e5f : FAILURE in 29m 02s\n\nWarning:\n  Comments left for invalid file designate/tests/unit/test_context.py\n  Comments left for invalid file doc/source/admin/config.rst\n  Comments left for invalid file tools/pretty_flake8.py\n  Comments left for invalid file designate/context.py\n  Comments left for invalid file designate/tests/__init__.py\n  Comments left for invalid file doc/ext/support_matrix.py\n  Comments left for invalid file designate/sqlalchemy/session.py\n  Comments left for invalid file designate/utils.py","accounts_in_message":[],"_revision_number":7},{"id":"dfd568ca224f9380895f48d3a8b5f876a3b3fb07","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-04-07 11:46:35.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"fd0a1f9d0700da567dcdf4478519fe4b5aba11a5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-04-07 17:24:06.000000000","message":"Patch Set 7:\n\n\u003e (1 comment)\n\nYep - I was just being lazy as I didn\u0027t know how it was supposed to be done in the plugin. Plus not even sure how our producer is ever working in tests, because it is configured to use memcached.","accounts_in_message":[],"_revision_number":7},{"id":"e5cd93fc1594681eab87d5fdd89a7a8efb5846dd","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-04-07 17:59:23.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a7083c0dc3a7c3d7ed5a282815b8ef64064ba682","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-04-07 20:28:09.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"38d9007a8e449225c75b26dd54d168ac84cf3638","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-04-07 20:28:55.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"a7828083bd8600a259ee08a2a2e6035829b31ca9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 21:46:41.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5e7d4b66de2542639a935e2005f007b6 : SUCCESS in 38m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bad2530e0e8b4a599a5daf6ea59b9920 : SUCCESS in 44m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b2034b3ca653414483f2d39a7e66a381 : SUCCESS in 1h 10m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b92955b1a06748eea450b3f0c2c0e40c : SUCCESS in 40m 33s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6d58089c51734401aba9f00bcd89175d : SUCCESS in 41m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dfce49ac1f304f35a8049554bfcbb367 : SUCCESS in 6m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0faa9b19867d40f397a4c9d56aa9e6ba : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b285c4598df4b48a0248552b764e065 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/002af549a228478e954b5dc2007969ee : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0ea4a49b6d71465ca4ece7c05a5745db : SUCCESS in 9m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f31b082f7b8b4ee39effe09cb53ebfe6 : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b47c6973cb64d74af51dc2715e16be1 : SUCCESS in 13m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/969d9e11306945c8b72587fa26b63d4a : SUCCESS in 33m 25s\n\nWarning:\n  Comments left for invalid file designate/tests/unit/test_context.py\n  Comments left for invalid file doc/source/admin/config.rst\n  Comments left for invalid file tools/pretty_flake8.py\n  Comments left for invalid file designate/context.py\n  Comments left for invalid file designate/tests/__init__.py\n  Comments left for invalid file doc/ext/support_matrix.py\n  Comments left for invalid file designate/sqlalchemy/session.py\n  Comments left for invalid file designate/utils.py","accounts_in_message":[],"_revision_number":9},{"id":"009789b9697831b5a5bb2acb266bdb124c242392","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-04-07 22:29:26.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"6f8a1ce951fc5f4d3380c714f4a2746fe69fd557","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 00:01:47.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/aec9f1f281714cd693c9dd223c2e73d5 : SUCCESS in 38m 45s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9e2f998417284bdd88c92e353e261f72 : SUCCESS in 38m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e3cecb37524d4faabe102776af85164f : SUCCESS in 1h 08m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d702300c2039451fa0d52c8c6275eb90 : SUCCESS in 50m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e7f16db6bc664287826b3f44a892c9be : SUCCESS in 45m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/729ebad759044866b63f3bf70e4eccbf : SUCCESS in 7m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3016433a1044ab9812fd8876c4ec46e : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03037957998c4daa99f4b6c28daebe35 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cddecf43612545e98184af0a59a31559 : SUCCESS in 7m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0125945ca8a34eeba823350799e36b1d : SUCCESS in 9m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b90059d98694ece8b0e90628db04a21 : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94606e2fa0c14a54bfb41b7488df222d : SUCCESS in 11m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b226e252e7d14cb2ac68278ebd4dc0bc : SUCCESS in 39m 13s\n\nWarning:\n  Comments left for invalid file designate/sqlalchemy/session.py\n  Comments left for invalid file designate/tests/unit/test_context.py\n  Comments left for invalid file doc/source/admin/config.rst\n  Comments left for invalid file tools/pretty_flake8.py\n  Comments left for invalid file doc/ext/support_matrix.py\n  Comments left for invalid file designate/utils.py\n  Comments left for invalid file designate/tests/__init__.py\n  Comments left for invalid file designate/context.py","accounts_in_message":[],"_revision_number":10},{"id":"8409b6afec21a554399066c74088b27ffc42eef7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-04-08 05:37:06.000000000","message":"Patch Set 10: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":10},{"id":"76174fc554887e211fbdee0bf7119e3d9637f7cd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-04-08 06:57:03.000000000","message":"Patch Set 10: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"bf73bf294ed0710f3f987d1c85174190319008ce","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-04-08 07:18:46.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"09c01c9ba76a920197f2dc03a216fbbf25416a28","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-04-08 08:18:49.000000000","message":"Patch Set 11: Code-Review+2 Review-Priority+1\n\nHumm, why doesn\u0027t review-priority persist across PSs?","accounts_in_message":[],"_revision_number":11},{"id":"a9ec8ee55d8a836d56e648d6108d73b96f258d08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 08:41:33.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2796b2e9b1c743539f884a8007bb42b9 : SUCCESS in 45m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4b973bade9984a0e9f8598fbe11a127b : SUCCESS in 48m 05s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/66625c5373074f81a1e40373c5e256f3 : SUCCESS in 1h 12m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/516a8e44e42a4c2eafedfc74e7b0e04b : SUCCESS in 36m 16s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8687aa5ceac54c20854c77ef81d44ca5 : SUCCESS in 41m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/400f61704b5c43778052a0d9a20da298 : SUCCESS in 7m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/860635c751a34cfda34b58e30c47c89f : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d96b1903f8f4a428762f195088f0381 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a253a2f093f4bcea4fd05e767fab632 : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a28209b7ddca4bba9d061eb0893e2049 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd77d6fc7c964ac9ad058d17b6e4cd90 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3eb6bd8555ac40b89e56e130a208d078 : SUCCESS in 10m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b0b4796e843b4965aa50c6cc37bad5e9 : SUCCESS in 35m 50s\n\nWarning:\n  Comments left for invalid file designate/tests/unit/test_context.py\n  Comments left for invalid file doc/source/admin/config.rst\n  Comments left for invalid file tools/pretty_flake8.py\n  Comments left for invalid file designate/tests/__init__.py\n  Comments left for invalid file doc/ext/support_matrix.py\n  Comments left for invalid file designate/utils.py\n  Comments left for invalid file designate/sqlalchemy/session.py\n  Comments left for invalid file designate/context.py","accounts_in_message":[],"_revision_number":11},{"id":"4e0b15a89bb08942c62e01f73290d636a945c037","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-04-08 11:50:45.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"c91f08e0fe0c85b03fa4bfbe1420e7238b2ba726","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2020-04-08 13:41:16.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"20f82a27b88b8686abc8ff30fbdbc95fb52de33b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 13:41:31.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"5143c1791862832a6b85b3778f8497ebbc011f8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 14:55:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"36b897a30aa7b631bf739ba35fe80fc72ce7fa5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 14:55:13.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a98a4052c42b4271b71b00dc1a25b867 : SUCCESS in 43m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b0e049ad959f472792f77934e93db1dd : SUCCESS in 37m 02s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c39478a84fc94a93a13b697b2689ab27 : SUCCESS in 1h 11m 23s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4fd4cd50cbf047fabe34beefaa81decf : SUCCESS in 46m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fea415cad2554c3a9413259d3b0a37cc : SUCCESS in 40m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41492190dbae4576909872c15bff256c : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d551a901d3c54f1394f58a95ad0e82bd : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/610eda9db6864299a2c64fba0cf6e3fb : SUCCESS in 6m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3fe9ca40e3f94d8c9a30564e764ba23d : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f700f63930e42479b408b882da0160e : SUCCESS in 11m 52s\n\nWarning:\n  Comments left for invalid file designate/tests/unit/test_context.py\n  Comments left for invalid file doc/source/admin/config.rst\n  Comments left for invalid file tools/pretty_flake8.py\n  Comments left for invalid file designate/tests/__init__.py\n  Comments left for invalid file doc/ext/support_matrix.py\n  Comments left for invalid file designate/utils.py\n  Comments left for invalid file designate/sqlalchemy/session.py\n  Comments left for invalid file designate/context.py","accounts_in_message":[],"_revision_number":11},{"id":"e91262d3ec1547fda41f79d2d2cc3434de5caaaf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 14:58:17.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23ffde1654814cefad53e59344a75675 : SUCCESS in 2m 06s","accounts_in_message":[],"_revision_number":11},{"id":"79e029b7676bad92e491181814735ebc16398d5d","author":{"_account_id":29647,"name":"Akradiy Shinkarev","email":"kewatut@gmail.com","username":"k3nny0ne"},"date":"2020-06-16 21:47:08.000000000","message":"Patch Set 11: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 6a48f54dfc74357e8eacb119f881f03db5e28550","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"f6090d885c8ebe00a89307a1e2f5621de75dcda8","revisions":{"c10442704ad251fde0353d8aee6c5bd2796eb09c":{"kind":"REWORK","_number":1,"created":"2020-04-07 05:40:30.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/717955/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/717955/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/717955/1"}}},"commit":{"parents":[{"commit":"19ec7d9cd17b4f972d96a8ccf5088842aa593757","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/19ec7d9cd17b4f972d96a8ccf5088842aa593757"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 05:39:36.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 05:40:14.000000000","tz":-420},"subject":"[WIP] Adding distributed locking to inc zone serial","message":"[WIP] Adding distributed locking to inc zone serial\n\nChange-Id: I98f7f80ce365cdee33528f9964c03274f62a795a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c10442704ad251fde0353d8aee6c5bd2796eb09c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c10442704ad251fde0353d8aee6c5bd2796eb09c"}]},"branch":"refs/heads/master"},"014036ea0d772da7e29f6f6361a30c7ab34fae38":{"kind":"REWORK","_number":2,"created":"2020-04-07 06:13:38.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/717955/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/717955/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/717955/2"}}},"commit":{"parents":[{"commit":"19ec7d9cd17b4f972d96a8ccf5088842aa593757","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/19ec7d9cd17b4f972d96a8ccf5088842aa593757"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 05:39:36.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 06:13:24.000000000","tz":-420},"subject":"[WIP] Adding distributed locking to inc zone serial","message":"[WIP] Adding distributed locking to inc zone serial\n\nChange-Id: I98f7f80ce365cdee33528f9964c03274f62a795a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/014036ea0d772da7e29f6f6361a30c7ab34fae38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/014036ea0d772da7e29f6f6361a30c7ab34fae38"}]},"branch":"refs/heads/master"},"109d3705862c0afce0e777e08028438316d87678":{"kind":"REWORK","_number":3,"created":"2020-04-07 06:16:42.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/717955/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/717955/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/717955/3"}}},"commit":{"parents":[{"commit":"19ec7d9cd17b4f972d96a8ccf5088842aa593757","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/19ec7d9cd17b4f972d96a8ccf5088842aa593757"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 05:39:36.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 06:16:33.000000000","tz":-420},"subject":"[WIP] Adding distributed locking to inc zone serial","message":"[WIP] Adding distributed locking to inc zone serial\n\nChange-Id: I98f7f80ce365cdee33528f9964c03274f62a795a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/109d3705862c0afce0e777e08028438316d87678"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/109d3705862c0afce0e777e08028438316d87678"}]},"branch":"refs/heads/master"},"019f5f2ccf0cc7770ef231df21d1182d4ad4901c":{"kind":"REWORK","_number":4,"created":"2020-04-07 08:09:49.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/717955/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/717955/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/717955/4"}}},"commit":{"parents":[{"commit":"19ec7d9cd17b4f972d96a8ccf5088842aa593757","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/19ec7d9cd17b4f972d96a8ccf5088842aa593757"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 05:39:36.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 08:09:34.000000000","tz":-420},"subject":"[WIP] Adding distributed locking to inc zone serial","message":"[WIP] Adding distributed locking to inc zone serial\n\nChange-Id: I98f7f80ce365cdee33528f9964c03274f62a795a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/019f5f2ccf0cc7770ef231df21d1182d4ad4901c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/019f5f2ccf0cc7770ef231df21d1182d4ad4901c"}]},"branch":"refs/heads/master"},"0d4a90c6112cd1474d681373d4cd95e81d398b01":{"kind":"REWORK","_number":5,"created":"2020-04-07 08:52:35.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/717955/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/717955/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/717955/5"}}},"commit":{"parents":[{"commit":"19ec7d9cd17b4f972d96a8ccf5088842aa593757","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/19ec7d9cd17b4f972d96a8ccf5088842aa593757"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 05:39:36.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 08:51:18.000000000","tz":-420},"subject":"Adding distributed locking to central","message":"Adding distributed locking to central\n\nThe current locking implementation is limited to\nthe running process. This introduces distributed\nlocking that will help prevent race conditions\nwhen there are many instances of designate\nrunning.\n\nChange-Id: I98f7f80ce365cdee33528f9964c03274f62a795a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0d4a90c6112cd1474d681373d4cd95e81d398b01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0d4a90c6112cd1474d681373d4cd95e81d398b01"}]},"branch":"refs/heads/master"},"92d3834df9f954783acefd32b89c226603040394":{"kind":"REWORK","_number":6,"created":"2020-04-07 09:20:28.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/717955/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/717955/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/717955/6"}}},"commit":{"parents":[{"commit":"19ec7d9cd17b4f972d96a8ccf5088842aa593757","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/19ec7d9cd17b4f972d96a8ccf5088842aa593757"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 05:39:36.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 09:19:44.000000000","tz":-420},"subject":"Adding distributed locking to central","message":"Adding distributed locking to central\n\nThe current locking implementation is limited to\nthe running process. This introduces distributed\nlocking that will help prevent race conditions\nwhen there are many instances of designate-central\nrunning.\n\nCloses-Bug: #1768618\nChange-Id: I98f7f80ce365cdee33528f9964c03274f62a795a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/92d3834df9f954783acefd32b89c226603040394"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/92d3834df9f954783acefd32b89c226603040394"}]},"branch":"refs/heads/master"},"4dbb29b9dbecaaaa297428b885c6e63c844b8db3":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-04-07 09:20:35.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/717955/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/717955/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/717955/7"}}},"commit":{"parents":[{"commit":"65736672da70cea9fc795c5f4c0d9a195f852ce8","subject":"Stop setting version specific basepython binaries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/65736672da70cea9fc795c5f4c0d9a195f852ce8"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 05:39:36.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 09:20:35.000000000","tz":0},"subject":"Adding distributed locking to central","message":"Adding distributed locking to central\n\nThe current locking implementation is limited to\nthe running process. This introduces distributed\nlocking that will help prevent race conditions\nwhen there are many instances of designate-central\nrunning.\n\nCloses-Bug: #1768618\nChange-Id: I98f7f80ce365cdee33528f9964c03274f62a795a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4dbb29b9dbecaaaa297428b885c6e63c844b8db3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4dbb29b9dbecaaaa297428b885c6e63c844b8db3"}]},"branch":"refs/heads/master"},"937ffc451dd408d2a4680cef21c47f6ea622efd3":{"kind":"REWORK","_number":8,"created":"2020-04-07 20:28:09.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/717955/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/717955/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/717955/8"}}},"commit":{"parents":[{"commit":"19ec7d9cd17b4f972d96a8ccf5088842aa593757","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/19ec7d9cd17b4f972d96a8ccf5088842aa593757"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 05:39:36.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 20:28:03.000000000","tz":-420},"subject":"Adding distributed locking to central","message":"Adding distributed locking to central\n\nThe current locking implementation is limited to\nthe running process. This introduces distributed\nlocking that will help prevent race conditions\nwhen there are many instances of designate-central\nrunning.\n\nCloses-Bug: #1768618\nChange-Id: I98f7f80ce365cdee33528f9964c03274f62a795a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/937ffc451dd408d2a4680cef21c47f6ea622efd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/937ffc451dd408d2a4680cef21c47f6ea622efd3"}]},"branch":"refs/heads/master"},"786078301f9c37b7468c725564f28bf116af064d":{"kind":"REWORK","_number":9,"created":"2020-04-07 20:28:55.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/717955/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/717955/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/717955/9"}}},"commit":{"parents":[{"commit":"19ec7d9cd17b4f972d96a8ccf5088842aa593757","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/19ec7d9cd17b4f972d96a8ccf5088842aa593757"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 05:39:36.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 20:28:42.000000000","tz":-420},"subject":"Adding distributed locking to central","message":"Adding distributed locking to central\n\nThe current locking implementation is limited to\nthe running process. This introduces distributed\nlocking that will help prevent race conditions\nwhen there are many instances of designate-central\nrunning.\n\nCloses-Bug: #1768618\nChange-Id: I98f7f80ce365cdee33528f9964c03274f62a795a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/786078301f9c37b7468c725564f28bf116af064d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/786078301f9c37b7468c725564f28bf116af064d"}]},"branch":"refs/heads/master"},"99890220dc165a9dc10a9fed6faf49ab25352d97":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2020-04-07 22:29:26.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/717955/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/717955/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/717955/10"}}},"commit":{"parents":[{"commit":"19ec7d9cd17b4f972d96a8ccf5088842aa593757","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/19ec7d9cd17b4f972d96a8ccf5088842aa593757"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 05:39:36.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 22:29:25.000000000","tz":0},"subject":"Adding distributed locking to central","message":"Adding distributed locking to central\n\nThe current locking implementation is limited to\nthe running process. This introduces distributed\nlocking that will help prevent race conditions\nwhen there are many instances of designate-central\nrunning.\n\nCloses-Bug: #1871332\nChange-Id: I98f7f80ce365cdee33528f9964c03274f62a795a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/99890220dc165a9dc10a9fed6faf49ab25352d97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/99890220dc165a9dc10a9fed6faf49ab25352d97"}]},"branch":"refs/heads/master"},"f6090d885c8ebe00a89307a1e2f5621de75dcda8":{"kind":"REWORK","_number":11,"created":"2020-04-08 07:18:46.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/55/717955/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/717955/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/717955/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/717955/11"}}},"commit":{"parents":[{"commit":"19ec7d9cd17b4f972d96a8ccf5088842aa593757","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/19ec7d9cd17b4f972d96a8ccf5088842aa593757"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-07 05:39:36.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-04-08 07:18:33.000000000","tz":-420},"subject":"Adding distributed locking to central","message":"Adding distributed locking to central\n\nThe current locking implementation is limited to\nthe running process. This introduces distributed\nlocking that will help prevent race conditions\nwhen there are many instances of designate-central\nrunning.\n\nCloses-Bug: #1871332\nChange-Id: I98f7f80ce365cdee33528f9964c03274f62a795a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f6090d885c8ebe00a89307a1e2f5621de75dcda8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f6090d885c8ebe00a89307a1e2f5621de75dcda8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
