)]}'
{"id":"openstack%2Fdesignate~720766","triplet_id":"openstack%2Fdesignate~stable%2Fqueens~I0a85c5df5c323d6f636c8cb92ee5201c14cdb306","project":"openstack/designate","branch":"stable/queens","topic":"bug/1785769-stable/rocky-stable/queens","hashtags":[],"change_id":"I0a85c5df5c323d6f636c8cb92ee5201c14cdb306","subject":"Worker should send NOTIFY also to all servers in \u0027also_notifies\u0027 pool settings.","status":"MERGED","created":"2020-04-17 16:05:17.000000000","updated":"2020-07-24 23:25:43.000000000","submitted":"2020-07-24 23:25:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720766-1595633143894-58a94d7c","meta_rev_id":"e8b41942e1247427596e5b9a4d049602e902ee98","_number":720766,"virtual_id_number":720766,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-24 23:25:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"recommended":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":1,"date":"2020-05-18 16:14:47.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2020-05-15 18:30:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":1,"date":"2020-05-18 15:05:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2020-06-15 19:40:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2020-07-24 22:20:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-15 18:30:08.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2020-05-18 15:05:11.000000000","updated_by":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"reviewer":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"state":"REVIEWER"},{"updated":"2020-05-18 16:14:47.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-07-24 22:20:01.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-07-24 23:25:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a1b62a7748355f2e992c34fc5e6376eef4b93679","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-17 16:05:17.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"76b426fa339b171cbf168297f8e6fe3f8cffc231","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 16:40:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-devstack-bind9 https://zuul.opendev.org/t/openstack/build/a9b694e249b747d58fde9b946469943e : FAILURE in 27m 06s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/5c8fedbcba3e46188364e7877e96da4e : FAILURE in 26m 28s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/05fef6ad11d149489a3c7bbdf809b345 : FAILURE in 23m 10s\n- designate-devstack-pdns4-manager-model https://zuul.opendev.org/t/openstack/build/9926235e8d6f4c758303399e45c8f8be : FAILURE in 18m 39s\n- designate-devstack-pdns4-py35 https://zuul.opendev.org/t/openstack/build/d27697bafd3d428eae01fc5643ab8b0b : FAILURE in 19m 39s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bbf8fdf0d6d245fc81eca1641c93a6bd : FAILURE in 18m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f449b13938b14fd28354c0f336e59822 : SUCCESS in 5m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/794270948afa42169d3230af0199e344 : SUCCESS in 5m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7696ab4e2a07493b9bb0725dc9414dc5 : SUCCESS in 7m 31s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9d634f35eb744ac2b8f7c0d078dc2e6c : SUCCESS in 7m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/00905c2913b4419cbd9434511e319507 : FAILURE in 23m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2f80c82e981450486f7f34829a718a0 : SUCCESS in 8m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0929ac473211a490f8de5f9e16cccd096b4dad72","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-23 15:38:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7232a6b8fd902bbb3e97203b73acc378026d2c85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 16:52:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-devstack-bind9 https://zuul.opendev.org/t/openstack/build/b00383c2ee4144a4b4aedd48836791c3 : FAILURE in 19m 57s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/141c78d1d78c449f93f3cddb703b4739 : FAILURE in 18m 26s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/152fab8a006f4a199240835d6dc3eee5 : FAILURE in 24m 27s\n- designate-devstack-pdns4-manager-model https://zuul.opendev.org/t/openstack/build/9ca59b9780e240a991aa77e4ce08c101 : FAILURE in 24m 28s\n- designate-devstack-pdns4-py35 https://zuul.opendev.org/t/openstack/build/baa6ac4df2624b80ae8514f364591480 : FAILURE in 23m 30s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/03ed57f9ee2f42ab96c41843a2fe9607 : FAILURE in 24m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb32634c760b4d82889cf187cc236768 : SUCCESS in 7m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0edd6d1ba0654eefb5dcab49082e26a7 : SUCCESS in 6m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/555c52efeaf04634adcbe70605a2ed9c : SUCCESS in 6m 57s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/df214eda66e647af8e78c9c4742e50be : SUCCESS in 6m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/91e10813d7bc4d9eb4ae95df34dec31a : FAILURE in 19m 40s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef56083f1f0144cc8595f7d88c2ea228 : SUCCESS in 8m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ac4486b580c4eb3911349baff082b74f94d3234e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-28 22:32:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3d4980b147257f1391572b0f32a4d9a8a0c517cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 23:06:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-devstack-bind9 https://zuul.opendev.org/t/openstack/build/fe05c6d9d53b4fe08a44b620aed21ac6 : FAILURE in 18m 09s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/8f68281885c14b03aa8e2117baf28df0 : FAILURE in 18m 07s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/3fa7c33fac9b489b8189b68bc9e0ac88 : FAILURE in 18m 44s\n- designate-devstack-pdns4-manager-model https://zuul.opendev.org/t/openstack/build/215b5eff1ce74ee39b2d8413db326b81 : FAILURE in 18m 03s\n- designate-devstack-pdns4-py35 https://zuul.opendev.org/t/openstack/build/4df623b94f9446b1b21bb8303690b3a0 : FAILURE in 24m 33s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8f6a5c89638348bb852fec5be393da42 : FAILURE in 22m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff40cd8fe10d45f3b8b11ab613a3011a : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0c10851aea0a4086a12dd430823c9ba3 : SUCCESS in 6m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3582dc9b90304230b875f5b93d3dc77f : SUCCESS in 5m 04s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9b7723b506a24b81ad4a27d99a3e3060 : SUCCESS in 6m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6cc4a24d91af427c8595e99eae2e5e3c : FAILURE in 23m 27s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccb9d2ee678943ac8415361f18e8f941 : SUCCESS in 6m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7f5659478c0346e4a84a405c42c55cea94517ca9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-04 17:26:01.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"f8100de90046107c6462bfdd165381492207550d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 18:34:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-devstack-bind9 https://zuul.opendev.org/t/openstack/build/06f96ed3424c48898d9d9966c844c010 : SUCCESS in 47m 17s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/0c8ba40c83934cd4bc0f0d6181d7648d : SUCCESS in 40m 14s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/06ae75aebc9e47c3a87c5633ac82e3fe : SUCCESS in 38m 08s\n- designate-devstack-pdns4-manager-model https://zuul.opendev.org/t/openstack/build/f180b7c6974245ae9785865a3038b2d7 : SUCCESS in 53m 09s\n- designate-devstack-pdns4-py35 https://zuul.opendev.org/t/openstack/build/e1fcf76731e44e76aa7c3a9d293d764c : SUCCESS in 37m 08s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ae0c7bb32716481da337ae582a3eb8f8 : FAILURE in 26m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cab5d3621dc045d485db4a75c6f59ae9 : SUCCESS in 5m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/61fc6e0fa8d04e068260cb00c2bf70fe : SUCCESS in 6m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b0e69970d3204a8eb23ef015973dcb41 : SUCCESS in 6m 50s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ecdfd768af6a45c9b2ab3530c003dfc6 : SUCCESS in 6m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/02eed654f7ef495680154d0ae9d3eaec : FAILURE in 21m 43s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b7d48dffa5846d78d10847f8425993a : SUCCESS in 7m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7a5418a232d117fc84ba6984a6a091d4be4f6729","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-11 16:49:55.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"4ea9f963421156506d5568ee0d80ec0461db9d13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 17:51:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-devstack-bind9 https://zuul.opendev.org/t/openstack/build/29aaa6e510f64cf4af6085a2c0ef4a77 : SUCCESS in 36m 09s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/a10ccff7ca3c42d8b0da19184372c641 : SUCCESS in 35m 53s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/e65aae5b726447dcbd21c0736b67358a : SUCCESS in 35m 35s\n- designate-devstack-pdns4-manager-model https://zuul.opendev.org/t/openstack/build/94b9eba07b6247eb90a16ebc752b4d8b : SUCCESS in 44m 13s\n- designate-devstack-pdns4-py35 https://zuul.opendev.org/t/openstack/build/6ba8045dd81044fb846e1ff1fa9253d0 : SUCCESS in 42m 36s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/92d7f6c7609344629476ea4b0c4c92a6 : SUCCESS in 52m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82510599bde749e983f4e3ddb4866161 : SUCCESS in 7m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f52ecf614baf4a4f81e6683638c925c5 : SUCCESS in 7m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/411f429a7eec4e1796dd41cf74b75928 : SUCCESS in 8m 24s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/fa01d35ae96f436faf483664976f7850 : SUCCESS in 7m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e66f7dd5f08443d89bbfc2a1156523a5 : FAILURE in 21m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29468e28d4364b7f9a4d492483217be1 : SUCCESS in 5m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c582bd4160bf30637ac9286cfe28adc876e32fae","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2020-05-15 18:30:08.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2960b9a24b0f09282bf61e4d1ef75d109a331d20","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2020-05-18 15:05:11.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"68e53a71b49e21281d0638d564ef26a273efd00f","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-05-18 16:14:47.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0a08e9e4bdb3b7bba6c710393c2a4881e7b41e64","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-06-15 19:40:56.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"69f2c3db8ee3d0d59ce0804649d4a27825650a87","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-07-24 22:20:01.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dc8d1ca6dc23cfb9c7b605cc4dd18559d1e1d0bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 22:20:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"505d86435769fd96adf871b745db7864d4c31ba8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 23:25:43.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-devstack-bind9 https://zuul.opendev.org/t/openstack/build/878eee5550e241c79ce6f7b32100cd7c : SUCCESS in 25m 43s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/1c8e670624684bb886121435163d24c9 : SUCCESS in 42m 59s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/3284c6c27589478fa7c36db0b5899f3e : SUCCESS in 33m 28s\n- designate-devstack-pdns4-manager-model https://zuul.opendev.org/t/openstack/build/e96344fc329949568d795e583d9b7678 : SUCCESS in 49m 39s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/25fe5ed9f5c54a6d8b72df509a9ec803 : SUCCESS in 1h 02m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/595ac8bf16ab4a2ca3dc867f559006cd : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/27adab436b664ebf9f13ec720d7e9236 : SUCCESS in 5m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/099a79bd13bd49bda15f20787e6f68e3 : SUCCESS in 7m 42s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4f6a3943947f404d9e5a281fcbdf8022 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":3},{"id":"dd7ad5393aaa3c55422ad3eb79a1e05fe5363b39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 23:25:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"076600dd9107baf567c3dc032a0901dfeda73784","revisions":{"234bfcf4adec94a62367245e926377da1b912ff1":{"kind":"REWORK","_number":1,"created":"2020-04-17 16:05:17.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/66/720766/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/66/720766/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/720766/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/720766/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/720766/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/66/720766/1"}}},"commit":{"parents":[{"commit":"c7987ef4b74962ccdc50b5aad76b6ed973f7e70c","subject":"Merge \"Fix DBDeadLock error resulting into 500\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c7987ef4b74962ccdc50b5aad76b6ed973f7e70c"}]}],"author":{"name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","date":"2018-08-07 09:22:31.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-17 16:05:17.000000000","tz":0},"subject":"Worker should send NOTIFY also to all servers in \u0027also_notifies\u0027 pool settings.","message":"Worker should send NOTIFY also to all servers in \u0027also_notifies\u0027 pool settings.\n\nChange-Id: I0a85c5df5c323d6f636c8cb92ee5201c14cdb306\nCloses-Bug: #1785769\n(cherry picked from commit 6e79c2e0816ca59cff9a8b0eb238e1e5add7b943)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/234bfcf4adec94a62367245e926377da1b912ff1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/234bfcf4adec94a62367245e926377da1b912ff1"}]},"branch":"refs/heads/stable/queens"},"a43b01a5f7a72bad9032745e3ebd08576c1c0b08":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-04 17:26:01.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/66/720766/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/66/720766/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/720766/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/720766/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/720766/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/66/720766/2"}}},"commit":{"parents":[{"commit":"2aaa40d62e636d6b66f79095572227d945c7cc9f","subject":"Pin stable/queens tempest tests to 0.7.0 tag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2aaa40d62e636d6b66f79095572227d945c7cc9f"}]}],"author":{"name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","date":"2018-08-07 09:22:31.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-05-04 17:26:01.000000000","tz":0},"subject":"Worker should send NOTIFY also to all servers in \u0027also_notifies\u0027 pool settings.","message":"Worker should send NOTIFY also to all servers in \u0027also_notifies\u0027 pool settings.\n\nChange-Id: I0a85c5df5c323d6f636c8cb92ee5201c14cdb306\nCloses-Bug: #1785769\n(cherry picked from commit 6e79c2e0816ca59cff9a8b0eb238e1e5add7b943)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a43b01a5f7a72bad9032745e3ebd08576c1c0b08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a43b01a5f7a72bad9032745e3ebd08576c1c0b08"}]},"branch":"refs/heads/stable/queens"},"076600dd9107baf567c3dc032a0901dfeda73784":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-05-11 16:49:55.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/66/720766/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/66/720766/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/720766/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/720766/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/66/720766/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/66/720766/3"}}},"commit":{"parents":[{"commit":"0dae66c1b4c87f054a183a4284f656113d0bed86","subject":"Pin stable/queens tempest tests to 0.7.0 tag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0dae66c1b4c87f054a183a4284f656113d0bed86"}]}],"author":{"name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","date":"2018-08-07 09:22:31.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-05-11 16:49:55.000000000","tz":0},"subject":"Worker should send NOTIFY also to all servers in \u0027also_notifies\u0027 pool settings.","message":"Worker should send NOTIFY also to all servers in \u0027also_notifies\u0027 pool settings.\n\nChange-Id: I0a85c5df5c323d6f636c8cb92ee5201c14cdb306\nCloses-Bug: #1785769\n(cherry picked from commit 6e79c2e0816ca59cff9a8b0eb238e1e5add7b943)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/076600dd9107baf567c3dc032a0901dfeda73784"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/076600dd9107baf567c3dc032a0901dfeda73784"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
