)]}'
{"id":"openstack%2Fdesignate~721036","triplet_id":"openstack%2Fdesignate~master~I4bd0de2f553c43a805670a424c44455d691891ac","project":"openstack/designate","branch":"master","topic":"unittest.mock","hashtags":[],"change_id":"I4bd0de2f553c43a805670a424c44455d691891ac","subject":"Use unittest.mock instead of third party mock","status":"MERGED","created":"2020-04-18 17:08:47.000000000","updated":"2020-06-03 14:46:23.000000000","submitted":"2020-06-03 14:44:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721036-1591195451432-1bb9092c","meta_rev_id":"9c2623eb45258873505d1ed7f3d42a817e802cc4","_number":721036,"virtual_id_number":721036,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-03 14:44:11.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2020-06-01 12:25:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-02 17:58:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":1,"date":"2020-06-03 12:53:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-02 17:58:32.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-06-03 12:53:45.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-06-03 14:44:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e38ff925f8ed98bd8233a672e27856eecb9da602","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-18 17:08:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e6c6a6442ec8f6792f17221c30696707dbd3ef2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 18:31:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e16e2eb3d4bb4d10b66a05556ed13b39 : SUCCESS in 6m 28s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f7c5df3d29a544308117da5d8e1ebafb : SUCCESS in 38m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5bbfa18ef6a942998726bb7eceb39a38 : SUCCESS in 38m 00s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3753951db4b349f5bd6ccff3a3df7a6b : SUCCESS in 1h 08m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2c8671a74f964faa9bfc917eeee348c3 : SUCCESS in 52m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/937130f55a3d4f89b44758ca87e12fcf : SUCCESS in 48m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a976d26470947fe951a0095a21e13f7 : FAILURE in 7m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc99824dbe264e3b81a118ad644aa55b : FAILURE in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcd2c553ab874646a703ce484cfd0f10 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbf72b1d0ce44a4b852d0df7be8a7c34 : FAILURE in 7m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9d8e06d5c69e479b94c21d6285b8acef : FAILURE in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33cd7a133fbd4f1aa3e927117df50adf : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0d6d45868734e538fa8ac4f751d6378 : SUCCESS in 9m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d34273e9a32d4b9b88ec1530de2847ae : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a107143d81b64ece9fe8e29ff9d7bd6a : SUCCESS in 8m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dc5eae2800bf498bbd636dfa94c7996d : SUCCESS in 45m 44s","accounts_in_message":[],"_revision_number":1},{"id":"d5afc49ada37bd4cc5818c67fd53642d6b622326","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-29 16:35:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e03c736ddbf24a16497c593e607b8b2a07030289","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 17:35:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2ba385c963b64cd6af2dc7a7ad622a41 : SUCCESS in 4m 15s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/81ee9c66f1ca4676888796a4f9a886f0 : SUCCESS in 44m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/978c5422ae364df8956bd1abde02834f : SUCCESS in 35m 31s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/dadede8566ef42589f070119e51f8455 : SUCCESS in 53m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f8e801c59a604a149c2a5a311d353f77 : SUCCESS in 38m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c89fd2a0399242dfa85d71317fd636ef : SUCCESS in 42m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/846d1acf555148e8be670152abe1af5e : SUCCESS in 5m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4cfa46e7e3040389696a5a70e6ad456 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28a9d0e6580143f994767dfee898ce19 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea744c76f28d4a9c8fffe9aaa0e332d6 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a8866ad92e9449fb82160e56b45bc50 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70eb9cf12d0343c2913554aa98fd81b2 : SUCCESS in 6m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8e4934c3609c4740ad89a70b31b4b0f4 : SUCCESS in 2m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32c110925cef477ba582785090f020a9 : SUCCESS in 4m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2cf232d0573c400aae7be295f69820b3 : SUCCESS in 36m 58s","accounts_in_message":[],"_revision_number":2},{"id":"0ab9ddfd7f6b4acfd2487859dbd3bd8db959e4d7","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-06-01 12:25:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"56de963a95f988c90bea8c81b8b1db83bf28a168","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-06-02 17:58:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"66ea68b01342bdb2f91e2cf98900ac993da1d9e0","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-06-03 12:53:45.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a654b065aeeaaf943e9b1ed5e87d15f162b2ef84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 12:53:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bd8dccecf129cce395fc7728da3fe0c6ab456aea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 14:44:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b7574c5b4e065a5ade848fc205605105bbbbd08f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 14:44:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/05c77c0b846d49298834e81669f58c8c : SUCCESS in 5m 34s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c2d0afdf63f741eea7fb9d6c47fefc9b : SUCCESS in 39m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6f6f49bafd964731a78d649a4ec58c8b : SUCCESS in 37m 09s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e5e2b5a6f6f149819217906c24e69783 : SUCCESS in 57m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/458d631ab3a04a3086c867ead3442ecb : SUCCESS in 36m 32s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1c2a0e74331243648e5c3420d2400879 : SUCCESS in 48m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5db1bfdb0a2f45ad85ee850a5985aadc : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7d946ab36bd4f50a0c40971c086dd93 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d68a354731284b329036789d8551864b : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aae6821b66674d92b9866c49f52eb01f : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92b8e5f9066d40b488113a7e4a1c2ba3 : SUCCESS in 10m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/01bd781bf23d402d862eeaf86c18a101 : SUCCESS in 2m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21af883d86e04ffdae32840493b7f915 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":2},{"id":"d032157e0497b443dcfdc0e7ea7f46cdce31c5e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 14:46:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/c7b4935b6efe45438a3b3d94e495c7e2 : SUCCESS in 1m 24s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92f38afc76834759bbf33276faed4c8e : SUCCESS in 1m 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6721ae2012264fe094bec6c7d86fdb15 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4b2c732df7dc21fccbf136995846970e80fceeba","revisions":{"3fb762742abb96162773e376a87862b35b97095e":{"kind":"REWORK","_number":1,"created":"2020-04-18 17:08:47.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/36/721036/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/36/721036/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/36/721036/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/36/721036/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/36/721036/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/36/721036/1"}}},"commit":{"parents":[{"commit":"beaf0de79f788622afd02cb07d506422783f70a3","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/beaf0de79f788622afd02cb07d506422783f70a3"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-18 16:52:22.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-18 17:08:36.000000000","tz":-300},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib.\n\nChange-Id: I4bd0de2f553c43a805670a424c44455d691891ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3fb762742abb96162773e376a87862b35b97095e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3fb762742abb96162773e376a87862b35b97095e"}]},"branch":"refs/heads/master"},"4b2c732df7dc21fccbf136995846970e80fceeba":{"kind":"REWORK","_number":2,"created":"2020-05-29 16:35:47.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/36/721036/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/36/721036/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/36/721036/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/36/721036/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/36/721036/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/36/721036/2"}}},"commit":{"parents":[{"commit":"59a2cdc24038db09223f1090c1e732707135dcb8","subject":"Merge \"Cap jsonschema 3.2.0 as the minimal version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59a2cdc24038db09223f1090c1e732707135dcb8"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-18 16:52:22.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-29 16:35:40.000000000","tz":-300},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib.\n\nChange-Id: I4bd0de2f553c43a805670a424c44455d691891ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b2c732df7dc21fccbf136995846970e80fceeba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b2c732df7dc21fccbf136995846970e80fceeba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
