)]}'
{"id":"openstack%2Fdesignate~722368","triplet_id":"openstack%2Fdesignate~stable%2Fussuri~I48094fc85c679aee5dd4179aacc53f16091b3ad7","project":"openstack/designate","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"I48094fc85c679aee5dd4179aacc53f16091b3ad7","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","status":"MERGED","created":"2020-04-23 15:33:34.000000000","updated":"2020-05-08 10:24:00.000000000","submitted":"2020-05-08 10:20:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722368-1588933256518-1288b4e0","meta_rev_id":"d8129f28fac6f8b552e3b89545db7f380747e0e3","_number":722368,"virtual_id_number":722368,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-08 08:01:09.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-08 10:20:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-04-30 05:30:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-04-30 05:30:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-30 05:30:29.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-05-08 08:01:09.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-05-08 10:20:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bd4653059bb6ec90436bd6b0ed79fe5424f9b689","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-23 15:33:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a3058dbf3411200b401bf24058b0404c21b215f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 16:51:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/29e41c678c8645508e530300d82359c5 : SUCCESS in 1h 07m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/507aa7daf8144644a38a1cb77a1ced9f : SUCCESS in 6m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10975940c84d45eea09787c653a28539 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d46bcc602b74aceb04bbb8e42f5a4cf : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3012337620b45cbae1618cbbb9058d6 : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/babe917d8b46439394efc00fc2f30057 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7524ec7a053f413dbb74494836404303 : SUCCESS in 8m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11c200b91da448379a8262641d892747 : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":1},{"id":"7b979cc7162234c3b117651e19383715b2b1a5dd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-04-30 05:30:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"97a53d7419fc0a0c5c248f7f6160840855642e6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 05:30:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"adccea4aaae0046567be2b8e079dbcbb29b54c1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 06:38:48.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"d7fb278df457f2f368e38175d51c768ac64c9400","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-08 08:01:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dd17af17f1a9b0a3450bb74ad3a3391dd65a9a9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 09:12:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6a03e7f676854de6a6c516f171455cc4 : SUCCESS in 42m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/83a8776b7d554143ad920477a6f41a4a : SUCCESS in 38m 52s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4eedc32851c54e3cb7fe0dfdd7f767f3 : SUCCESS in 1h 06m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/714335a16951415182ce83b5b2bb31f6 : SUCCESS in 40m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a1a67bd551db4039b3ce8bb5562f3249 : SUCCESS in 40m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52ccc652078e49d69eb104423d3ac9a4 : SUCCESS in 6m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/811608393af44c578ed3e8c2629471d9 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/284448456c9a47a4bc9e3cd8de0eaa06 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b292165f11a401f955e047091eccdeb : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9d067389a3624df6b4285789d8ce9f0c : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16b6c14f8614465baee1e48d8a000645 : SUCCESS in 7m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a30fb5e3954744709d0747aee2a88661 : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":1},{"id":"495fe75d073bfb2a33be5e3bec302d05b97f0ecb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 09:12:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"251e1faa13c616ee6fe699755b2ccf118c520ee3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 10:20:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e0cb27a598249f2c567eb5f4589f40c1090ca719","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 10:20:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/26c147f9c9b84de68f69032bc0d41d9e : SUCCESS in 42m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/95caad35fde1419f96c56fde0c340145 : SUCCESS in 36m 07s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c6252a573b9e4faeab2816bfec02eafc : SUCCESS in 1h 05m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ecd8fe2aae7f468fa86b165396de3d53 : SUCCESS in 35m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9230c9ab206a424f9f685283c7b85c69 : SUCCESS in 39m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37f92ffa04944768ac3812686c3e41e3 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f13bf7fc184435a9d08797d5300d819 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78cc3eea317440408d4504859dbf971b : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7e24f5b8ae134c0b87f0ee6c67a0ab67 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ce0c235713a4ab391721667c1195b59 : SUCCESS in 8m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9f946266f57414a8d18ef64c454b977 : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":1},{"id":"29be3e574984e93ee608a8ca692127d3e2c7d507","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 10:24:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80579fa05da44805a54638c7d0f14a8b : SUCCESS in 2m 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73cd21967a5a4ac3bef504a1bcfa7e7b : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a61a0c393d1fee9ec4b1f0dc051256d118bb136f","revisions":{"a61a0c393d1fee9ec4b1f0dc051256d118bb136f":{"kind":"REWORK","_number":1,"created":"2020-04-23 15:33:34.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/68/722368/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/722368/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/722368/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/722368/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/722368/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/722368/1"}}},"commit":{"parents":[{"commit":"5a97672d65397ebc80f9e9c2200a4a82a49ead30","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5a97672d65397ebc80f9e9c2200a4a82a49ead30"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-23 15:33:32.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-23 15:33:32.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","message":"Update TOX_CONSTRAINTS_FILE for stable/ussuri\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/ussuri branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I48094fc85c679aee5dd4179aacc53f16091b3ad7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a61a0c393d1fee9ec4b1f0dc051256d118bb136f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a61a0c393d1fee9ec4b1f0dc051256d118bb136f"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
