)]}'
{"id":"openstack%2Fdesignate~725345","triplet_id":"openstack%2Fdesignate~stable%2Frocky~I74d822250c02002d5e4c58b7b85543d1ad6f9173","project":"openstack/designate","branch":"stable/rocky","hashtags":[],"change_id":"I74d822250c02002d5e4c58b7b85543d1ad6f9173","subject":"Pin stable/rocky tempest tests to 0.7.0 tag","status":"MERGED","created":"2020-05-04 17:21:49.000000000","updated":"2020-05-12 19:07:46.000000000","submitted":"2020-05-12 19:07:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725345-1589310466882-5004bf0a","meta_rev_id":"614fc0318b20b9852c2a8da8a46512d3e9a2cacb","_number":725345,"virtual_id_number":725345,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-12 19:07:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2020-05-12 07:47:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-12 17:56:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-12 17:56:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-12 07:47:13.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-05-12 17:56:07.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-05-12 19:07:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c6016ef7c1625c3eff68ca1faafadb2cd7e4f14b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-04 17:21:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35d966f1a8c9cfab3cf10415d02d6ed29d05bf75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 17:22:08.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/designate on branch stable/rocky.  The error was:\n\n  mapping values are not allowed here\n  in \"openstack/designate/.zuul.yaml@stable/rocky\", line 20, column 26","accounts_in_message":[],"_revision_number":1},{"id":"8352a2d0ff6cfb1ed02204355365c3e739506f23","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-04 17:23:40.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"01fc0a283c3ee9f8fc8eb22be33bf8af61ac0811","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 18:12:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cf4acaacc4aa423b82eafe111fee438a : SUCCESS in 42m 39s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/d12ab9a4b69b434da16e56596470358c : SUCCESS in 40m 52s\n- designate-bind9-py35 https://zuul.opendev.org/t/openstack/build/2a936b5fa0a444a2a72591bc1f5ee44d : SUCCESS in 41m 42s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/4065cfc278064f7fb3456ac7e4dcc8ca : SUCCESS in 40m 07s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/19785245f7354440a66f769aa2a6c2bb : FAILURE in 20m 58s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/22cde13f97e44a7798abea557f4c593e : FAILURE in 19m 42s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d25fe0b715e64dfca0be857c36dd660a : FAILURE in 20m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b75215694deb4288a57a7ecf9f3f67ac : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c862ea3e0324547804a483f7ec7ddb8 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bfd9c0dca18a4bf394dedaeff76fd774 : SUCCESS in 5m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/25cb25f6920e41bdad706c7f055f0ee7 : SUCCESS in 9m 26s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/be588089519142789f835ae7c2253a89 : SUCCESS in 5m 59s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/21d89d15d2f7450ca287b54c044d905c : SUCCESS in 32m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09313d63bcb34eb9a86257af5ee60745 : SUCCESS in 4m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09a34bccbed84870b7bcf3f6e6766bca : SUCCESS in 7m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fc2d5e64c873a8b4bc0b0d8630a9b9bec37fc782","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-08 20:34:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"af501e2eb50327acfae4c61daec7b9a8f6294381","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-08 20:47:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"18434c8fd2154f79deb834223e7f43d08ba44967","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 21:43:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d5c12f6048ed49f7b5700bd820fc1778 : SUCCESS in 44m 14s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/7ae9a5403b9a4b93bc00b80026864b45 : SUCCESS in 38m 32s\n- designate-bind9-py35 https://zuul.opendev.org/t/openstack/build/a8720667a6c24cc5a984b4c658a974e9 : SUCCESS in 40m 45s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/e37fed65f9764f32a36b0c93639c8334 : SUCCESS in 41m 51s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/af1b0c546a2c4b62ba3a67c470be2186 : SUCCESS in 40m 55s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/ba8ed538d720405ba4ec4ab5800905aa : SUCCESS in 38m 28s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3ce6e2a275ac4022b51999278174bf2b : SUCCESS in 53m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/155661ab2fd14af3a06a2379c3d2c900 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42151305151743d09c8b2d6906020ffb : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4a333ca54d504702a064c075bf7fcc25 : SUCCESS in 5m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2080d62d71694c968f80c1c736220cae : SUCCESS in 5m 28s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a9fb37d839e8405fbb716058436b9b05 : SUCCESS in 5m 47s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/07e5bc932e8540fc9fb1a86391374fc1 : SUCCESS in 36m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19d3f9b0de2449b8b8deaa5ce702eb90 : SUCCESS in 4m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c80ae74f3a04cbe88510ef41d6794c7 : SUCCESS in 5m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d557da779b053db29f793cbb7b08fda15c302f46","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-05-12 07:47:13.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"5ccc850edcd132abe32c20eba8ee795ec6d7913e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-05-12 17:56:07.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"48e9340b38c5280169c76bcbe94e67820a4f50a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 17:56:23.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"50c91bb05b3449fc081cfe39ad65106681008f74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 19:07:46.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3fbc6a325c6b46a696a77760866549d5 : SUCCESS in 42m 20s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/1d971fbdb3a14ee4ba7b47ec2f3c56d0 : SUCCESS in 43m 21s\n- designate-bind9-py35 https://zuul.opendev.org/t/openstack/build/5a5d73528cdf4bb9b15ac5492a1925fd : SUCCESS in 40m 03s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/d15006d45f364fc1a323e3a19abbc5f3 : SUCCESS in 39m 47s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/1ab3beeb913e40a3b5b05d0f0eb38d2b : SUCCESS in 41m 22s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/9fd2504fbb4b4bd5990b20b46de3b9c9 : SUCCESS in 42m 40s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/999b9dbb6eff4458bdd46e0930e44dc6 : SUCCESS in 1h 03m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/941aa05f316e4be09a2ee5af37f28a7f : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/052788a2a447401f9f7768158dc7be3d : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3519072ab487412b81890051d4d1c13b : SUCCESS in 5m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1176772cf853430c804729226f498747 : SUCCESS in 7m 11s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1286e8cf2f3143c3946d82a94e56f6d6 : SUCCESS in 5m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/551632e266c54bc0a9dabd8cd36e17d9 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":4},{"id":"614fc0318b20b9852c2a8da8a46512d3e9a2cacb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 19:07:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6eb9305fccd7ac71858c93ce4b895088f5e9a151","revisions":{"328126dad46d13db394a6358dd37f3d416ef6bdb":{"kind":"REWORK","_number":1,"created":"2020-05-04 17:21:49.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/45/725345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/45/725345/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/725345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/725345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/725345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/45/725345/1"}}},"commit":{"parents":[{"commit":"34160d0602abe7fbc566191126698d8bac23ea9d","subject":"Merge \"Clean up zone locking\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/34160d0602abe7fbc566191126698d8bac23ea9d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-05-04 17:19:13.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-05-04 17:21:47.000000000","tz":-420},"subject":"Pin stable/rocky tempest tests to 0.7.0 tag","message":"Pin stable/rocky tempest tests to 0.7.0 tag\n\nMaster of designate-tempest-plugin has removed support for python\n2.7, but the stable/rocky designate runs on python2.7.\nThis patch will pin the tempest jobs to use the 0.7.0 tag for\ndesignate-tempest-plugin as it was the last tag that supports\npython2.7.\n\nChange-Id: I74d822250c02002d5e4c58b7b85543d1ad6f9173\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/328126dad46d13db394a6358dd37f3d416ef6bdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/328126dad46d13db394a6358dd37f3d416ef6bdb"}]},"branch":"refs/heads/stable/rocky"},"75b8b97b8b734aa5767cc37aed9c4a0c0094fa6c":{"kind":"REWORK","_number":2,"created":"2020-05-04 17:23:40.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/45/725345/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/45/725345/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/725345/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/725345/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/725345/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/45/725345/2"}}},"commit":{"parents":[{"commit":"34160d0602abe7fbc566191126698d8bac23ea9d","subject":"Merge \"Clean up zone locking\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/34160d0602abe7fbc566191126698d8bac23ea9d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-05-04 17:19:13.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-05-04 17:23:38.000000000","tz":0},"subject":"Pin stable/rocky tempest tests to 0.7.0 tag","message":"Pin stable/rocky tempest tests to 0.7.0 tag\n\nMaster of designate-tempest-plugin has removed support for python\n2.7, but the stable/rocky designate runs on python2.7.\nThis patch will pin the tempest jobs to use the 0.7.0 tag for\ndesignate-tempest-plugin as it was the last tag that supports\npython2.7.\n\nChange-Id: I74d822250c02002d5e4c58b7b85543d1ad6f9173\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75b8b97b8b734aa5767cc37aed9c4a0c0094fa6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/75b8b97b8b734aa5767cc37aed9c4a0c0094fa6c"}]},"branch":"refs/heads/stable/rocky"},"23ff12dc920e7cfe07268809451170846e40e1d1":{"kind":"REWORK","_number":3,"created":"2020-05-08 20:34:41.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/45/725345/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/45/725345/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/725345/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/725345/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/725345/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/45/725345/3"}}},"commit":{"parents":[{"commit":"34160d0602abe7fbc566191126698d8bac23ea9d","subject":"Merge \"Clean up zone locking\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/34160d0602abe7fbc566191126698d8bac23ea9d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-05-04 17:19:13.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-05-08 20:34:40.000000000","tz":-420},"subject":"Pin stable/rocky tempest tests to 0.7.0 tag","message":"Pin stable/rocky tempest tests to 0.7.0 tag\n\nMaster of designate-tempest-plugin has removed support for python\n2.7, but the stable/rocky designate runs on python2.7.\nThis patch will pin the tempest jobs to use the 0.7.0 tag for\ndesignate-tempest-plugin as it was the last tag that supports\npython2.7.\nThis patch also forces the grenade job to use python2.\n\nChange-Id: I74d822250c02002d5e4c58b7b85543d1ad6f9173\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/23ff12dc920e7cfe07268809451170846e40e1d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/23ff12dc920e7cfe07268809451170846e40e1d1"}]},"branch":"refs/heads/stable/rocky"},"6eb9305fccd7ac71858c93ce4b895088f5e9a151":{"kind":"REWORK","_number":4,"created":"2020-05-08 20:47:53.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/45/725345/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/45/725345/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/725345/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/725345/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/725345/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/45/725345/4"}}},"commit":{"parents":[{"commit":"34160d0602abe7fbc566191126698d8bac23ea9d","subject":"Merge \"Clean up zone locking\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/34160d0602abe7fbc566191126698d8bac23ea9d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-05-04 17:19:13.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-05-08 20:47:47.000000000","tz":-420},"subject":"Pin stable/rocky tempest tests to 0.7.0 tag","message":"Pin stable/rocky tempest tests to 0.7.0 tag\n\nMaster of designate-tempest-plugin has removed support for python\n2.7, but the stable/rocky designate runs on python2.7.\nThis patch will pin the tempest jobs to use the 0.7.0 tag for\ndesignate-tempest-plugin as it was the last tag that supports\npython2.7.\nThis patch also forces the grenade job to use python2.\n\nChange-Id: I74d822250c02002d5e4c58b7b85543d1ad6f9173\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6eb9305fccd7ac71858c93ce4b895088f5e9a151"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6eb9305fccd7ac71858c93ce4b895088f5e9a151"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
