)]}'
{"id":"openstack%2Fdesignate~725347","triplet_id":"openstack%2Fdesignate~stable%2Fqueens~I87d2a9de9178b0c8b41d4ac4c8b674b7900b3546","project":"openstack/designate","branch":"stable/queens","hashtags":[],"change_id":"I87d2a9de9178b0c8b41d4ac4c8b674b7900b3546","subject":"Pin stable/queens tempest tests to 0.7.0 tag","status":"MERGED","created":"2020-05-04 17:25:05.000000000","updated":"2020-05-15 19:34:28.000000000","submitted":"2020-05-15 19:34:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725347-1589571268972-63667995","meta_rev_id":"cb45d1c2b009a2757bc67fe9d6b4398d597343aa","_number":725347,"virtual_id_number":725347,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"date":"2020-05-11 18:33:19.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-15 19:34:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":2,"date":"2020-05-15 18:29:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-12 19:01:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":1,"date":"2020-05-15 18:29:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-12 19:01:49.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-05-15 18:29:56.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2020-05-15 19:34:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3f215584f269574be67e138b4383e1943361b5a5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-04 17:25:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf567abf30023bc594f2ac26c9e21c363ce7d377","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 18:16:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-devstack-bind9 https://zuul.opendev.org/t/openstack/build/2e949866385f4a138b66bbef2f6b43d9 : SUCCESS in 43m 12s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/cebe5a4d94b54af7a28a9f7a587b3d5e : SUCCESS in 42m 57s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/4bdfae89dc9c47078e78ecd0957c3657 : SUCCESS in 43m 41s\n- designate-devstack-pdns4-manager-model https://zuul.opendev.org/t/openstack/build/74b34ebe182c438e9684242832719bc7 : FAILURE in 40m 54s\n- designate-devstack-pdns4-py35 https://zuul.opendev.org/t/openstack/build/6ac424b6b4f944f697fb8afe1c113f6f : SUCCESS in 37m 52s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/919f2794f61243dab51fc4366f1fb5f5 : FAILURE in 21m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b2a4a91cf1642acb24792edc9f96d54 : SUCCESS in 8m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1ac870a44a4f44789319fd8cce65322e : SUCCESS in 5m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/213035b68e4a43879212dd875c1c71b1 : SUCCESS in 5m 43s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d05c0dc2dca64eb496676adaeafc7742 : SUCCESS in 5m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/44ead69d9ff04b6ab7b9206c87188050 : FAILURE in 23m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e4d5e8ee2434892b4fd47b8789957b2 : SUCCESS in 7m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"82900d1711d976dcf964b39f22644dc65eab83e7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-08 20:17:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ae21c32b8cda00cc9177124d6a9737272bd77d44","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-08 20:47:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4329c7e2ebeaedd01da84aedb3ee96c2c80e74f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 21:56:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-devstack-bind9 https://zuul.opendev.org/t/openstack/build/0837514d37244f4b9e56dc815512d4d5 : SUCCESS in 31m 14s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/55360a7f6ec94fc2be03d3439dabc02e : SUCCESS in 43m 48s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/c710ba35c8054bb089f302951af5a6be : FAILURE in 46m 38s\n- designate-devstack-pdns4-manager-model https://zuul.opendev.org/t/openstack/build/cc4437e74e3f4ae59e3366a07d22b8f8 : FAILURE in 52m 18s\n- designate-devstack-pdns4-py35 https://zuul.opendev.org/t/openstack/build/8fd30ccfb6364f2faa83f68f38fbad42 : FAILURE in 52m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/43364f6e045b4e0ca95ffe3d29d9f261 : SUCCESS in 1h 01m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf0475806b5647798f3061390726235f : SUCCESS in 6m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4b4eb60a85b240bf8a4da8916ce0c352 : SUCCESS in 4m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7d2794e91d3047a7b58e42e57636c10f : SUCCESS in 5m 38s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/17bc30ca04e747f8b401430e5451e1ec : SUCCESS in 7m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/77eec5ce8ecd4efbb36a6f0e1c891732 : FAILURE in 22m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7563ee0df4704522bbf2f92952d79b71 : SUCCESS in 6m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"15f53ddd3cf6d9daf13d063df4f33f4d644eb03e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-11 16:49:45.000000000","message":"Patch Set 3:\n\nrecheck\nrabbit timeouts","accounts_in_message":[],"_revision_number":3},{"id":"f9fc17e6fdf7e8833fd9a5e8b235730f04d27086","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 18:01:56.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-devstack-bind9 https://zuul.opendev.org/t/openstack/build/fed2f97d4c434695b963a9f15f9d6143 : SUCCESS in 40m 37s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/6eea51afa99e442487e680f9b1a4932c : SUCCESS in 28m 39s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/90fa3f2434044796830a22c99f84546c : SUCCESS in 40m 37s\n- designate-devstack-pdns4-manager-model https://zuul.opendev.org/t/openstack/build/3dd39da34b0a4a56a8621af9a29ba5ef : FAILURE in 42m 00s\n- designate-devstack-pdns4-py35 https://zuul.opendev.org/t/openstack/build/de615504b1d945e78eae5e8f7f95a41f : SUCCESS in 39m 53s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cdddb133c6d94c0bb66578a3fc8af88e : SUCCESS in 1h 01m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a746ffa6b8d54f42bd1e51b3b1650db4 : SUCCESS in 5m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d302685639142e2b068e11789302b43 : SUCCESS in 5m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3968fbc32a004fbb8f8f358dd569fb64 : SUCCESS in 6m 57s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/5f455172bc274d05a8690ac34656c0d6 : SUCCESS in 6m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5a0e1d50b7c04e2994b387a5126c2582 : FAILURE in 27m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/480b322ffdfc48519c57ebd14179449b : SUCCESS in 8m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4ef2caefe7662b0375b7577855c1a3dd306eec23","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-11 18:33:19.000000000","message":"Patch Set 3:\n\nrecheck\ntimeout, but child patch passed, so...","accounts_in_message":[],"_revision_number":3},{"id":"d7ddbc15831cc43a16cb5e1c80cb4a9ba6bcd298","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 19:43:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-devstack-bind9 https://zuul.opendev.org/t/openstack/build/bc17a997fe3d40fd8ee110369fb232e9 : SUCCESS in 39m 51s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/0a9d28e819454a56958d403a113b44bd : SUCCESS in 37m 16s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/6c94e06fab05439a95f84aa79b695655 : SUCCESS in 35m 26s\n- designate-devstack-pdns4-manager-model https://zuul.opendev.org/t/openstack/build/254af11a2bea45ab8d9ade7faabe36a0 : SUCCESS in 42m 35s\n- designate-devstack-pdns4-py35 https://zuul.opendev.org/t/openstack/build/0bd2314d309b4cfdba2bcf38ab6592e9 : SUCCESS in 46m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7d7a2d33814244ec8473ca4e280f61c5 : SUCCESS in 57m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31fc51fb377042c987e34877a74e5825 : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/22b537b111424cf3abbcaf02fb835a23 : SUCCESS in 5m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/255be869c85b454382892b6d3cbec28d : SUCCESS in 7m 12s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/df76fcbe840e46fb8885aaa8debd6c2c : SUCCESS in 5m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/205f9075c0d8492f98269bf37332ef33 : FAILURE in 25m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10c11de4fd39495cbff0453eaebb84eb : SUCCESS in 13m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e88a7921ddd5831e62e0adf76c70f121c821446a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-05-12 19:01:49.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7ddbf9be570884f40ed8efe17ad5950e2f4425a9","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2020-05-15 18:29:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"981bb6a978ed39c0133212abb3a5038c83ae4a85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 18:30:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"889cf642180d04e8a62b7c2729a5a5ba22569c5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 19:34:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-devstack-bind9 https://zuul.opendev.org/t/openstack/build/3b8cce5d9e5f435581770a4655789c07 : SUCCESS in 39m 07s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/80a0e54c0a7a45afb042b32e3c017401 : SUCCESS in 43m 19s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/09a5d13afdcb41d2a7797d827860ebd3 : SUCCESS in 37m 57s\n- designate-devstack-pdns4-manager-model https://zuul.opendev.org/t/openstack/build/38b589e3136e4fc19ee2fdfa89b90ecd : SUCCESS in 48m 04s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c500b91219144368a0a8f8567ef4eab8 : SUCCESS in 58m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b94c0e23d99474f879ca2316cc6a8c9 : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a5e15c27978642d19b10ddd90eedfaa5 : SUCCESS in 6m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8548c2c5e35a42c081a13d110f7cf8ff : SUCCESS in 7m 05s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/19a43c8e37164ab79997760eacb8a02b : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":3},{"id":"cb45d1c2b009a2757bc67fe9d6b4398d597343aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 19:34:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0dae66c1b4c87f054a183a4284f656113d0bed86","revisions":{"2aaa40d62e636d6b66f79095572227d945c7cc9f":{"kind":"REWORK","_number":1,"created":"2020-05-04 17:25:05.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/47/725347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/47/725347/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/725347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/725347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/725347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/47/725347/1"}}},"commit":{"parents":[{"commit":"c7987ef4b74962ccdc50b5aad76b6ed973f7e70c","subject":"Merge \"Fix DBDeadLock error resulting into 500\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c7987ef4b74962ccdc50b5aad76b6ed973f7e70c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-05-04 17:24:29.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-05-04 17:25:04.000000000","tz":-420},"subject":"Pin stable/queens tempest tests to 0.7.0 tag","message":"Pin stable/queens tempest tests to 0.7.0 tag\n\nMaster of designate-tempest-plugin has removed support for python\n2.7, but the stable/queens designate runs on python2.7.\nThis patch will pin the tempest jobs to use the 0.7.0 tag for\ndesignate-tempest-plugin as it was the last tag that supports\npython2.7.\n\nChange-Id: I87d2a9de9178b0c8b41d4ac4c8b674b7900b3546\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2aaa40d62e636d6b66f79095572227d945c7cc9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2aaa40d62e636d6b66f79095572227d945c7cc9f"}]},"branch":"refs/heads/stable/queens"},"b364f32ff7125dca4e66b84a67d6c5afb4453147":{"kind":"REWORK","_number":2,"created":"2020-05-08 20:17:08.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/47/725347/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/47/725347/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/725347/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/725347/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/725347/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/47/725347/2"}}},"commit":{"parents":[{"commit":"c7987ef4b74962ccdc50b5aad76b6ed973f7e70c","subject":"Merge \"Fix DBDeadLock error resulting into 500\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c7987ef4b74962ccdc50b5aad76b6ed973f7e70c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-05-04 17:24:29.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-05-08 20:17:07.000000000","tz":-420},"subject":"Pin stable/queens tempest tests to 0.7.0 tag","message":"Pin stable/queens tempest tests to 0.7.0 tag\n\nMaster of designate-tempest-plugin has removed support for python\n2.7, but the stable/queens designate runs on python2.7.\nThis patch will pin the tempest jobs to use the 0.7.0 tag for\ndesignate-tempest-plugin as it was the last tag that supports\npython2.7.\nThis patch also forces the grenade job to use python2.\n\nChange-Id: I87d2a9de9178b0c8b41d4ac4c8b674b7900b3546\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b364f32ff7125dca4e66b84a67d6c5afb4453147"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b364f32ff7125dca4e66b84a67d6c5afb4453147"}]},"branch":"refs/heads/stable/queens"},"0dae66c1b4c87f054a183a4284f656113d0bed86":{"kind":"REWORK","_number":3,"created":"2020-05-08 20:47:15.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/47/725347/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/47/725347/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/725347/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/725347/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/725347/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/47/725347/3"}}},"commit":{"parents":[{"commit":"c7987ef4b74962ccdc50b5aad76b6ed973f7e70c","subject":"Merge \"Fix DBDeadLock error resulting into 500\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c7987ef4b74962ccdc50b5aad76b6ed973f7e70c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-05-04 17:24:29.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-05-08 20:47:09.000000000","tz":-420},"subject":"Pin stable/queens tempest tests to 0.7.0 tag","message":"Pin stable/queens tempest tests to 0.7.0 tag\n\nMaster of designate-tempest-plugin has removed support for python\n2.7, but the stable/queens designate runs on python2.7.\nThis patch will pin the tempest jobs to use the 0.7.0 tag for\ndesignate-tempest-plugin as it was the last tag that supports\npython2.7.\nThis patch also forces the grenade job to use python2.\n\nChange-Id: I87d2a9de9178b0c8b41d4ac4c8b674b7900b3546\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0dae66c1b4c87f054a183a4284f656113d0bed86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0dae66c1b4c87f054a183a4284f656113d0bed86"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
