)]}'
{"id":"openstack%2Fdesignate~726347","triplet_id":"openstack%2Fdesignate~master~Ic488ccebee6d3206d6b38eededbb27fcb46694a4","project":"openstack/designate","branch":"master","topic":"dnspython","hashtags":[],"change_id":"Ic488ccebee6d3206d6b38eededbb27fcb46694a4","subject":"Switch to dnspython package","status":"MERGED","created":"2020-05-08 11:20:08.000000000","updated":"2020-05-14 14:32:15.000000000","submitted":"2020-05-14 14:26:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726347-1589466419268-d71d2c75","meta_rev_id":"2d9a20eec9662e8e80a5b9d354c1fbc00487716f","_number":726347,"virtual_id_number":726347,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-14 14:26:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2020-05-14 12:01:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2020-05-14 08:21:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2020-05-14 13:19:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-08 16:20:29.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-05-14 12:01:58.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-05-14 13:19:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-05-14 14:26:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b084a74384be9902e8d185021efbc993ae20f3fe","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-08 11:20:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9980d2ca65e1f56038e3eb196b2ce544b4e6fa8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 12:22:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ed875a98f1b344f19489c76b27669973 : SUCCESS in 40m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/249965605ca549b88deb9ec701785b13 : SUCCESS in 32m 40s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f68b470b876e4ea09a38fa15e95ccff5 : SUCCESS in 57m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c7225e499b984eeaae563e5b32dbee5d : SUCCESS in 43m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/aea92c8f115b4775a63bb1eee3798bb5 : SUCCESS in 38m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40f6a0653fd04b218e2ae0fbe4f6bae1 : SUCCESS in 8m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74c416fd38784e5997d9cd502ac4e1a8 : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae921af7170144ba87ccc609c8ae3cf1 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c975709f1e864b16aadccafeebbc5c53 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acbf60c65c5244c784c26256ca085eed : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c953052ec6b3491793bf25a28cc86ed4 : SUCCESS in 10m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a4829333332244629f65d2597172894a : FAILURE in 2m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/304202b8d70b4bbaa90484f912ef31bd : SUCCESS in 39m 52s","accounts_in_message":[],"_revision_number":1},{"id":"eef2c38fec67be7b023c0726132a0e633442865d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-08 16:17:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"98134c0e8186c8c821b4b0793eb41eb585112468","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-08 16:20:29.000000000","message":"Patch Set 2: Code-Review+1\n\nYep, just pushed a patch for this too (abandoned).","accounts_in_message":[],"_revision_number":2},{"id":"45ac8ede3102c6b2dbca068b1e11f5348da7c8f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 17:18:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/acff378005044954a32de4f5ee280245 : SUCCESS in 34m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3b032beabdea4b76845cb4f6a59519ba : SUCCESS in 44m 53s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7528a53a668241858cf9ede91d82fded : SUCCESS in 58m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/92b5e4bebb55459fb3c6b19683d386ca : SUCCESS in 42m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6e344573c096431cbfb5756dedf1f801 : SUCCESS in 37m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f6eb60cf0704dc2a71bdf20d39a3921 : SUCCESS in 6m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a919f3fec6a04e00a28920b72b9b8ff5 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/662e33e830354912b4a4398424ce655e : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16bdef15469f446e9d80b386ca89341c : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/604af57a8fad448e9a9766997eede296 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1674489c8c754e8fa7a3c5c4215fa0b8 : SUCCESS in 7m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/28ec47feb3d64ff2bcd0b75f06591084 : FAILURE in 2m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e06f46f0025540ed8bdb4bb07e42c0c2 : SUCCESS in 37m 29s","accounts_in_message":[],"_revision_number":2},{"id":"59316a6068517018f3da6b631ff3ad27aae28ea5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-08 19:28:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c9a169f953033d1bf3f3eca081ff7377e0312026","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 21:24:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/385e58645a3d48a79ba8aa3eb493ad1f : SUCCESS in 41m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/985f9a05793644a59568e70b868e100a : SUCCESS in 37m 15s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5f2dd2c9667848648ac72e24e09a5464 : SUCCESS in 1h 05m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/934bf0530c5643c0b982c3c50c753f41 : SUCCESS in 39m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/98ef687d42364050ad1f05961da94e7d : SUCCESS in 37m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3256893be86e4cb8a99ea3efcdb8151c : SUCCESS in 8m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2daa9974afc3478a82aa931929d6d8b8 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11327b4e93144b5c9cff8c5a29ff482b : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8b91a488210414b986bca0c29210fda : SUCCESS in 8m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e606b0b88204af7beb00662b035dcf3 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cfb5c7ee6d649468f93fa314bb3afe7 : SUCCESS in 16m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d15e3f9bc19948ffba791f42d67460b7 : SUCCESS in 2m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e1102ffefd7f4ea49fce30a5c701fec4 : SUCCESS in 33m 27s","accounts_in_message":[],"_revision_number":3},{"id":"0a453f7c324ecf5c21409cbcdeacb46318459964","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-05-14 08:21:50.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ee3180b0fc9a0c4b7238d8567c9b90c29e971871","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-05-14 12:01:58.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"66418d4ac73f891fb2fb2cd2c7c583bda24d9298","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-05-14 13:19:44.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"285d3e42efbe6542139e96a56be41153df10ca09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 13:19:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"16327cd5e107bcd88a654e9026685fd63bd2a21b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 14:26:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"bfcf528c02068db89c93931829759879d689fdd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 14:26:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c96d09c50e4d4e488410e27733bfd07b : SUCCESS in 47m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2967d3eb68864b69b5657db3dffdfd45 : SUCCESS in 39m 05s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0885a0eaf6da413494705cb36dc354a0 : SUCCESS in 59m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bf6adeb3933b43149a45c661784dbccd : SUCCESS in 42m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d87043abaae54381a06963f4a4b3b4fd : SUCCESS in 41m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/476668ef80024ed799dcb3b4b7fe372b : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a23900f635894738aef9112637570872 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fb756d39b8440f7a4e16f59b6cea181 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11cbd6cf1e9d45beb055af5d42b65d9c : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f543199006b549e88d36ca6a0ffa32f9 : SUCCESS in 9m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ea69c8d5b482454bb41bd0de6e437b76 : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":3},{"id":"524ced319600c0b2c4cbb232562df0e2632788f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 14:32:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf1f521353dd471cba518e46ad5ad7b6 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"57858d9121f1a0848392a56cb86a958b71deb168","revisions":{"a81e80524e0debf25cd2f6b8902cbb944d8d2fa5":{"kind":"REWORK","_number":1,"created":"2020-05-08 11:20:08.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/47/726347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/47/726347/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/726347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/726347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/726347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/47/726347/1"}}},"commit":{"parents":[{"commit":"814cc0aa5157112da36af8d13c1bb2b5605bc02b","subject":"Merge \"Add Python3 victoria unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/814cc0aa5157112da36af8d13c1bb2b5605bc02b"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-08 11:18:57.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-08 11:18:57.000000000","tz":-300},"subject":"Switch to dnspython package","message":"Switch to dnspython package\n\nThe dnspython3 package is no longer needed as py3 support has been added\nto dnspython, and dnspython3 just installs dnspython now as a\ndependency.\n\nChange-Id: Ic488ccebee6d3206d6b38eededbb27fcb46694a4\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a81e80524e0debf25cd2f6b8902cbb944d8d2fa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a81e80524e0debf25cd2f6b8902cbb944d8d2fa5"}]},"branch":"refs/heads/master"},"6eedad1b64e78f7ff6081516c67811d12bbb1aa9":{"kind":"REWORK","_number":2,"created":"2020-05-08 16:17:36.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/47/726347/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/47/726347/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/726347/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/726347/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/726347/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/47/726347/2"}}},"commit":{"parents":[{"commit":"814cc0aa5157112da36af8d13c1bb2b5605bc02b","subject":"Merge \"Add Python3 victoria unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/814cc0aa5157112da36af8d13c1bb2b5605bc02b"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-08 11:18:57.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-08 16:17:10.000000000","tz":-300},"subject":"Switch to dnspython package","message":"Switch to dnspython package\n\nThe dnspython3 package is no longer needed as py3 support has been added\nto dnspython, and dnspython3 just installs dnspython now as a\ndependency.\n\nDepends-on: https://review.opendev.org/#/c/726353/\n\nChange-Id: Ic488ccebee6d3206d6b38eededbb27fcb46694a4\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6eedad1b64e78f7ff6081516c67811d12bbb1aa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6eedad1b64e78f7ff6081516c67811d12bbb1aa9"}]},"branch":"refs/heads/master"},"57858d9121f1a0848392a56cb86a958b71deb168":{"kind":"REWORK","_number":3,"created":"2020-05-08 19:28:03.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/47/726347/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/47/726347/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/726347/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/726347/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/726347/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/47/726347/3"}}},"commit":{"parents":[{"commit":"814cc0aa5157112da36af8d13c1bb2b5605bc02b","subject":"Merge \"Add Python3 victoria unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/814cc0aa5157112da36af8d13c1bb2b5605bc02b"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-08 11:18:57.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-08 19:27:59.000000000","tz":-300},"subject":"Switch to dnspython package","message":"Switch to dnspython package\n\nThe dnspython3 package is no longer needed as py3 support has been added\nto dnspython, and dnspython3 just installs dnspython now as a\ndependency.\n\nDepends-on: https://review.opendev.org/#/c/726353/\n\nChange-Id: Ic488ccebee6d3206d6b38eededbb27fcb46694a4\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/57858d9121f1a0848392a56cb86a958b71deb168"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/57858d9121f1a0848392a56cb86a958b71deb168"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
