)]}'
{"id":"openstack%2Fdesignate~728583","triplet_id":"openstack%2Fdesignate~stable%2Frocky~Iaeca2be8b38075e3e7e8f79621b4b41cbe9934f7","project":"openstack/designate","branch":"stable/rocky","hashtags":[],"change_id":"Iaeca2be8b38075e3e7e8f79621b4b41cbe9934f7","subject":"Adding the option to set the zone quota to unlimited","status":"ABANDONED","created":"2020-05-15 22:04:18.000000000","updated":"2021-12-24 04:30:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d340f7efa662df50c2f6016b05b4c7881ee67f83","_number":728583,"virtual_id_number":728583,"owner":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-15 23:15:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-05-15 23:34:49.000000000","_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":-1,"date":"2020-05-18 16:40:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-15 23:15:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-18 16:40:18.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"}],"messages":[{"id":"91cb2e4ad03a1c696d03f50b6d57ab2dc46bcca1","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-05-15 22:04:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"100708ba98d8d3c3b1e5b72c36c637f3f5e80f9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 23:15:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/525d9dbf4c7341afa711ab4c18e48c58 : SUCCESS in 39m 23s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/a77de2ac890a4e01a0ed58331ee7c15e : SUCCESS in 42m 42s\n- designate-bind9-py35 https://zuul.opendev.org/t/openstack/build/0427a3115ffd44f4bed87d5c68c28b51 : SUCCESS in 41m 25s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/edfad0ae2c5f4f1bb87d48577a5d4eab : SUCCESS in 43m 00s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/97d87de0c0df4888b0544dc5cf2f2089 : SUCCESS in 39m 22s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/045322c2837f4ba983b917046b01e286 : FAILURE in 39m 02s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d9a4609fb8494d2d8bb5146b52838845 : SUCCESS in 1h 06m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/030d6fa61bbe423ebe16fb83b9a82d16 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36588a63484743708f7e2f7272d08dd8 : SUCCESS in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/429cc735b76f416ea27977743c86df39 : SUCCESS in 7m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/13233d4a26df42ddb405fa91b504ecd1 : SUCCESS in 7m 04s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1aa4176f74aa43a1b1dd9b3bda940942 : SUCCESS in 8m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c20756792333475984b6204f2f0c3498 : SUCCESS in 7m 38s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/99519e7cdc56408fa02e37999ae2da4b : SUCCESS in 31m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4912e200a17d471abd5219b56661e9bf : SUCCESS in 4m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ea7546354b148619ccfc7c5f70accc1 : SUCCESS in 7m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"30c0dfc7824474264f4039a129820a8ef1f0c797","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-05-15 23:34:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2c290f20b39b7a63a768d80c2fe6afdf54f65db5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 00:33:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6e1283f70c954847b718ec10a2a6284a : SUCCESS in 49m 15s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/0bb4cac55d8b4885bc97f66bc8e10bcc : SUCCESS in 49m 59s\n- designate-bind9-py35 https://zuul.opendev.org/t/openstack/build/0c3fa0bc715f46e08b2a2a9601e2f094 : SUCCESS in 49m 26s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/dfb5182240b94fd494600b4850601c7c : SUCCESS in 39m 55s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/2d261a43b9d04f6fbd4fb5dd898e6cb0 : SUCCESS in 45m 20s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/95c6e511eda74e7ba55a0c9f7d424803 : SUCCESS in 37m 32s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5bfa1ebbfe2b4d0d80d9cc9bdbe310ae : FAILURE in 56m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a50e66b688414f62948ba9dd914018b0 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4d933d6f6e64eaeb4d5e55e17d9be0b : SUCCESS in 4m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d89335eab8f04ba1bb3d1d06d4d09c66 : SUCCESS in 5m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d7947a1a90ce4cdeb49d417d761307b7 : SUCCESS in 6m 30s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f038f8a47bf1431e8bcb60da99a63e77 : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0cd211a73574fda9fceb9f0be3ea278 : SUCCESS in 6m 16s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/b41aac56075b476dacf6dd9c49d7edab : FAILURE in 30m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9da7155e1cd4b66926d127f3c56a3d0 : SUCCESS in 3m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c483f7eaadd347aca03b597f62adf00b : SUCCESS in 5m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c2703bf38ec4eaa63b882d10fee02b8e963fd0a3","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2020-05-18 16:40:18.000000000","message":"Patch Set 1: Code-Review-1\n\nThis a feature backport - so under the stable branch policy, we can\u0027t backport it unfortunately","accounts_in_message":[],"_revision_number":1},{"id":"d340f7efa662df50c2f6016b05b4c7881ee67f83","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-24 04:30:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0a9052472ece1f1ba15594fb0b842f1109b9d00e","revisions":{"0a9052472ece1f1ba15594fb0b842f1109b9d00e":{"kind":"REWORK","_number":1,"created":"2020-05-15 22:04:18.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/83/728583/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/83/728583/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/83/728583/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/83/728583/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/83/728583/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/83/728583/1"}}},"commit":{"parents":[{"commit":"6eb9305fccd7ac71858c93ce4b895088f5e9a151","subject":"Pin stable/rocky tempest tests to 0.7.0 tag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6eb9305fccd7ac71858c93ce4b895088f5e9a151"}]}],"author":{"name":"hamza alqtaishat","email":"hamalq@verizonmedia.com","date":"2020-05-07 17:31:18.000000000","tz":0},"committer":{"name":"hamza alqtaishat","email":"hamalq@verizonmedia.com","date":"2020-05-15 22:04:08.000000000","tz":0},"subject":"Adding the option to set the zone quota to unlimited","message":"Adding the option to set the zone quota to unlimited\n\nby assigning negative value for the resource like -1\n\nChange-Id: Iaeca2be8b38075e3e7e8f79621b4b41cbe9934f7\nCloses-Bug: #1876198\n(cherry picked from commit 12d50c7f5bafa113e70569c805d844b449976f36)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0a9052472ece1f1ba15594fb0b842f1109b9d00e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0a9052472ece1f1ba15594fb0b842f1109b9d00e"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
