)]}'
{"id":"openstack%2Fdesignate~728735","triplet_id":"openstack%2Fdesignate~master~I8881b891d062703d8136ee8eb459ca293cba8c35","project":"openstack/designate","branch":"master","topic":"hacking-fix","hashtags":[],"change_id":"I8881b891d062703d8136ee8eb459ca293cba8c35","subject":"Fix hacking min version to 3.0.1","status":"MERGED","created":"2020-05-18 01:28:24.000000000","updated":"2020-05-24 10:27:47.000000000","submitted":"2020-05-24 10:26:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728735-1590315977902-8638809e","meta_rev_id":"6f42b6cb2571cb40f71c3e4b83c582e05dd9acc8","_number":728735,"virtual_id_number":728735,"owner":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-24 10:26:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2020-05-18 11:45:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2020-05-24 09:32:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2020-05-24 09:32:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-18 11:45:48.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-05-24 09:32:15.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-05-24 10:26:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5989e5dfca83973e9efb8cd18d3aa2ebd1a7dfbb","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-05-18 01:28:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"835793e5a2e6941d120af32f89c87da6f72e91f8","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-05-18 01:36:05.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4dc5d0a804f931a1f6ef6bcbf8f16e0dfa8bf36f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 02:34:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cf5c9984d76c47e09b373742085043a1 : SUCCESS in 4m 34s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2ad76cb6d7524b7896d45fbd5aa84cb0 : SUCCESS in 36m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cc7b33829197494f9a2a2ccede4dbe67 : SUCCESS in 36m 43s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/adbc03c44f4f46cc8f4efd4282499f43 : SUCCESS in 52m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bd63d4b0b3cf49d0b22f5bd60b307276 : SUCCESS in 36m 16s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7959b89c409741959343951f661d4850 : SUCCESS in 38m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58dfebdd0ca742849e3cf89e5b1f37a1 : SUCCESS in 5m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbf3dca5f96a4c97931d2239a4a6a22f : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7808aa95d5f432a880fb02c5d479d8a : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21f6b01cda4846e39aaeeba71f7be745 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/568dd4041b08456886e6744baedb431e : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a6df5af663f4d4ab630ae6dff2af3be : SUCCESS in 7m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a6ba149fec494a2eb971983236af374d : SUCCESS in 2m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38a77b2643314487baa970550c32ed22 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"42de47d90eca0ead27a7e92074cfdd905b88388a","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-05-18 11:45:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5f7a5bb95d52721782b746dcb594c2e568a16905","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-05-24 09:32:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"37164c2a9c0f43b7486b26e98937799ae60684c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 09:32:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"945342d43289e617c4ae7d0e8fc128b9e52dd305","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 10:26:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/648b1246552f4bb699fdde8c322431e4 : SUCCESS in 5m 24s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b907eafb25db42c9b67bbc799aa3f501 : SUCCESS in 41m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/28ee861f8e86416b906790b475f58ab5 : SUCCESS in 40m 28s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/54aeb945c9f742a6af3de7e927ffc5d7 : SUCCESS in 52m 23s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7819f957e1b0464ca908645da99579ea : SUCCESS in 39m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0b5f07a399554fd28afc2a346eb0eb64 : SUCCESS in 41m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4dbd297751a40729f677f11fb014c08 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3706e2f513594a0bb931e851c06f0710 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ded993de7b14ae59e022a289d6ac5d7 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b75987323ca74b6b91fd838dc9323383 : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95ed532abe8a4b93ae224e2e93fe7054 : SUCCESS in 10m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/25be268dbb4142bd8e73ebfe0ccc953e : SUCCESS in 2m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba35b69c4690482cba9ecab29bb6f2ab : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":2},{"id":"ca997a42dc8814b2e4b2bc71a1f3eb9e053309be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 10:26:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"aa86b8b840a9a5a3e9f09a187b85b08be24d67e4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 10:27:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/b4bfcdc3b11b4594a4ab33345acc4cfb : SUCCESS in 1m 13s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ebb320fcb214d4cb7d2ced2914963af : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fcd8f97cea7442ab6c6dc923855ea38 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d5ee2ede9341a79a873efbe4279f53a5bb01ef36","revisions":{"33a3a229125403bf65f03950b411c0ab235e149e":{"kind":"REWORK","_number":1,"created":"2020-05-18 01:28:24.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/35/728735/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/35/728735/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/728735/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/728735/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/728735/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/35/728735/1"}}},"commit":{"parents":[{"commit":"9a52ea1cdfad3d208b97cc1a3a08df558bae8d41","subject":"Merge \"Adding the option to set the zone quota to unlimited\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9a52ea1cdfad3d208b97cc1a3a08df558bae8d41"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-05-18 01:24:45.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-05-18 01:28:21.000000000","tz":480},"subject":"Bump hacking min version to 3.0.1","message":"Bump hacking min version to 3.0.1\n\nhacking 3.0.1 fix the pinning of flake8 to avoid bringing in a new\nversion with new checks.\n\nbumping the min version for hacking so that any older hacking versions\nwhich auto adopt the new checks are not used.\n\nChange-Id: I8881b891d062703d8136ee8eb459ca293cba8c35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/33a3a229125403bf65f03950b411c0ab235e149e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/33a3a229125403bf65f03950b411c0ab235e149e"}]},"branch":"refs/heads/master"},"d5ee2ede9341a79a873efbe4279f53a5bb01ef36":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-18 01:36:05.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/35/728735/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/35/728735/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/728735/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/728735/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/728735/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/35/728735/2"}}},"commit":{"parents":[{"commit":"9a52ea1cdfad3d208b97cc1a3a08df558bae8d41","subject":"Merge \"Adding the option to set the zone quota to unlimited\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9a52ea1cdfad3d208b97cc1a3a08df558bae8d41"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-05-18 01:24:45.000000000","tz":480},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2020-05-18 01:36:01.000000000","tz":0},"subject":"Fix hacking min version to 3.0.1","message":"Fix hacking min version to 3.0.1\n\nflake8 new release 3.8.0 added new checks and gate pep8\njob start failing. hacking 3.0.1 fix the pinning of flake8 to\navoid bringing in a new version with new checks.\n\nThough it is fixed in latest hacking but 2.0 and 3.0 has cap for\nflake8 as \u003c4.0.0 which mean flake8 new version 3.9.0 can also\nbreak the pep8 job if new check are added.\n\nTo avoid similar gate break in future, we need to bump the hacking min\nversion.\n\n- http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014828.html\n\nChange-Id: I8881b891d062703d8136ee8eb459ca293cba8c35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d5ee2ede9341a79a873efbe4279f53a5bb01ef36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d5ee2ede9341a79a873efbe4279f53a5bb01ef36"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
