)]}'
{"id":"openstack%2Fdesignate~729514","triplet_id":"openstack%2Fdesignate~master~I0cf71327ca5d9da1d88a826225374c8801849c36","project":"openstack/designate","branch":"master","topic":"reno-openstackdocstheme","hashtags":[],"change_id":"I0cf71327ca5d9da1d88a826225374c8801849c36","subject":"Switch to newer openstackdocstheme and reno versions","status":"MERGED","created":"2020-05-20 06:14:17.000000000","updated":"2020-05-20 20:58:04.000000000","submitted":"2020-05-20 20:55:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729514-1590008135242-0ef704bd","meta_rev_id":"a034dfefa80074e17619a1603fe098db3d5b0861","_number":729514,"virtual_id_number":729514,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-20 20:55:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2020-05-20 15:34:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2020-05-20 19:52:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":1,"date":"2020-05-20 13:13:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2020-05-20 19:52:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-20 13:13:04.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-05-20 15:34:12.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-05-20 19:52:36.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2020-05-20 20:55:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d62919fb66fe83a3f1a3e32cca19a8d8d9fedf6b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-20 06:14:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8961799ee0af52df4b719db4828c7b9fb22cb6df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 07:12:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6f3a7205cde241e69018cc11243378be : SUCCESS in 6m 13s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/782ccc596f824ee2bdd4be04952123b4 : SUCCESS in 45m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/725b14d6d3d64ec9bd9c79e25d8da3cf : SUCCESS in 43m 02s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f5fd087b60364a60a0f6df6e1fe7abb4 : SUCCESS in 54m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/562b3d2a10e6422c9e898c1e660eb5eb : SUCCESS in 42m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3728a32d53cc40b1a0285546cf06c53b : SUCCESS in 42m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50e3d8703a7c43989866556da1c08381 : SUCCESS in 8m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd6d332b234c456dbba914e24fd400e1 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30ee8230e6f74515808d8645b1e97c4f : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d303dfc94401435883b57786a73c9b03 : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93dd29c8540d4a6291b3513893b498f9 : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e16573763bf403e85fb83dbbf0d9118 : SUCCESS in 11m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/27cea0eb740e4ca8a7f11c94b62bc92c : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd45a738b8744284866530fbaa554281 : SUCCESS in 6m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c6de89237e794b91875144ebbe344178 : SUCCESS in 38m 05s","accounts_in_message":[],"_revision_number":1},{"id":"33f098fe8f2291c80d5a1f5afca69578e4f80531","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-20 13:13:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bb4251e282e52483ce70f6890d6fb4ee999010bf","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-05-20 15:34:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e32d45f27afb3f8537d18c2398fa847064e77549","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2020-05-20 19:52:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"982a69810f5a2c0bd7a9425c797894a179d36874","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 19:52:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"706e7d7fc27ff544e0d9fb48343e0fd83b0be10e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 20:55:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b577b3ff0d0a41d7813b08501644db61 : SUCCESS in 5m 12s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/03b87ad9d52741edae6db330f1af4233 : SUCCESS in 45m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bff999ccaad7460c8211514da32ee7fc : SUCCESS in 41m 32s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a8a1d3f86d9649c6a65b1849b72ca08e : SUCCESS in 59m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/15b6c657dab84e35b05fa47cbbc7602d : SUCCESS in 43m 10s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/be8dafed474d44428aaf0e965b24ee37 : SUCCESS in 39m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40b3fc73bc2646ba83f7c8626d5cd619 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c13b1b6c72a4ae19bfd08831d6a3612 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2635a23c60542c2954f944cd1f4c07d : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2e18efe050544489d127ac92226ef6f : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e43c23be4554442b8769ad0e27c1036 : SUCCESS in 9m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/895034c767064a22bec60bf36d8c1a52 : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd4f863a2e38402cbbffb5c0f3783d3c : SUCCESS in 8m 03s","accounts_in_message":[],"_revision_number":1},{"id":"b78d40a35a9aee12193e2bdb53a937ca84b11f35","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 20:55:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"886cb260d6779dbdc5ef30f59f3475de05fab4ac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 20:58:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/197868c638e4470c8253a9c86757463a : SUCCESS in 1m 58s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81f0d8bcedc543cd8608e4194711dca6 : SUCCESS in 2m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5775fe4c06f428fb0a8233a9e013f22 : SUCCESS in 2m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"663e4206436dc1a9993cf3a43c950a22cf38c436","revisions":{"663e4206436dc1a9993cf3a43c950a22cf38c436":{"kind":"REWORK","_number":1,"created":"2020-05-20 06:14:17.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/14/729514/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/729514/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/729514/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/729514/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/729514/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/729514/1"}}},"commit":{"parents":[{"commit":"9a52ea1cdfad3d208b97cc1a3a08df558bae8d41","subject":"Merge \"Adding the option to set the zone quota to unlimited\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9a52ea1cdfad3d208b97cc1a3a08df558bae8d41"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-20 06:10:03.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-20 06:13:38.000000000","tz":120},"subject":"Switch to newer openstackdocstheme and reno versions","message":"Switch to newer openstackdocstheme and reno versions\n\nSwitch to openstackdocstheme 2.2.0 and reno 3.1.0 versions. Using\nthese versions will allow especially:\n* Linking from HTML to PDF document\n* Allow parallel building of documents\n* Fix some rendering problems\n\nUpdate Sphinx version as well.\n\nSet openstackdocs_pdf_link to link to PDF file. Note that\nthe link to the published document only works on docs.openstack.org\nwhere the PDF file is placed in the top-level html directory. The\nsite-preview places the PDF in a pdf directory.\n\nSet openstackdocs_auto_name to use \u0027project\u0027 as name.\n\nChange pygments_style to \u0027native\u0027 since old theme version always used\n\u0027native\u0027 and the theme now respects the setting and using \u0027sphinx\u0027 can\nlead to some strange rendering.\n\nRemove docs requirements from lower-constraints, they are not needed\nduring install or test but only for docs building.\n\nopenstackdocstheme renames some variables, so follow the renames\nbefore the next release removes them. A couple of variables are also\nnot needed anymore, remove them.\n\nAdd openstackdocstheme options to api-ref.\n\nChange-Id: I0cf71327ca5d9da1d88a826225374c8801849c36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/663e4206436dc1a9993cf3a43c950a22cf38c436"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/663e4206436dc1a9993cf3a43c950a22cf38c436"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
