)]}'
{"id":"openstack%2Fdesignate~730373","triplet_id":"openstack%2Fdesignate~master~I4ef18b2b2d2a4d73960d556229d4eedc74ed43ff","project":"openstack/designate","branch":"master","hashtags":[],"change_id":"I4ef18b2b2d2a4d73960d556229d4eedc74ed43ff","subject":"Remove the unused coding style modules","status":"ABANDONED","created":"2020-05-22 16:12:26.000000000","updated":"2021-12-24 04:29:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3bcea699844bbbc01ead2166ac1456b81cd93102","_number":730373,"virtual_id_number":730373,"owner":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-22 17:07:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-22 17:07:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5c1b08742e80d1764f98ee6ce942567bc1f52c0b","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-05-22 16:12:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d64abdb8b2388e22d74dd605ba2b78e827b1a0ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 17:07:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9be28ce9146e4a8f8314158e86aaa131 : SUCCESS in 45m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/05fd9e27040542be9bffb103b99ad4eb : SUCCESS in 48m 58s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ede63d33e3ac4a8ca2d9744fdb2d7897 : SUCCESS in 50m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b0224a6ecb49480ab1cae8f985642a37 : SUCCESS in 50m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b476e613bdee4043bb39080e974d68f9 : SUCCESS in 30m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5d36163aedb426bac3923c3287f2861 : SUCCESS in 10m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef43bf1b84614a8499df6fe6d6728bef : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/223ec4f7bab14992ab524609d0a2a774 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0413e4eb7a34f0f83a40aa6d91313b0 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/115a85a21ae04230904f4d757174472f : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca4e1740a9a64eafa50f675dc9d4af68 : SUCCESS in 11m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/395f9232672644e79cef563867e11308 : SUCCESS in 3m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/119bbe6b99ad4fa4ae8c2a1678f78754 : SUCCESS in 41m 13s","accounts_in_message":[],"_revision_number":1},{"id":"3bcea699844bbbc01ead2166ac1456b81cd93102","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-24 04:29:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7250d977f5b3a59818dcd2dac4e2d2e586fd0d60","revisions":{"7250d977f5b3a59818dcd2dac4e2d2e586fd0d60":{"kind":"REWORK","_number":1,"created":"2020-05-22 16:12:26.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/73/730373/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/730373/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/730373/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/730373/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/730373/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/730373/1"}}},"commit":{"parents":[{"commit":"663e4206436dc1a9993cf3a43c950a22cf38c436","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/663e4206436dc1a9993cf3a43c950a22cf38c436"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-05-22 16:11:50.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-05-22 16:11:50.000000000","tz":480},"subject":"Remove the unused coding style modules","message":"Remove the unused coding style modules\n\nPython modules related to coding style checks (listed in blacklist.txt in\nopenstack/requirements repo) are dropped from lower-constraints.txt\nas they are not actually used in tests (other than pep8).\n\nmore info: https://github.com/openstack/requirements/blob/master/blacklist.txt\n\nChange-Id: I4ef18b2b2d2a4d73960d556229d4eedc74ed43ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7250d977f5b3a59818dcd2dac4e2d2e586fd0d60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7250d977f5b3a59818dcd2dac4e2d2e586fd0d60"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
