)]}'
{"id":"openstack%2Fdesignate~730944","triplet_id":"openstack%2Fdesignate~master~I8a1efc079591da32f9b1cb200b6507e5ed968673","project":"openstack/designate","branch":"master","topic":"cap_jsonschema","hashtags":[],"change_id":"I8a1efc079591da32f9b1cb200b6507e5ed968673","subject":"Cap jsonschema 3.2.0 as the minimal version","status":"MERGED","created":"2020-05-26 19:48:19.000000000","updated":"2020-05-29 12:32:24.000000000","submitted":"2020-05-29 12:30:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730944-1590755446154-60d2867d","meta_rev_id":"ce8483a0f73aa42e35940324bb22b1b5385c8e5d","_number":730944,"virtual_id_number":730944,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"date":"2020-05-27 08:29:08.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-29 12:30:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2020-05-27 11:29:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-27 18:41:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":1,"date":"2020-05-29 11:24:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-27 18:41:40.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-05-29 11:24:58.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-05-29 12:30:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0b9c98df771a3f466805a74c42433e7fbeaeceec","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-26 19:48:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d6f25779f163cede7233cf0c4f92ea1ad677352","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 20:53:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9ff56d0467dd42859e6333411f486fae : FAILURE in 43m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a7cbe2c6459840d0b36641e02f93b872 : SUCCESS in 41m 36s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/51ba621c5ed045c1ac5c4154370c121e : SUCCESS in 1h 03m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/69b1ee80efd1472288ec26c21d2e6e03 : SUCCESS in 39m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2f57c49d05624062b9ea64de47954231 : SUCCESS in 45m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/834c4e63a473403ba3d1b8bf80ac30ba : SUCCESS in 7m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92b7f47b741649229195774bdc5c111a : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5a0c677f1bb4638b2e9c24f0094cadf : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8cb532fd91cf4ac59f9266b5fc0811bf : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a91febd88ad349228567ead21215d57e : SUCCESS in 7m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bac5e899e53a488280de49dc7253f8cd : SUCCESS in 9m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/38df7da94cef420e81a05669185655c3 : SUCCESS in 5m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bc6a4a7de2eb4dcebbf21a4d8ffc2ae6 : SUCCESS in 38m 23s","accounts_in_message":[],"_revision_number":1},{"id":"978494c3289733db9b93352bc608fd884429a160","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-27 08:29:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"afe43276b207d5bfdc2107c4dc7912c0307932f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 09:59:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/95729dd114a7439ea709bdcc0c305a43 : SUCCESS in 57m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bf47e1e285cd45c8a234bd5ca7d2d632 : SUCCESS in 39m 16s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4ac125e45bdc4035bb51598977e0fca6 : SUCCESS in 1h 23m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/376a8c0f13c74022a94af86cd7020cb6 : SUCCESS in 39m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3301bcb009a945da816623dd4a614b81 : SUCCESS in 37m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57a62ee6b28c477b9fb3109192eb49d2 : SUCCESS in 6m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2cde20602fa9492da5b68c334e9ffe49 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4711151d61af47bb8df1d5de0d4f43cd : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90fb7963ee89415c9545a776ee9321f2 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ca02f25c7e64522b8f259a8e5403123 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cc7591ecd324f09b3de873e917c9c73 : SUCCESS in 7m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5dfd546a825542eeb6eb26f9097fb65f : SUCCESS in 3m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4d5082eba69844d9b6de2039139bf24b : SUCCESS in 37m 32s","accounts_in_message":[],"_revision_number":1},{"id":"7ba86b88f36db8b59971248f3cb3d4283b70fb3f","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-05-27 11:29:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a9468538c311edd4a69f4574038cf22041d6fbf4","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-05-27 18:41:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"55ba0c9012a0169354c9aa1d61995c0dc8f8b090","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-05-29 11:24:58.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"640dc44d02bf7caa51a77b88ca49aae032a27264","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 11:25:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7d02bd0261c0973b516635264a874e2335ec968e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 12:30:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2441c887ff8949928461aaaf85d20633 : SUCCESS in 43m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/43db53333e1043018951beaae8da2f9f : SUCCESS in 41m 37s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9abee1183139446c9d0c278fbba24b99 : SUCCESS in 58m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2f7e0adcb73d4d6dac4ec272de08420c : SUCCESS in 34m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/69a2c148f2ed49469cc0c238bce3dad9 : SUCCESS in 35m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a16e83da6c9e4846aa96c6d19b9957d4 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ebabc6205bf485a9941d9f6fbe98bae : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d77e15aa75fc4b619e793bdf12f43662 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a64a80f20254d2593130fdb27b52596 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc4a07984c3e4ac5814291e0c89a877d : SUCCESS in 12m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6e3b7fc285a94b70b79f7cdf7f29c653 : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":1},{"id":"d30fae54e3001fd4934943c0a620a40b6d9d0919","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 12:30:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"382772121039b2618278570ad263cf42087ff3d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 12:32:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9424416b796c4d2198fda318b45383fa : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0a411a5097ef8a9c881405a654cd881b93fc7c77","revisions":{"0a411a5097ef8a9c881405a654cd881b93fc7c77":{"kind":"REWORK","_number":1,"created":"2020-05-26 19:48:19.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/44/730944/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/44/730944/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/730944/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/730944/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/730944/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/44/730944/1"}}},"commit":{"parents":[{"commit":"7428bcef5b261f7606ad1e3d7fa31d85d4b76f76","subject":"Add py38 package metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7428bcef5b261f7606ad1e3d7fa31d85d4b76f76"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-26 19:48:12.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-26 19:48:12.000000000","tz":120},"subject":"Cap jsonschema 3.2.0 as the minimal version","message":"Cap jsonschema 3.2.0 as the minimal version\n\nPrevious versions of jsonschema (\u003c3.2.0) doesn\u0027t support python 3.8 [1].\nPython 3.8 is part of the victoria supported runtimes [2] so we now force\nto use jsonschema version 3.2.0 to avoid issues, remove ambiguity and ensure\nthat everything works with python 3 in general.\n\n[1] https://github.com/Julian/jsonschema/pull/627\n[2] https://governance.openstack.org/tc/reference/runtimes/victoria.html#python-runtimes-for-victoria\n\nChange-Id: I8a1efc079591da32f9b1cb200b6507e5ed968673\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0a411a5097ef8a9c881405a654cd881b93fc7c77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0a411a5097ef8a9c881405a654cd881b93fc7c77"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
