)]}'
{"id":"openstack%2Fdesignate~732408","triplet_id":"openstack%2Fdesignate~master~I00ae6227edf21007c39a7d55221898ae5c10dda2","project":"openstack/designate","branch":"master","topic":"bug/1880230","hashtags":[],"change_id":"I00ae6227edf21007c39a7d55221898ae5c10dda2","subject":"pdns4 backend: check if zone exists before attempting delete","status":"MERGED","created":"2020-06-01 15:49:52.000000000","updated":"2020-10-14 23:39:06.000000000","submitted":"2020-06-02 17:55:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732408-1591120514564-2f67a625","meta_rev_id":"391bc0ae865dd95a80429a564c1b8af69eb74d43","_number":732408,"virtual_id_number":732408,"owner":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-02 17:55:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1998,"name":"andrewbogott","username":"andrewbogott","inactive":true},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2020-06-02 16:23:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2020-06-02 16:17:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1998,"name":"andrewbogott","username":"andrewbogott","inactive":true},{"value":2,"date":"2020-06-02 16:43:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2020-06-02 16:51:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1998,"name":"andrewbogott","username":"andrewbogott","inactive":true},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2020-06-02 16:17:41.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1998,"name":"andrewbogott","username":"andrewbogott","inactive":true},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1998,"name":"andrewbogott","username":"andrewbogott","inactive":true},{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-01 15:49:52.000000000","updated_by":{"_account_id":1998,"name":"andrewbogott","username":"andrewbogott","inactive":true},"reviewer":{"_account_id":1998,"name":"andrewbogott","username":"andrewbogott","inactive":true},"state":"REVIEWER"},{"updated":"2020-06-02 16:23:21.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-06-02 16:43:06.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-06-02 16:51:51.000000000","updated_by":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"reviewer":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"state":"REVIEWER"},{"updated":"2020-06-02 17:55:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"40ee9e941129ebf4052bb9da271d62cd22290372","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2020-06-01 15:49:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92ac5d3d36601b1391ecefad0453fc3be5bc05b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 17:10:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a7a288f8e5f14a3d9a463485e56cb029 : SUCCESS in 32m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b61f543fc4b34eef9b485b3d3bfa833e : TIMED_OUT in 1h 13m 23s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8539ea27a5474aa1ad92cadc4173e7e1 : FAILURE in 1h 07m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9f0b810a52474d38bbadd899fb9dcfe2 : FAILURE in 1h 14m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c3e140f564774b39a0ae7d836e2d4f38 : SUCCESS in 41m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69579bb8fc164643824e8ac345234744 : FAILURE in 6m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6b560dd44dc4f2b8e7f0d8960abb9ec : FAILURE in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb4f19810a38494bbc03f86d7921f690 : FAILURE in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5cbbe82f8f84bfdb72ea9302a83980b : FAILURE in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57e5a21df6f147198f5de3bfef2fe316 : FAILURE in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e238b8fb674492eb4de7826dcf8ec0c : SUCCESS in 7m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a905fcefcd2d44e3a5c5e0d93d0d1c69 : SUCCESS in 37m 44s","accounts_in_message":[],"_revision_number":1},{"id":"a9d6236b7a3c926c16c8e7ac0ec76ede04097297","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2020-06-01 18:59:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9fad1d2636c2ea8cf8bf65f8c195879b185afdc7","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2020-06-01 19:14:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"63d9c8a1e42dcf298589192e18d88e3b50779f87","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2020-06-01 19:15:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0e8c01309fc005955bdae698616022357534af22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 20:16:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/37e0554dcbd3406ba5d4c276216ca8c8 : SUCCESS in 37m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2b01923da4024699b940867f84194981 : SUCCESS in 34m 56s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1c390fdfa17b4b6da1863b545449c11f : SUCCESS in 59m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7e0dabea5b4241099eab707e6553f68e : SUCCESS in 45m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bb33d75d7f1c49aab49812bfc37541a6 : SUCCESS in 35m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c4224e9d3f6496db9495330b44b65b4 : FAILURE in 5m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70e9dd67513347909d58749dfdc09126 : FAILURE in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5131271eb35146699a45d876e4d69c8b : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9bbe60ae93a8459f800212a543859545 : FAILURE in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e62ed2b66ff246a896f58d46d4bf82d6 : FAILURE in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6f36d17c7d546bc8453f8021d4983a3 : SUCCESS in 7m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3347bed3bc464883a8a761cf52c1525c : SUCCESS in 30m 19s","accounts_in_message":[],"_revision_number":4},{"id":"d3474e95652cc454f9af4cb355e49eb83d868c6c","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2020-06-02 16:04:39.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"03e910c15c22db28bbccd479cf09858d4dbf39b8","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2020-06-02 16:17:41.000000000","message":"Patch Set 5: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"dbb12b25dd60a8d00b10a2fe8465d2240286926e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-06-02 16:18:02.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"dd9567158bc736fdcdb3ec174c28cf45a70b69f6","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-06-02 16:23:21.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"b493e44741af69523b20cf9a957924b1e9322abd","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-06-02 16:42:41.000000000","message":"Patch Set 5: -Code-Review","accounts_in_message":[],"_revision_number":5},{"id":"dcde4ba8185f3f02c3a8cefff0bfcf29ae8d05bf","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-06-02 16:43:06.000000000","message":"Patch Set 5: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":5},{"id":"2a577a26a95b875d7a15d0550de997465dd5feb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 16:50:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/060f593813784f9ca4f576cd06df3ca3 : SUCCESS in 26m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/96fdf572eb3a4e53a8699c6957dc0c01 : SUCCESS in 44m 57s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1520b08d543a41ecb800185982c8d78a : SUCCESS in 39m 53s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/df843ceb5e3d43fe8f73528d053c7512 : SUCCESS in 29m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/610f9828d40a43cfba72b419b6a93491 : SUCCESS in 27m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9d1541a131d4aa998bb96d6ad145acd : SUCCESS in 5m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e876d870382b43209a49eb713ca374a0 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0aaba703e10415597ca5a46b9decf43 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47f5e150a0474dad8a005ca936b500e1 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a42b52318df74e00b51888c790dbfeab : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d19b0e5ac90a4ffdb37859499b43a8e4 : SUCCESS in 8m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5ddc21aa00da40339bf201cacab2bc77 : SUCCESS in 32m 33s","accounts_in_message":[],"_revision_number":5},{"id":"15fae079cd0b5f8031d55191f29b02f4006502fb","author":{"_account_id":8099,"name":"Graham Hayes","email":"gr@ham.ie","username":"graham"},"date":"2020-06-02 16:51:51.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c412f693a94810abbfcdf2a0920e6fa0c2762067","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 16:52:03.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"94ff4ccf9e7ceb8f788f0ad953eda95b5fb4d15c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 17:55:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"909d656f2eeb42d345290b074625e74d96387652","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 17:55:14.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8383575994944aafa21a1d9a04916838 : SUCCESS in 44m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ff620af7a95b4aa1a21ff3a48aeda793 : SUCCESS in 42m 03s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e27ddd919bdc4f959ec8777f873c9e47 : SUCCESS in 1h 01m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fb6368f6c1954dbdacf58443951c67f9 : SUCCESS in 43m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/284d8c1393094265aed1835fd431732b : SUCCESS in 39m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9118771ae3cd42d5af751375a490afba : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/604a6bc5ace74e0f9318b5ad7f0c55ba : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a55d65264764701a466c54c712eadd0 : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1eca70f339e44f68da7f88687f0dcc2 : SUCCESS in 8m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d83dfa24980e44919ef482db4fa2bdf2 : SUCCESS in 13m 24s","accounts_in_message":[],"_revision_number":5},{"id":"86c3a0257a75519add7dce36851682dc4e193c9a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 17:56:40.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/425b13b9a4e04bd08e2ef1e08487ae54 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":5},{"id":"4e4acf9b40d9f38c9f0d7f8770d6b2b3b3b36107","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-14 23:38:33.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 9a1bb9d34620c1f6245c442215a2143c322ef565","accounts_in_message":[],"_revision_number":5},{"id":"9f9b11fb62d5044b8ef3857c14fd8d144091ac8b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-14 23:39:06.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 0dd12f3746e52dba2e6cc78984525845c473e836","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5acba9443e2fc1f6e78f9a32cd85eb168d2a7cb2","revisions":{"9a113b7e445d2d5c00dec73ed5623c6935a740fd":{"kind":"REWORK","_number":1,"created":"2020-06-01 15:49:52.000000000","uploader":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"ref":"refs/changes/08/732408/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/732408/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/732408/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/732408/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/732408/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/732408/1"}}},"commit":{"parents":[{"commit":"59a2cdc24038db09223f1090c1e732707135dcb8","subject":"Merge \"Cap jsonschema 3.2.0 as the minimal version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59a2cdc24038db09223f1090c1e732707135dcb8"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2020-06-01 15:04:12.000000000","tz":-300},"committer":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2020-06-01 15:49:51.000000000","tz":-300},"subject":"pdns4 backend: check if zone exists before attempting delete","message":"pdns4 backend: check if zone exists before attempting delete\n\nThis should prevent us from obsessing over zones that are in the\ndesignate DB but not in the pdns backend due to races/incomplete\ntransactions/etc.\n\nBug: 1880230\nChange-Id: I00ae6227edf21007c39a7d55221898ae5c10dda2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9a113b7e445d2d5c00dec73ed5623c6935a740fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9a113b7e445d2d5c00dec73ed5623c6935a740fd"}]},"branch":"refs/heads/master"},"0546b937c6eef8cd2c85cd52a89c140c63c55597":{"kind":"REWORK","_number":2,"created":"2020-06-01 18:59:56.000000000","uploader":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"ref":"refs/changes/08/732408/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/732408/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/732408/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/732408/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/732408/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/732408/2"}}},"commit":{"parents":[{"commit":"59a2cdc24038db09223f1090c1e732707135dcb8","subject":"Merge \"Cap jsonschema 3.2.0 as the minimal version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59a2cdc24038db09223f1090c1e732707135dcb8"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2020-06-01 15:04:12.000000000","tz":-300},"committer":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2020-06-01 18:59:52.000000000","tz":-300},"subject":"pdns4 backend: check if zone exists before attempting delete","message":"pdns4 backend: check if zone exists before attempting delete\n\nThis should prevent us from obsessing over zones that are in the\ndesignate DB but not in the pdns backend due to races/incomplete\ntransactions/etc.\n\nBug: 1880230\nChange-Id: I00ae6227edf21007c39a7d55221898ae5c10dda2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0546b937c6eef8cd2c85cd52a89c140c63c55597"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0546b937c6eef8cd2c85cd52a89c140c63c55597"}]},"branch":"refs/heads/master"},"df0f51b5b4d530747d6cf5cd78cfd99755371f80":{"kind":"REWORK","_number":3,"created":"2020-06-01 19:14:16.000000000","uploader":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"ref":"refs/changes/08/732408/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/732408/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/732408/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/732408/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/732408/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/732408/3"}}},"commit":{"parents":[{"commit":"59a2cdc24038db09223f1090c1e732707135dcb8","subject":"Merge \"Cap jsonschema 3.2.0 as the minimal version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59a2cdc24038db09223f1090c1e732707135dcb8"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2020-06-01 15:04:12.000000000","tz":-300},"committer":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2020-06-01 19:13:42.000000000","tz":-300},"subject":"pdns4 backend: check if zone exists before attempting delete","message":"pdns4 backend: check if zone exists before attempting delete\n\nThis should prevent us from obsessing over zones that are in the\ndesignate DB but not in the pdns backend due to races/incomplete\ntransactions/etc.\n\nBug: 1880230\nChange-Id: I00ae6227edf21007c39a7d55221898ae5c10dda2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/df0f51b5b4d530747d6cf5cd78cfd99755371f80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/df0f51b5b4d530747d6cf5cd78cfd99755371f80"}]},"branch":"refs/heads/master"},"6c828170f0e042290f0d09f847edcb87870df645":{"kind":"REWORK","_number":4,"created":"2020-06-01 19:15:40.000000000","uploader":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"ref":"refs/changes/08/732408/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/732408/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/732408/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/732408/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/732408/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/732408/4"}}},"commit":{"parents":[{"commit":"59a2cdc24038db09223f1090c1e732707135dcb8","subject":"Merge \"Cap jsonschema 3.2.0 as the minimal version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/59a2cdc24038db09223f1090c1e732707135dcb8"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2020-06-01 15:04:12.000000000","tz":-300},"committer":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2020-06-01 19:15:37.000000000","tz":-300},"subject":"pdns4 backend: check if zone exists before attempting delete","message":"pdns4 backend: check if zone exists before attempting delete\n\nThis should prevent us from obsessing over zones that are in the\ndesignate DB but not in the pdns backend due to races/incomplete\ntransactions/etc.\n\nBug: 1880230\nChange-Id: I00ae6227edf21007c39a7d55221898ae5c10dda2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6c828170f0e042290f0d09f847edcb87870df645"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6c828170f0e042290f0d09f847edcb87870df645"}]},"branch":"refs/heads/master"},"5acba9443e2fc1f6e78f9a32cd85eb168d2a7cb2":{"kind":"REWORK","_number":5,"created":"2020-06-02 16:04:39.000000000","uploader":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"ref":"refs/changes/08/732408/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/732408/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/732408/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/732408/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/732408/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/732408/5"}}},"commit":{"parents":[{"commit":"b9c4e9962ea2ef64b0cfa1348badcc9742adb784","subject":"Merge \"Fix some spelling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b9c4e9962ea2ef64b0cfa1348badcc9742adb784"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2020-06-01 15:04:12.000000000","tz":-300},"committer":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2020-06-02 16:03:21.000000000","tz":-300},"subject":"pdns4 backend: check if zone exists before attempting delete","message":"pdns4 backend: check if zone exists before attempting delete\n\nThis should prevent us from obsessing over zones that are in the\ndesignate DB but not in the pdns backend due to races/incomplete\ntransactions/etc.\n\nBug: 1880230\nChange-Id: I00ae6227edf21007c39a7d55221898ae5c10dda2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5acba9443e2fc1f6e78f9a32cd85eb168d2a7cb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5acba9443e2fc1f6e78f9a32cd85eb168d2a7cb2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
