)]}'
{"id":"openstack%2Fdesignate~733506","triplet_id":"openstack%2Fdesignate~master~Ia502bdf69650b0ac24fda657b849a061f1172086","project":"openstack/designate","branch":"master","topic":"reset_conf","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2021-12-31 06:16:32.000000000","reason":"Change was abandoned"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2021-12-31 06:16:32.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia502bdf69650b0ac24fda657b849a061f1172086","subject":"Always reset CONF when starting the wsgi app","status":"ABANDONED","created":"2020-06-04 09:28:47.000000000","updated":"2021-12-31 06:16:32.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"24e60ea1826ce3033d14bfcb4bcd9fd2bf9fb87b","_number":733506,"virtual_id_number":733506,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-12-24 05:34:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-08 15:55:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-09 06:53:26.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-07-09 08:12:07.000000000","updated_by":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"reviewer":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"state":"REVIEWER"},{"updated":"2020-07-13 21:11:31.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-07-22 16:36:11.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-07-28 14:46:21.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"}],"messages":[{"id":"5b934317b8d7d9dcfa9c1103dc06208ceaf1e5e8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-04 09:28:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2277574011b21467deefb6ac036cc47153273408","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-04 09:40:15.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9a61045bd311f1f057ef5118441aab329dc7ea6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 10:40:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e768d9f375314eedbd820ea09fe8097a : FAILURE in 29m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4545f323ef434dc3a3391ad96035b7f1 : FAILURE in 27m 54s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f786f8e646474d0c824c3182511e8e69 : FAILURE in 57m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/20514926a54b4eefbf8aee553b263378 : FAILURE in 39m 29s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3b1065bc1cd04d6ebfa56168f70f5721 : FAILURE in 22m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0e2d44f555544d381866357f5819f6e : SUCCESS in 6m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/926217dd55e44ef9b919a3e3fb45d43f : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea3b7a92850e48d0b50cdda373f9efc2 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c84b669450a842608e00dec2e5b99d6b : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67e531953b324d5c8fa6ad02cdcb446e : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcef0751647944beb61e0e67e74caeef : SUCCESS in 7m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/72a56323094149bfb77022045ae52f38 : FAILURE in 20m 39s","accounts_in_message":[],"_revision_number":2},{"id":"4b6f1ee870c00ee4fd59c091810ff79a5300408d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-04 12:03:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7c8a5ec75bae7cd623aff5c8572bc0124b03b18f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 13:02:36.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3c0b3178621a49eca241790793444137 : FAILURE in 38m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4fafdeffd24448aa8eca71975ff2d831 : FAILURE in 37m 52s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5b2907881d6446458150a68deb0a6564 : FAILURE in 54m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cb02230226f64e1b8e1b566598c03c6e : FAILURE in 35m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b944cb0b16f04aea963bde421727c4de : FAILURE in 38m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c442b72e5ced4f3fba639973b2e777bc : SUCCESS in 8m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f535cc0673124cda98efb29e2af14fb2 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94dd21ac922d4cf59df440fc07af8bb4 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/950e77f8549c41e4abd6e5726d93898c : SUCCESS in 7m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74e69de25476419e91c0bd3551cb1cf9 : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd56789339894c1b81795554b71327dc : SUCCESS in 10m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bccb32884fd84033a7ef3914d77214df : SUCCESS in 38m 17s","accounts_in_message":[],"_revision_number":2},{"id":"0be0643fd76a947ca3c23f2e85084ca7e646e0fb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-04 13:48:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4461b8467e62b32d01c4039f764b148fad69d4c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 14:59:22.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3ace5193f0b94a988812c746aa4c02f9 : FAILURE in 45m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2630147daab3434d8c4e1587ea84c14a : FAILURE in 41m 15s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/620c1d68ecdf48e4924645216a8d091d : FAILURE in 1h 04m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f3c936d2ad6c4cd7a68bdbf60850d654 : FAILURE in 45m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/99634e9ebf59477dafdbba8043eb2c5f : FAILURE in 36m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/387ff399acce410f933c237ed3d1f1e0 : SUCCESS in 10m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b37b245c47e943ef9ed920cbd279ed8e : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca77aed2b10a49e19ea1474689682e39 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a03c739e78d4b08820dee7eae8f985a : SUCCESS in 9m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4f3c63b53084119aeaf58e32afb3f9e : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5657cb1197d446b98c7396af8bf604c7 : SUCCESS in 8m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/07f982950dc64f82b243a6c23b273573 : SUCCESS in 40m 30s","accounts_in_message":[],"_revision_number":2},{"id":"f039d185ffcf00709e51b1ea22ec7c885a73e8e7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-08 14:46:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bd54c2c76e0ea68310dd5e1a8e75c9d49240926f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 15:55:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e5f9c45dfd9649b0bc656415f022a59c : FAILURE in 39m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fa92f641967c4452802137ce63b79a8a : FAILURE in 27m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4e4884ea6f59432cbac4ba07baf86fac : FAILURE in 1h 06m 49s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6df6867ab77b40fc84098988c53d033f : FAILURE in 39m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4a4a7dd6a86f4672a48503c43587bc5b : FAILURE in 41m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22356a26985f4709a0b3ac75b761f656 : SUCCESS in 8m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9b0c5196f0246699ca82d7adf3b823d : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5d61a7e2e2e42d39b09151570080317 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ede4ad946c754d748a7019270da150db : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2154c6b245b044d4b27723a486f84440 : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1675c45dabcb47bfb4fcfd920e734c74 : SUCCESS in 9m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f247872da622446cb2afdd9cd5534499 : SUCCESS in 35m 42s","accounts_in_message":[],"_revision_number":3},{"id":"31520133d8b2703983df116abea6d9a3c7c9dcd1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-24 08:17:41.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ad4585755f75742affd6e04622eac8f2376df1b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 09:25:25.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/abc2239750bb4bc286182ff3e7d616aa : FAILURE in 41m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7b957fbee5cd4275b1550cf6aed4c1c8 : FAILURE in 29m 48s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/89ac00a3001c447fb8b8bf58759996e2 : FAILURE in 59m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bf0be842fb2447b1a86e82cc64f2ead3 : FAILURE in 35m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c8ddef2516b04d31907bf0eda461a573 : FAILURE in 39m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd40407cdb7b45aeb4a0252de60ae517 : SUCCESS in 5m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/961306144c62449b813407ea528b3697 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dd61b0e88df46679e2e3d119d7240b4 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf88274cbb334c6cb345c8d92c6096e9 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07c474a75b504ac19987ff0cbf0561ad : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bd0cf6cc36044e9af3a7e1ca4fbe8aa : SUCCESS in 9m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b2a9acfb098241379b8ebeb76dedcd1e : SUCCESS in 34m 48s","accounts_in_message":[],"_revision_number":3},{"id":"896e48fa21feaf77a4ed0ace988807ed29179cc2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-29 07:48:05.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"bfef1ee45e157386b92f9d45a5454841e1831b66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 08:55:47.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/85fd5dc75e694315b1fad26108a3e4e2 : FAILURE in 47m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/48d28d2cef3d41f8aab8d04e751f992c : FAILURE in 38m 55s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3df65e6e0d324d06917301c7eae69108 : FAILURE in 58m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a1f7ebe86d524282a80ac0be3d36860a : FAILURE in 44m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8928115fd9284898826f6c459ebc5463 : SUCCESS in 40m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24a30471fe9c47dc9b432fdf4e3652d8 : SUCCESS in 7m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a91db639d0442edbbccada0d6d628f2 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8776d231896943cfb413208f7b68c406 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7948effe8d4043ad8dfb193d82ff3d9e : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a25df18abf804c8ba0fc594b972933ff : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e62445d7549840409eae52174317dff8 : SUCCESS in 7m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/61be8ce4c7794c59bcb371c86edea9fa : SUCCESS in 37m 26s","accounts_in_message":[],"_revision_number":3},{"id":"f4eafbe541a5afe155d17130080ac9ad602a7e29","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2020-07-06 03:02:24.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1d12b8b29dbb1e4fe52134a7b1fd4c723c65e87c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 03:44:15.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/80dd66402f524391a5ed420cc5d6ce6c : SUCCESS in 34m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ca75c3ec695a4bcb865d750a621a7ba0 : FAILURE in 32m 48s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/60051696e5ab4a56ae54a1c5cc1ff670 : SUCCESS in 39m 49s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0e179c263d4547919250ba3eb81d6c9e : FAILURE in 41m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cd741426150342cbb38bee47a813fb94 : FAILURE in 40m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75384caefeca4d5aba4f26e10a794f58 : SUCCESS in 4m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/119065217a154cce96fc5c75f6e4bf69 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b95ecfdebba4d3a8aab893af75e1fb7 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f455000199d4c08a6271fa213260ef6 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ec02d07913f48d4aaf05fd38e90ad45 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83f85bd96184480aadfc7826338e696e : SUCCESS in 6m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/414cd40fc9864469b094200bbe73f4f6 : SUCCESS in 32m 29s","accounts_in_message":[],"_revision_number":3},{"id":"80111845e0b2108381690427d4acd301f09f4dd1","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2020-07-08 03:15:17.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"02c12d83a7a51c2a045d25ea1353fd9c9f531471","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 04:15:36.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3ecde4b46201479f886d1972c7d866c9 : FAILURE in 37m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/17f0c7060c4a402a818fe64820ec59df : FAILURE in 40m 38s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fc7af6d504644029acbe54b6623b5640 : FAILURE in 53m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5f26aa625cc3429b9358d102518e6999 : FAILURE in 35m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2f8bf73bf22e463cbcfb4e6a6fc3d829 : FAILURE in 34m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae5b9c91569a4052a0ea25c3ddebeccd : FAILURE in 4m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49cdb7a5007e44669877eec8ea8ddf1d : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68e70fb50a4746c38749daf8c8299415 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0b5a9c3e8554c84aeb2c8d1dcb956c7 : FAILURE in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/341587aabd884dd991c1ac7832abad98 : FAILURE in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/845c1854c86b4662b6e47ee2fd3a8f7d : SUCCESS in 8m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8214226465db4b96a8dc6a90af562939 : SUCCESS in 30m 37s","accounts_in_message":[],"_revision_number":3},{"id":"43d1efad810e355605cc50662f95dc82a459ea02","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2020-07-09 05:24:21.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1b03963a476fd9a08fb9bf1e619bf9154aee1d8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 06:20:31.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/53972a453f3043a3aa55a13f2ab10db8 : SUCCESS in 25m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d310924314e542419748919e879abd76 : FAILURE in 26m 06s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/44c92e02de7741a98d12fc76d224ee5f : FAILURE in 44m 01s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/aa0aa5e63ec147ec9089b364fa8516f6 : FAILURE in 26m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e9849c7d1fa34edd9a3d27c24ddfd0cb : FAILURE in 27m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf8196df96994e33bda252bd876e3aaa : SUCCESS in 5m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/efae869319d8490e8fe80b28990f8796 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/733a496ded3d4e75ba484140577f747f : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44bb045f75ef435888f3544d4788504f : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8051c6c497834b41a018adbf1d9b2ae5 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7d3a885b8d74a4593373b6ecacd2d94 : SUCCESS in 7m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/76a85a09a3404f44b8242f555a9d595f : SUCCESS in 23m 04s","accounts_in_message":[],"_revision_number":3},{"id":"95b381e9f6c79b859b17beed8c9947d486f7b67a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-07-09 06:53:26.000000000","message":"Patch Set 3:\n\nThis seems to be a bug fix that might need to get backported, please create a bug report and reference it.\n\nI\u0027d also like to see a way to trigger the issue in order to be able to verify that this patch fixes it, ideally as a functional test.\n\nI\u0027ve seen https://bugs.launchpad.net/nova/+bug/1882094 but not sure if you just want to extend it to designate or create a separate bug in order to allow for independent discussion.","accounts_in_message":[],"_revision_number":3},{"id":"ec405352935d42f75b712d091f030933488811c1","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2020-07-09 08:12:07.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"cbb8f6c94b7ddd74d9a147adb1905d28c38f2b8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 09:16:31.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1047d9a0a2c34e6e9637366f9d1d1005 : FAILURE in 51m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4add5d3957ce4916842cc1ad4d6c73a0 : SUCCESS in 40m 06s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6d554c2cc96c49f79876adb2fbcad16c : FAILURE in 56m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/edfdc9efbbd343c39ef320d5c875c7c5 : FAILURE in 30m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d5f4563afe5543e6aceed66cc3e94666 : FAILURE in 53m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79d46acbcfbf4a3bbc53243146c74cfb : SUCCESS in 8m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c70deec11fd349b7bff0f108f30ca57f : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/572c4e4a67e64762804e08401383dd5a : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2114a787b30840d782fbce110590cd47 : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2414c6e1591147f3bc69d5be6cf0a670 : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dff10acaa2bb4eadae6fc4e53d9e5923 : SUCCESS in 9m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4eab1cbe88be4eda9867aece433037b9 : SUCCESS in 34m 40s","accounts_in_message":[],"_revision_number":3},{"id":"f509d42b32c908f112ef11b7f91255d775a51862","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-07-13 21:11:31.000000000","message":"Patch Set 3:\n\nI feel like we are hitting some sort of bug in Designate here. Not sure why this change would cause these errors.","accounts_in_message":[],"_revision_number":3},{"id":"082fc84351b838ff00b0ff2be0f60a0953699685","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-07-22 16:36:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"043cc57197c7787113fe7037aabec474cde9272f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-24 04:33:29.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"ffcb1f43fde9296fc912e295b9396df39a996bfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 05:34:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f6c15c34c813467aa1cd5b7e88da28c8 : FAILURE in 38m 20s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/6a17acfa732f4080a183f3662239369c : FAILURE in 38m 23s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/529e1b24a3024841a7dfde08774f92fc : FAILURE in 36m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8ce4d60d67f2449abaf248ec3bfa65ca : FAILURE in 27m 45s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fe0f7062fa20438bb46f10efee72a39a : SUCCESS in 58m 57s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8f544703688a4f51b3506c085ba8a98e : SUCCESS in 28m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/576aa2fba7f449158ca43a71ccf1c9fe : FAILURE in 33m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba020d4040df48e89aebce9da2fcaeb2 : SUCCESS in 5m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d27e0d4fcb940e0b24fa7293032e573 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bba6ac6d3d6f4fc79d2e2c4d0def76e0 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a41a4d7bde944c95940455a4c534fdcb : SUCCESS in 23m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34b0e0d622314160bae1ed358d96c784 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b32b186ac1f14da59af26c967e3a74aa : SUCCESS in 8m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b0b10c030bfd4d619f83f25ddcf2eca7 : SUCCESS in 29m 54s","accounts_in_message":[],"_revision_number":4},{"id":"57208c9bce87c4fa6326ac4b1ef3999d9e6463f2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-27 07:15:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cf4d4f493b6d383de7a89697333054524cb96bf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-27 08:17:53.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ab274ae00b8f4892bb130f185fb82c2e : FAILURE in 38m 32s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/fb7c8e2a7df4406691247beab15081d1 : FAILURE in 49m 24s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/2efdf57052a04a64a0f4c35cc3a787f9 : FAILURE in 38m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/66877e6f168a455abf9ee55f6f20d6fa : FAILURE in 38m 36s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9294578a11ac4aa3b2dd0282bd9fac2f : SUCCESS in 1h 00m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/83abfcdee4dc494585a4c1f72962ff59 : FAILURE in 34m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/757cb0c2fb014b8a8c8b3138cb6b5703 : FAILURE in 29m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d7822b1ba0e4c3389328875011e2e3b : SUCCESS in 6m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9b4859c30144279bdcecf290f8a0401 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3818508d4e44edeb571f45dfc325878 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f995b069a86e468eb49c8218842d8a7e : SUCCESS in 7m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69089776033745c4b25d74afa1bf82d2 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34ded93f0c164437a7f095ef9e949f05 : SUCCESS in 7m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7ed013af4b6b4f758bd3bc3062990989 : SUCCESS in 25m 44s","accounts_in_message":[],"_revision_number":4},{"id":"8d541872ce0457d201a0bd039f75d2db6c1ed442","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-27 13:57:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"24e60ea1826ce3033d14bfcb4bcd9fd2bf9fb87b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-31 06:16:32.000000000","message":"Abandoned\n\nGoing to close this one, but Herve feel free to re-open this one if you want to continue working on it.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6221a8bd2f8552f8b4ff605ef24cf8c3f647ed97","revisions":{"495c5706fef8f3d5c2e68b687b92512f1db75a63":{"kind":"REWORK","_number":1,"created":"2020-06-04 09:28:47.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/06/733506/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/733506/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/733506/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/733506/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/733506/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/733506/1"}}},"commit":{"parents":[{"commit":"21f94dea86abecd0f238f01c4d39d2fafb84d832","subject":"Cleanup docs building","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/21f94dea86abecd0f238f01c4d39d2fafb84d832"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-04 09:27:49.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-04 09:28:32.000000000","tz":120},"subject":"Always reset conf.CONF when starting the wsgi app","message":"Always reset conf.CONF when starting the wsgi app\n\nThis ensures that options loaded during any prior run of the\napplication are dropped before being added again during init_application.\n\nChange-Id: Ia502bdf69650b0ac24fda657b849a061f1172086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/495c5706fef8f3d5c2e68b687b92512f1db75a63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/495c5706fef8f3d5c2e68b687b92512f1db75a63"}]},"branch":"refs/heads/master"},"5823ce3f4d0a81e4c4f1b5d1f4b24db213202e2c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-04 09:40:15.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/06/733506/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/733506/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/733506/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/733506/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/733506/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/733506/2"}}},"commit":{"parents":[{"commit":"21f94dea86abecd0f238f01c4d39d2fafb84d832","subject":"Cleanup docs building","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/21f94dea86abecd0f238f01c4d39d2fafb84d832"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-04 09:27:49.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-04 09:40:01.000000000","tz":120},"subject":"Always reset CONF when starting the wsgi app","message":"Always reset CONF when starting the wsgi app\n\nThis ensures that options loaded during any prior run of the\napplication are dropped before being added again during init_application.\n\nChange-Id: Ia502bdf69650b0ac24fda657b849a061f1172086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5823ce3f4d0a81e4c4f1b5d1f4b24db213202e2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5823ce3f4d0a81e4c4f1b5d1f4b24db213202e2c"}]},"branch":"refs/heads/master"},"2a4c19c0f0fc1506ab7afa459654509f669e0e6e":{"kind":"REWORK","_number":3,"created":"2020-06-08 14:46:49.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/06/733506/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/733506/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/733506/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/733506/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/733506/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/733506/3"}}},"commit":{"parents":[{"commit":"21f94dea86abecd0f238f01c4d39d2fafb84d832","subject":"Cleanup docs building","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/21f94dea86abecd0f238f01c4d39d2fafb84d832"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-04 09:27:49.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-08 14:46:39.000000000","tz":120},"subject":"Always reset CONF when starting the wsgi app","message":"Always reset CONF when starting the wsgi app\n\nThis ensures that options loaded during any prior run of the\napplication are dropped before being added again during init_application.\n\nChange-Id: Ia502bdf69650b0ac24fda657b849a061f1172086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2a4c19c0f0fc1506ab7afa459654509f669e0e6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2a4c19c0f0fc1506ab7afa459654509f669e0e6e"}]},"branch":"refs/heads/master"},"6221a8bd2f8552f8b4ff605ef24cf8c3f647ed97":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-12-24 04:33:29.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/06/733506/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/733506/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/733506/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/733506/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/733506/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/733506/4"}}},"commit":{"parents":[{"commit":"5f87d207b448ec4e24988af5671c3b3593b13804","subject":"Fix support for scoped tokens and default roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f87d207b448ec4e24988af5671c3b3593b13804"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-04 09:27:49.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2021-12-24 04:33:29.000000000","tz":0},"subject":"Always reset CONF when starting the wsgi app","message":"Always reset CONF when starting the wsgi app\n\nThis ensures that options loaded during any prior run of the\napplication are dropped before being added again during init_application.\n\nChange-Id: Ia502bdf69650b0ac24fda657b849a061f1172086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6221a8bd2f8552f8b4ff605ef24cf8c3f647ed97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6221a8bd2f8552f8b4ff605ef24cf8c3f647ed97"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
