)]}'
{"id":"openstack%2Fdesignate~735701","triplet_id":"openstack%2Fdesignate~master~I6b7e9d3b63afe5fc8c658fa903bbf65fde9f26f6","project":"openstack/designate","branch":"master","topic":"allow-creation-of-pools-with-id","hashtags":[],"change_id":"I6b7e9d3b63afe5fc8c658fa903bbf65fde9f26f6","subject":"Allow the creation of pools with a specific ID","status":"ABANDONED","created":"2020-06-15 20:42:03.000000000","updated":"2022-04-21 19:23:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"02fc480469f977b9a712f01d8867f6a940a41c7a","_number":735701,"virtual_id_number":735701,"owner":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-10-05 12:30:05.000000000","_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-15 21:28:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-15 21:28:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-05 12:30:05.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"}],"messages":[{"id":"8b4e1a7c39cf3e925df9964bf9facf53947b8250","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-06-15 20:42:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d857f7706e812a5af6f7ac699b53a1c61c2b6ffd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 21:28:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6c5832613c8e4c0daa2ba16d0b30b7f1 : FAILURE in 34m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3584b218c02c4a0da1b1b71403b0ca57 : FAILURE in 31m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7388cf89778d4885a52a1dfcfb8a39c6 : FAILURE in 17m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4a874b4d9fe8466182af301950159797 : FAILURE in 28m 41s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1f938aa94d3648768c29b125c68c1168 : FAILURE in 33m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de862636295f410f98af9e9021a2e3eb : SUCCESS in 8m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee42a8e4b92f400780e3e4e771703b42 : SUCCESS in 9m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b38477cff4be4eb688bbb4975178594b : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07914dc9a6334962bb0b5ebe264a4f68 : SUCCESS in 8m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa7c3018d08a4f0a98fe309019c712e3 : SUCCESS in 10m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddfc21642c2d4aa9abb7a96e98de6b2a : SUCCESS in 12m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c356690a12d84bd1b5adb77d9690b3a7 : FAILURE in 30m 51s","accounts_in_message":[],"_revision_number":1},{"id":"77d33730548bb390fa5b3d79b6a76716f405af23","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-05 12:30:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9e2527232d0983ee55ba1dff9d3e9773a0d06acd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 13:38:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/aca184d95b174a4e8fe946e29d17afe3 : SUCCESS in 38m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/638db13d0cc94dda8dc7ef48b702bf04 : SUCCESS in 41m 39s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1ff18617a44d4904b4d3e3364c003baa : SUCCESS in 57m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b1f3508570bb4da5a6c45f01e0ed9c6a : FAILURE in 41m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0b6c7a3c43cb46898a033c230e15003a : SUCCESS in 40m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27ab1339a7d04ab8ac21a135f8d57af4 : SUCCESS in 6m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc849354cbfb48c88fb6d8e467b52abb : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9904bdc67ec549e0b168b0ad42b49706 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbbdaa6df05345b8a2fe194c9a79464c : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/479039aac04f40798c8ac6ee614d41ee : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e5a36f112be46a29ede980f1af7951c : SUCCESS in 7m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/93c696991b514c5399266f9b488092a2 : SUCCESS in 37m 30s","accounts_in_message":[],"_revision_number":1},{"id":"02fc480469f977b9a712f01d8867f6a940a41c7a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2022-04-21 19:23:40.000000000","message":"Abandoned\n\nout of date","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1ba3fb478cd89fd1d7baef4db25080eeca7a53b5","revisions":{"1ba3fb478cd89fd1d7baef4db25080eeca7a53b5":{"kind":"REWORK","_number":1,"created":"2020-06-15 20:42:03.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/01/735701/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/01/735701/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/01/735701/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/01/735701/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/01/735701/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/01/735701/1"}}},"commit":{"parents":[{"commit":"7869cc9df017a58514f3508830c6a71d2d875ad1","subject":"Merge \"Use unittest.mock instead of third party mock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7869cc9df017a58514f3508830c6a71d2d875ad1"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2020-06-15 20:40:05.000000000","tz":120},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2020-06-15 20:42:01.000000000","tz":120},"subject":"Allow the creation of pools with a specific ID","message":"Allow the creation of pools with a specific ID\n\nCurrently it is not possible to create new pools with a specific ID.\n\nWith this change it is possible to specify the ID to be used in\npools.yaml already during the creation.\n\nChange-Id: I6b7e9d3b63afe5fc8c658fa903bbf65fde9f26f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ba3fb478cd89fd1d7baef4db25080eeca7a53b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1ba3fb478cd89fd1d7baef4db25080eeca7a53b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
