)]}'
{"id":"openstack%2Fdesignate~737729","triplet_id":"openstack%2Fdesignate~stable%2Ftrain~Iec79c9cec72e74baab9f002b13d426facff96923","project":"openstack/designate","branch":"stable/train","topic":"fix-tempest","hashtags":[],"change_id":"Iec79c9cec72e74baab9f002b13d426facff96923","subject":"Pin version of designate-tempest-plugin","status":"MERGED","created":"2020-06-24 09:50:31.000000000","updated":"2020-06-25 08:10:25.000000000","submitted":"2020-06-25 08:06:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737729-1593072415623-dd5307e8","meta_rev_id":"04612b7b067b716e8c6651419280fb41977bf9d9","_number":737729,"virtual_id_number":737729,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-25 08:06:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"recommended":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":1,"date":"2020-06-24 22:04:52.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2020-06-25 06:50:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-24 22:30:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2020-06-25 06:50:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-24 22:04:52.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-06-24 22:30:30.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-06-25 08:06:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7a1fa57ca72ce83114b7a604980bac146fc70996","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-06-24 09:50:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"382aef07ee317cd4b99e1025bb3bc22b50a765c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 10:45:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/56c307231ff74f23bf1eb493716514e0 : SUCCESS in 42m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c2bea93a361042e6be05fbd2f65dadb3 : SUCCESS in 33m 56s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/f92d380a6766450db4f366ad73887f3a : FAILURE in 15m 36s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/88fc9f80ea39491ba789703e3e8197e6 : FAILURE in 23m 05s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/16c929b95e1d483c888d17ec25545dae : FAILURE in 30m 19s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0b6449254f0f431eb3d646478c01c6f2 : FAILURE in 25m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0c3470217488401c9a09c5936bbb5a4d : SUCCESS in 40m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/376b7788194944e6872e7dd7e4978ff4 : SUCCESS in 47m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57657c364070404fad116472eca5184c : SUCCESS in 6m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/379e33fcb2924dc2991f1b0a56984095 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8d4aa83f68d43559992d5bd9e8e6d97 : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2317554e826e484f917c253ca199e135 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0be3a711703a4176900dc5ed1d0c726f : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d7feba0701d04d9db474c2d4d36207d8 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6fda5e4cd5c480f9398cee0b0510fe0 : SUCCESS in 9m 52s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/234f4d0d4ff7449ca6b38b378832e8f1 : SUCCESS in 33m 54s","accounts_in_message":[],"_revision_number":1},{"id":"637094be924457eb7a0d0b53c933603615932c07","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-06-24 11:12:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"00c112b335c3da5a6f8cdac54c31c0c98016fcd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 11:56:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/22fef9cf5aea44ab9eeb23fbabd35904 : SUCCESS in 37m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/08a8c7afd24a4d07b3d766bf1fecbd7c : SUCCESS in 38m 20s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/de846b72f8594bdab4d7f87c151e47f5 : FAILURE in 18m 15s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/67923bb5afba41afa62d894f96ba6ce6 : FAILURE in 19m 36s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/0772908ebaa74830ab568de223862c17 : FAILURE in 28m 12s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6b8a9ee693c445b19e17fe3383d6ba1f : FAILURE in 25m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1af6c9caa3be4955886fe9983bf43466 : SUCCESS in 37m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e59121ffefab4f2ca65e9a805077b242 : SUCCESS in 41m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5dcc6366321485994ffdaff90ee6c4c : SUCCESS in 6m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b24a82db507d4836ad33007e9edb1ca8 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c73dbc7405943e0a401739ddd88d11d : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/15a0dbbc141944a9a85b0db472035097 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bff7386a52ae4300a986a65e3f100f19 : SUCCESS in 6m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef9b38b96d7f434ba757f2c668e0e3ce : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d7c17026c4345a5bf6ef3e44d54589d : SUCCESS in 6m 41s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/9d7ac25fdc8842b8bda597995c5abdbf : SUCCESS in 38m 30s","accounts_in_message":[],"_revision_number":2},{"id":"4e3aa027c8c4c15b5f58936a32ba3c25d23b5859","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-06-24 17:49:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1fa1381d1dc7d9ee81a54513d25ccc5d33fe446f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-06-24 17:57:38.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"000c3a5200a4bb448bc67364727b81808211b647","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-06-24 17:58:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"39da7eea88442ae64983e2121f76746e1b1ff774","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-06-24 17:58:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ccaab2dd9e8bb1f00eff322f16c7f4bc0dbc1d3c","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-06-24 18:44:13.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f3430195a3487546952d35f32d06c5092457cbcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 18:49:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8bbf0180a74c43b98434d984461beb2b : SUCCESS in 45m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1c0549ec50b84b14908a67839aa29e45 : SUCCESS in 52m 49s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/83d551bd357b44d2a5fe4cf36e1d587e : SUCCESS in 40m 39s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/6a27062a73a549ce8048cedaf7214f61 : SUCCESS in 38m 20s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/431f8a628e8744e08fff58733c285acf : FAILURE in 33m 03s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b508f79df0124dc1b45e62a3229ff888 : FAILURE in 29m 53s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/16b3f7d0b37a4eafa7dc35659d606b09 : SUCCESS in 40m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1883ad498fcd45118e0d4b07b167ed6a : SUCCESS in 42m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/435798ea323148199fe4fe21d84a8134 : SUCCESS in 6m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/762447176c154d66986a0fab87d6a120 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa46fecd6134416d9c1d28d273aaec81 : SUCCESS in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d4a1fc33b9bb4b1cb7a9612cd7075bab : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4af338fedfef490c8f8e61ab9c87c0d9 : SUCCESS in 6m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb9f89cfa7b04501a2d3225d4a1e89d0 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d767d4eccac045d292cb3991cf8732dc : SUCCESS in 8m 43s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/286df38924714958b4a4b9acb38d5335 : SUCCESS in 26m 32s","accounts_in_message":[],"_revision_number":3},{"id":"37899def3a3d19c566b914ca30215e3569a9bf4d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-06-24 19:06:26.000000000","message":"Patch Set 3:\n\ngrenade failed with the same error\n\u003e designate-tempest-plugin requires Python \u0027\u003e\u003d3.6\u0027 but the running Python is 2.7.17","accounts_in_message":[],"_revision_number":3},{"id":"64bd5124d1fb1b11118f2894abe175bad300fa03","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-06-24 19:07:31.000000000","message":"Patch Set 3: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bd9b9766f699dbc9d91abff9518b255d78d154eb","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-06-24 19:08:47.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"e63f2a83df80909d1811898197c74a91bef853d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 20:13:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fe2f696fdeee47d391d83e9fca6032c8 : SUCCESS in 1h 02m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f2242e3ed65a4017a2662bb4b2e2eb8b : SUCCESS in 50m 21s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/f958dfd44bef4c4d9a875518639dcfd7 : SUCCESS in 49m 40s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/08a7669d1b704009b2b9edd649109b8b : SUCCESS in 41m 10s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/f379d7a00492407b8ecb642aac466145 : FAILURE in 28m 26s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e7a562ac4adb46e4b31ccf4ebe03bb69 : SUCCESS in 59m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/754d8b46e4444f4e95a9f015babc3f29 : SUCCESS in 43m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/caee8d3f9b5d44098f49dca73eae5e9a : SUCCESS in 39m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e19c9631920d4abdb663b8633d8bce16 : SUCCESS in 5m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e94586879ae435897aa9738cdf054bf : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f3d0e69428f467bb8a78aa19ad94339 : SUCCESS in 6m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ff6c14b0aebb42529557a002cfdc6836 : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/284554170d574c498f246d6e1f5dcfa0 : SUCCESS in 5m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/513ad525103b4c6aa105f6933c20a20d : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7551c8be0b924c98b1602fe689995389 : SUCCESS in 10m 13s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/a79a9c790e164905a95ef72e8cb2f5ec : SUCCESS in 48m 19s","accounts_in_message":[],"_revision_number":4},{"id":"476c1cd19340aa51949446472d885f93b66f53ea","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-06-24 22:04:52.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"f304ee35997652ea2950b4d773fbd524b22e981e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-06-24 22:30:30.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"56aa6d6fc3a616173df315ba55179315b162f373","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-06-25 06:50:20.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nOh, great, I was thinking we\u0027d need to fix grenade on the \"old\" side, i.e. stein, but with the job override here it looks fine.","accounts_in_message":[],"_revision_number":4},{"id":"dc4b660463e30f0cac9e7a9405885e1e9ebe4695","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 06:50:36.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6282fcee2a2646d61d7ce34bc08e0fa950306675","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 08:06:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"d292c1898263a146f311da8eb4d9a5e53f746a52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 08:06:55.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e33d74eb3e894f6881f5e8becdb93c20 : SUCCESS in 36m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7f87b33f90504667b857c5556fe8d0b0 : SUCCESS in 38m 38s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/0586903918ea4fddabb4a767916859a1 : SUCCESS in 53m 13s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/f5b14b1172a2429aba0ef0b46a574b81 : SUCCESS in 49m 06s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fa697f3ed6754e5da6a62428cc4c787c : SUCCESS in 1h 13m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/36c86eda22f34bb898c1200a63736db8 : SUCCESS in 45m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/113105cfb1fd43058a1c97e5aab14e7e : SUCCESS in 44m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e9edb3f834e47419ab141d0414f3705 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/809acb676fe6445eae9f0c5c3e347502 : SUCCESS in 6m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0b13751519504659b9b7cc8773d2bdd4 : SUCCESS in 8m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c858df2416704f808accc6cb1e99253b : SUCCESS in 9m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/894e929662e84a039b83e4052a1a0e32 : SUCCESS in 8m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cb1692db02748418ec8259de0bee1c1 : SUCCESS in 9m 52s","accounts_in_message":[],"_revision_number":4},{"id":"8ef3454e892e6229184cb7e65f73b3792308e965","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 08:10:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/999becb01b244a87be7a7959450f5ce6 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9eedcf9658b08ce15b3f6286e4816435a94aca9b","revisions":{"ffaeecf5dec055453e71fd78f06db0c4c0741f85":{"kind":"REWORK","_number":1,"created":"2020-06-24 09:50:31.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/29/737729/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/29/737729/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/737729/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/737729/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/737729/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/29/737729/1"}}},"commit":{"parents":[{"commit":"20addf0d521a950577ee80af6d5e3644fd424f87","subject":"Merge \"Improve sink recordset creation\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/20addf0d521a950577ee80af6d5e3644fd424f87"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-06-24 09:49:27.000000000","tz":120},"committer":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-06-24 09:50:27.000000000","tz":120},"subject":"Pin version of designate-tempest-plugin","message":"Pin version of designate-tempest-plugin\n\nMaster dropped support for py2 but we still need it for some jobs.\n\nChange-Id: Iec79c9cec72e74baab9f002b13d426facff96923\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ffaeecf5dec055453e71fd78f06db0c4c0741f85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ffaeecf5dec055453e71fd78f06db0c4c0741f85"}]},"branch":"refs/heads/stable/train"},"1915bdf54a8506b2ec7b20569fc3441fe6e27a20":{"kind":"REWORK","_number":2,"created":"2020-06-24 11:12:04.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/29/737729/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/29/737729/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/737729/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/737729/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/737729/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/29/737729/2"}}},"commit":{"parents":[{"commit":"20addf0d521a950577ee80af6d5e3644fd424f87","subject":"Merge \"Improve sink recordset creation\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/20addf0d521a950577ee80af6d5e3644fd424f87"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-06-24 09:49:27.000000000","tz":120},"committer":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-06-24 11:11:37.000000000","tz":120},"subject":"Pin version of designate-tempest-plugin","message":"Pin version of designate-tempest-plugin\n\nMaster dropped support for py2 but we still need it for some jobs.\n\nChange-Id: Iec79c9cec72e74baab9f002b13d426facff96923\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1915bdf54a8506b2ec7b20569fc3441fe6e27a20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1915bdf54a8506b2ec7b20569fc3441fe6e27a20"}]},"branch":"refs/heads/stable/train"},"6974918a1f8314640361b894af983bd73493401c":{"kind":"REWORK","_number":3,"created":"2020-06-24 17:49:36.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/29/737729/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/29/737729/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/737729/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/737729/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/737729/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/29/737729/3"}}},"commit":{"parents":[{"commit":"20addf0d521a950577ee80af6d5e3644fd424f87","subject":"Merge \"Improve sink recordset creation\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/20addf0d521a950577ee80af6d5e3644fd424f87"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-06-24 09:49:27.000000000","tz":120},"committer":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-06-24 17:49:07.000000000","tz":120},"subject":"Pin version of designate-tempest-plugin","message":"Pin version of designate-tempest-plugin\n\nMaster dropped support for py2 but we still need it for some jobs.\n\nChange-Id: Iec79c9cec72e74baab9f002b13d426facff96923\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6974918a1f8314640361b894af983bd73493401c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6974918a1f8314640361b894af983bd73493401c"}]},"branch":"refs/heads/stable/train"},"9eedcf9658b08ce15b3f6286e4816435a94aca9b":{"kind":"REWORK","_number":4,"created":"2020-06-24 19:08:47.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/29/737729/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/29/737729/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/737729/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/737729/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/29/737729/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/29/737729/4"}}},"commit":{"parents":[{"commit":"20addf0d521a950577ee80af6d5e3644fd424f87","subject":"Merge \"Improve sink recordset creation\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/20addf0d521a950577ee80af6d5e3644fd424f87"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-06-24 09:49:27.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-06-24 19:08:44.000000000","tz":0},"subject":"Pin version of designate-tempest-plugin","message":"Pin version of designate-tempest-plugin\n\nMaster dropped support for py2 but we still need it for some jobs.\n\nChange-Id: Iec79c9cec72e74baab9f002b13d426facff96923\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9eedcf9658b08ce15b3f6286e4816435a94aca9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9eedcf9658b08ce15b3f6286e4816435a94aca9b"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
