)]}'
{"id":"openstack%2Fdesignate~739394","triplet_id":"openstack%2Fdesignate~master~Id78112b49d5c0e99064fcb271aaded871da3f633","project":"openstack/designate","branch":"master","topic":"context_project","attention_set":{},"removed_from_attention_set":{"6994":{"account":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"last_update":"2022-02-17 23:43:57.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-02-17 23:43:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id78112b49d5c0e99064fcb271aaded871da3f633","subject":"Move context from tenant to project_id","status":"MERGED","created":"2020-07-06 04:14:27.000000000","updated":"2022-02-17 23:43:57.000000000","submitted":"2022-02-17 23:43:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739394-context_project","meta_rev_id":"33785cde8b42bd6e697e2eeb2d3931483dff072e","_number":739394,"virtual_id_number":739394,"owner":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-17 23:43:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2022-02-17 23:43:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-17 21:21:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-17 21:21:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-06 07:13:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-14 12:23:54.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2022-02-17 21:21:53.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"eb447412374a4b5523390feb7f439b8474a28520","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2020-07-06 04:14:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd0b02083c01398b36a0d35f3dfb06fe395fbde2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 05:20:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fae7b420272c4aaa8d3c39c56e7d5f00 : FAILURE in 58m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ab4a19fd032e4d2980ae98d5c8340ef5 : SUCCESS in 39m 11s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c1f78cbec2534f0a9a6eea38ba423eac : SUCCESS in 56m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3fd0574b411a4f97810630039f8958b5 : SUCCESS in 50m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5908cdbccf174434b7c6696f5b681890 : SUCCESS in 38m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/934a4679024c4a629be92889b7a72cc7 : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93bb5d4fb3b5444cb04031c08eb38de6 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d6bfde9c4bc493193aea468e117e519 : FAILURE in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/549c1796eca940d19194c7d5763334ce : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/242920c0ef754e7390c036358fc355c7 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0525edfaffe450990a5d04332292a35 : SUCCESS in 15m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0a912c78f56b47ccb3d2965ba80d97dc : SUCCESS in 32m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e4c4c3de62d7b76f0d0fe788afc13648f649a6a4","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2020-07-06 06:06:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bbe646fd0f987ecc2307cb90ae83838924f42f8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 07:13:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/960a67d8f0ab4fd1945660e90b33e823 : SUCCESS in 43m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/977b8951e28448efa44a34f4a813c3e2 : SUCCESS in 43m 34s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1ef099b376dd431f90d42a20e5521c1a : SUCCESS in 57m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2953cd9da5364a40a53ab76197e88543 : SUCCESS in 39m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/963e808e66964d9097b3155948975b38 : SUCCESS in 46m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c21181687e504bf1b4a8fd999d1a34f5 : SUCCESS in 10m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d68450eae21c4d979f3a8fcd5ba389c1 : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/242542dc26ce453d97a6ad3970157af6 : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca1b28f702444278b5901b1f52ad5b20 : SUCCESS in 9m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8507c7ff138c40b8b4c38a3b266e753f : SUCCESS in 9m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77c2dd030c04438691d75a2cfb43a5d6 : SUCCESS in 9m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bf17511665b74c17953a06f9ea0df7ab : SUCCESS in 40m 38s","accounts_in_message":[],"_revision_number":2},{"id":"9ca0c7ef408cd6c755a549101caf9c633d320090","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-14 12:23:54.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":2},{"id":"1cfbf0ddc0aa354647fb52e352b055ba761e64df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-01-13 05:00:35.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"4fbd9e55caf8f767bc04ef3567ce2c52865d2abc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 06:09:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/69e506657c26424ea4f2b6d77ced16ed : SUCCESS in 35m 25s\n- designate-bind9-centos8stream-fips https://zuul.opendev.org/t/openstack/build/fcf78c667acd4d05a39838cb0e4a8d00 : SUCCESS in 45m 31s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/62bdeb962b3b4053881e901f98eae80d : SUCCESS in 38m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d363874d4a9f460aab1a0bbe65629812 : SUCCESS in 46m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/eba385fc68704b978f2ee362183c9d31 : SUCCESS in 1h 05m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0ed6680a07d240288f571b3205bb585f : SUCCESS in 36m 33s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/73562c43e2334848acf97f6579a3bc52 : SUCCESS in 39m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b694f5141d994228bddc09cee76bc012 : SUCCESS in 6m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ef2ca3fbe024957b3764664071d791c : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e9a90f3768a42a9a1efc8fb10424aaa : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3d826dc4b1846ae8c2caac206165c20 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e716272480cc4ce8975761f67f56c8cf : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9bae1e055f648c08a4f54657ab6c44d : SUCCESS in 7m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/11a05e7f48064e74be7784cd8744c732 : SUCCESS in 40m 50s","accounts_in_message":[],"_revision_number":3},{"id":"328ae757ae1da2e8c7e3960a57689d5a07de4867","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-02-17 21:21:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7898908e1446091e301cfd86e5157d10734224b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 21:22:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"04800058147e41224d3a101027869074884ab101","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 23:43:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ea7dd450b04544d887fecabcefbb1520 : SUCCESS in 51m 34s\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f20bb556bf2641768145499cc3a97efb : SUCCESS in 47m 39s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/dae34884ec2747d3837e57eca81f97ae : SUCCESS in 39m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/dda2c1f6addb414dbb64527f8d97c4df : SUCCESS in 45m 56s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cd4dcc00ad304377bbc7f787a0ceee0b : SUCCESS in 1h 02m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e75e8b4871d34ac1adc2228f065bba60 : SUCCESS in 47m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/33393cbde73546dab8c75f46cf8da269 : SUCCESS in 46m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9d0ec9c842147c8be3b90715ce27671 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c1443bbde044e7dbc88398a74b5052e : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3571d381d52945b9a341b40dfdcb96f3 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aedf25f8f6124993a8b2c8fdca78aeaf : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56a05b49d08f40f6890af403f7a80e85 : SUCCESS in 7m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a202b50c0d8243a2a7c8bb27170a04e2 : SUCCESS in 30m 04s","accounts_in_message":[],"_revision_number":3},{"id":"3d3b13e7e3a2acab80d2083ab2f515a3a70a5c3b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 23:43:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6c80f6161aa01a1eb7da419f5d396ed60ab08066","revisions":{"a726f6238be13e2e3d9704f4c39eb633a9809d86":{"kind":"REWORK","_number":1,"created":"2020-07-06 04:14:27.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/94/739394/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/739394/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/739394/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/739394/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/739394/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/739394/1"}}},"commit":{"parents":[{"commit":"eac003cb3a0fa79f0201a7b89c07b111481385e5","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eac003cb3a0fa79f0201a7b89c07b111481385e5"}]}],"author":{"name":"michaeltchapman","email":"woppin@gmail.com","date":"2020-07-06 04:10:04.000000000","tz":600},"committer":{"name":"michaeltchapman","email":"woppin@gmail.com","date":"2020-07-06 04:13:07.000000000","tz":600},"subject":"Move context from tenant to project_id","message":"Move context from tenant to project_id\n\nin the sudo context unit test:\nDeprecationWarning: Property \u0027tenant\u0027 has moved to \u0027project_id\u0027\nin version \u00272.6\u0027 and will be removed in version \u00273.0\u0027\n\noslo_context is deprecating the tenant field, so move to using\nproject_id. This does leave a less than ideal state where both\nproject_id and tenant are referenced (since all_tenants remains\nuntouched) but replacing \u0027tenant\u0027 with \u0027project\u0027 across the\nentire codebase is a much larger task.\n\nChange-Id: Id78112b49d5c0e99064fcb271aaded871da3f633\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a726f6238be13e2e3d9704f4c39eb633a9809d86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a726f6238be13e2e3d9704f4c39eb633a9809d86"}]},"branch":"refs/heads/master"},"0ccbd21b77ad173e797f713ad75a51b46d26178d":{"kind":"REWORK","_number":2,"created":"2020-07-06 06:06:55.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/94/739394/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/739394/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/739394/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/739394/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/739394/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/739394/2"}}},"commit":{"parents":[{"commit":"eac003cb3a0fa79f0201a7b89c07b111481385e5","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eac003cb3a0fa79f0201a7b89c07b111481385e5"}]}],"author":{"name":"michaeltchapman","email":"woppin@gmail.com","date":"2020-07-06 04:10:04.000000000","tz":600},"committer":{"name":"michaeltchapman","email":"woppin@gmail.com","date":"2020-07-06 06:06:39.000000000","tz":600},"subject":"Move context from tenant to project_id","message":"Move context from tenant to project_id\n\nin the sudo context unit test:\nDeprecationWarning: Property \u0027tenant\u0027 has moved to \u0027project_id\u0027\nin version \u00272.6\u0027 and will be removed in version \u00273.0\u0027\n\noslo_context is deprecating the tenant field, so move to using\nproject_id. This does leave a less than ideal state where both\nproject_id and tenant are referenced (since all_tenants remains\nuntouched) but replacing \u0027tenant\u0027 with \u0027project\u0027 across the\nentire codebase is a much larger task.\n\nChange-Id: Id78112b49d5c0e99064fcb271aaded871da3f633\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0ccbd21b77ad173e797f713ad75a51b46d26178d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0ccbd21b77ad173e797f713ad75a51b46d26178d"}]},"branch":"refs/heads/master"},"6c80f6161aa01a1eb7da419f5d396ed60ab08066":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-01-13 05:00:35.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/94/739394/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/739394/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/739394/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/739394/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/739394/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/739394/3"}}},"commit":{"parents":[{"commit":"bff3d5f6e31fe595a77143ec4ac779c187bf72a8","subject":"Fix designate-manage pool update bugs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bff3d5f6e31fe595a77143ec4ac779c187bf72a8"}]}],"author":{"name":"michaeltchapman","email":"woppin@gmail.com","date":"2020-07-06 04:10:04.000000000","tz":600},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-01-13 05:00:35.000000000","tz":0},"subject":"Move context from tenant to project_id","message":"Move context from tenant to project_id\n\nin the sudo context unit test:\nDeprecationWarning: Property \u0027tenant\u0027 has moved to \u0027project_id\u0027\nin version \u00272.6\u0027 and will be removed in version \u00273.0\u0027\n\noslo_context is deprecating the tenant field, so move to using\nproject_id. This does leave a less than ideal state where both\nproject_id and tenant are referenced (since all_tenants remains\nuntouched) but replacing \u0027tenant\u0027 with \u0027project\u0027 across the\nentire codebase is a much larger task.\n\nChange-Id: Id78112b49d5c0e99064fcb271aaded871da3f633\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6c80f6161aa01a1eb7da419f5d396ed60ab08066"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6c80f6161aa01a1eb7da419f5d396ed60ab08066"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
