)]}'
{"id":"openstack%2Fdesignate~739894","triplet_id":"openstack%2Fdesignate~master~Idef54da1134e14511f336911b7c82f6a0bb282ce","project":"openstack/designate","branch":"master","topic":"mock_removal","hashtags":[],"change_id":"Idef54da1134e14511f336911b7c82f6a0bb282ce","subject":"Replace mock with unittest.mock","status":"ABANDONED","created":"2020-07-08 05:31:39.000000000","updated":"2020-07-08 08:52:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c1e21edeb1c1931af90493440d7ec030a8939d0a","_number":739894,"virtual_id_number":739894,"owner":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-07-08 08:31:57.000000000","_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-08 06:35:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":1,"date":"2020-07-08 08:00:03.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-08 06:35:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-08 08:00:03.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"7a0c167f6d8ecccf66ce3aaec97b07dc4ea5183a","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2020-07-08 05:31:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50b6af3577afb3d9abd7454b0193561f0a25d304","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 06:35:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ec3981a7d01548bca206a43417f05e16 : SUCCESS in 39m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/eb1ed6d11dc644b4a549f3d8f5eeebad : SUCCESS in 37m 30s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cae9addcadb349afa42bc7c67a71348e : SUCCESS in 58m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/73ab2337e4cd499c9899e353b9920174 : SUCCESS in 36m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c2f4401dd9334f368f10c90facb8be6b : SUCCESS in 43m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a37316c60bab4296a937edffcc2ae914 : SUCCESS in 5m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cab790d0f91945fe98023716c7bd5792 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14dba7bc0246443b9fc9a486e454d3d3 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7f46a9c570347b3b95dfb0b4ce87a9e : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9488ee41b33246abb06df0c02dd86730 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30bc7685ed1546a79cd1d88c26d97d6f : SUCCESS in 7m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/338b68c31cd94ba99275cbb06c99ebc7 : SUCCESS in 34m 12s","accounts_in_message":[],"_revision_number":1},{"id":"604cd112d5e00fe217b1ba7c0753870aa6ee4e60","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-08 08:00:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"93aa3b849c4bccf1ce402a161796eb58b331a464","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2020-07-08 08:31:57.000000000","message":"Patch Set 1:\n\nAddressed here: https://review.opendev.org/#/c/734329/1","accounts_in_message":[],"_revision_number":1},{"id":"a1d0de5d53730da7d916e470bd4388b881e01cb2","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2020-07-08 08:52:29.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"03fb2bf8605b6ab92bca346a4d00c4e08370fb38","revisions":{"03fb2bf8605b6ab92bca346a4d00c4e08370fb38":{"kind":"REWORK","_number":1,"created":"2020-07-08 05:31:39.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/94/739894/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/739894/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/739894/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/739894/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/739894/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/739894/1"}}},"commit":{"parents":[{"commit":"eac003cb3a0fa79f0201a7b89c07b111481385e5","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eac003cb3a0fa79f0201a7b89c07b111481385e5"}]}],"author":{"name":"michaeltchapman","email":"woppin@gmail.com","date":"2020-07-08 05:29:15.000000000","tz":600},"committer":{"name":"michaeltchapman","email":"woppin@gmail.com","date":"2020-07-08 05:29:15.000000000","tz":600},"subject":"Replace mock with unittest.mock","message":"Replace mock with unittest.mock\n\nUnittest is in the std library since py3\n\nChange-Id: Idef54da1134e14511f336911b7c82f6a0bb282ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/03fb2bf8605b6ab92bca346a4d00c4e08370fb38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/03fb2bf8605b6ab92bca346a4d00c4e08370fb38"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
