)]}'
{"id":"openstack%2Fdesignate~740655","triplet_id":"openstack%2Fdesignate~stable%2Fstein~I9693b71038f91d976ac82e768697001a4dc80e05","project":"openstack/designate","branch":"stable/stein","topic":"grenade-em-nv","hashtags":[],"change_id":"I9693b71038f91d976ac82e768697001a4dc80e05","subject":"Revert \"Make greande jobs n-v for EM and oldest stable\"","status":"ABANDONED","created":"2020-07-13 09:59:34.000000000","updated":"2021-12-24 04:32:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"25d033ecd4b441cb271c8b5f40c0a9f6b10e1f75","_number":740655,"virtual_id_number":740655,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-18 21:18:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-13 10:32:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-13 21:12:17.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-10-05 12:24:36.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"}],"messages":[{"id":"1d3e4d23a2fd8ce87dc803c1746215885ee8d36c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-13 09:59:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b7dcb08c5e29fbf70e37d17964117f12f445809","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 10:32:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e3d7b1faa9e143188e38919f2dabfc9e : FAILURE in 13m 38s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/ff1c8a25779a4922b767d28be8f8082a : FAILURE in 18m 36s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/263daff3cea04e1dbe887f9e991284e8 : FAILURE in 19m 31s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/763ae04a5d324585a080e5519a94ed0d : FAILURE in 17m 03s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/c52777aadffe493fa72bf5e544076566 : FAILURE in 17m 42s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0c7d8559a97d47bbb0ea231c3c6013db : FAILURE in 19m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d4f921b80414c959b32f41ef00290e1 : SUCCESS in 6m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1f733068d764398bc158ffe58cf3bcc : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9adbb384245b449fbab6957644abbb1f : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6932227f9f7544e9a220feb6fb29d821 : SUCCESS in 5m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c96ca01e5892476c8fa06a168cd76ce7 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6394ad726b2f43ab9f6095e07f7f7d5c : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/15cccda23d9447d9aef6067af8c53969 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a01bf918fa345e9b520b80e708f2d9e : SUCCESS in 5m 07s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/7008b4c60f70402091470de522114b5b : SUCCESS in 31m 24s","accounts_in_message":[],"_revision_number":1},{"id":"3cf77e490a438824761286179345fa9d7df54f77","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-07-13 21:12:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8b52fdca73875fe424ed0a614252344be2b936d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 21:45:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c78aa13c2c5640be9e2c5c0738e1489c : FAILURE in 21m 07s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/f94b3c71d31b47b8a620a52e5f784185 : FAILURE in 24m 42s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/3639b3d1a9134e74b634ef96a5091ed9 : FAILURE in 15m 16s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/f960b310103a4677a01b7a992813f68a : FAILURE in 14m 20s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/f24f8707214e41aabc2a8b2f0aa212b7 : FAILURE in 27m 29s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/55673678496e4880b9a40dbe044dfe53 : FAILURE in 27m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d87984df2cd645e3b711fda4fd224ac3 : SUCCESS in 5m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c143f4a14084ebda5f9eeffec7df684 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/794976b3a38d4c9f875e1835d375d693 : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7d08062f7db54e48b754126fa63142f5 : SUCCESS in 4m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8e6f6550f50d4708abb41a3ec1245fa1 : SUCCESS in 8m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e8f1c4a01cb4262b8ea22edec1fedb9 : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/70784af5177e41908a1a4d202ee0726e : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da974e8c0dc24738bb5bdfaad3cbe5e3 : SUCCESS in 5m 25s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/3e378aeca86b4e8a9da96930a22a19ec : SUCCESS in 25m 12s","accounts_in_message":[],"_revision_number":1},{"id":"731d33a2e3c2a87dd5e4f866d3a0e4f92c31320d","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-05 12:24:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"23ac2780a82fdfdd3b5fab2f9965f4017a948d84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 13:09:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/09ac2a57378e4461b945b7d015d7c725 : FAILURE in 22m 33s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/c7347f5a209145d09313997d58c0ca9a : FAILURE in 21m 52s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/eaba3cdee4b34e59b8eda37e2f8e61b3 : FAILURE in 20m 20s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/c5fb87a08f0d40cdbc03d6f5da0aefed : FAILURE in 22m 55s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/bf7906b8e31c4d6fa768923f575499f7 : FAILURE in 23m 12s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d3f27b813bb74344b989b87cd2819394 : FAILURE in 24m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa4fc73b132f444995b9b2ae32819514 : SUCCESS in 5m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b947535757ca4a9abfb2c50715732a98 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81d06dd5cfaf4998903e2d7019ee1df6 : SUCCESS in 6m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ac820aa1ec4c434884b4a460276e0196 : SUCCESS in 5m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/77e81d754cbf4feca38a28b91211d6c3 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9bb2db08791d47b49e10a0229f325523 : SUCCESS in 7m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf27768a08ee4547a7642062493ed87f : FAILURE in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ee9bdfe5ced4e51a4ada11665b7829f : SUCCESS in 6m 06s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/b3541083738c4ee9b675488e27cc3cf0 : SUCCESS in 36m 32s","accounts_in_message":[],"_revision_number":1},{"id":"1f36adb889b9e1e4093784a16f05c66c406acadc","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-01-13 12:04:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bae813acda5a7a4684f16985a8c2a527d3ec2e5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 12:45:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b4ab0c66d5d845679515446e4dcc6f9a : FAILURE in 17m 42s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/08b02cd805cb4d8d987dce37c4a55611 : FAILURE in 16m 47s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/6fc26edb1f5d4df2a0e1f421b6f4e3ac : FAILURE in 17m 36s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/10f69b67d0e34a7784c35c710c73be6e : FAILURE in 19m 06s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/8df9dea93a114aad831a1235ebaa61de : FAILURE in 20m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/412f02e76f5148feb7c4064e3d14600a : FAILURE in 26m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2cc46a0444f84d7eb5ceb06cece09adb : SUCCESS in 6m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/948835f6f039489694df5a211f9575b8 : FAILURE in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a29960fa3ff425e8fb89854bce5dc7a : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d7b28427a334446788d7ec048f0eb6cc : FAILURE in 3m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6506792785e4424c96fe0aaa39e0ab84 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a44594f960a54d7182eacf5a30d21c9a : SUCCESS in 6m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7341e0f1318646d8b6b50045c62f8b37 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df489895ca6c4e5a92d02742acb67545 : SUCCESS in 5m 03s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/7ce6ed0fb2d942ca81863f5020840e6a : SUCCESS in 29m 41s","accounts_in_message":[],"_revision_number":1},{"id":"881185e3b6574ee323f27efa638f88863ccf083a","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-18 20:16:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5eaf17f536b9432c7dfeb978fe24408d496c8548","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-18 20:35:19.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"80a3b0899c6dc9406d0258876b52dfb85bb5134a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 21:18:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/218f41e170a243c398240cb273c18fa4 : FAILURE in 21m 17s\n- designate-bind9-manager-model https://zuul.opendev.org/t/openstack/build/af9dca19b8d643e0a6256323bb3a376c : FAILURE in 18m 55s\n- designate-bind9-py36 https://zuul.opendev.org/t/openstack/build/4637a618b2b841c79f2f9aa227c51b27 : FAILURE in 19m 44s\n- designate-devstack-pdns4 https://zuul.opendev.org/t/openstack/build/9920ab5f3b8d495eadaaf9f7f490d12d : FAILURE in 28m 20s\n- designate-devstack-pdns4-postgres https://zuul.opendev.org/t/openstack/build/1d21e302bd3b4bc3a4945063891a1a05 : FAILURE in 19m 21s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a7a5be8ef5614ae4951edbd15317ba9e : FAILURE in 20m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aad9b53437544e5e8fa174f633c66618 : SUCCESS in 6m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f84afeeeae440779840d74bbb5e5e28 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d666131b960146d3a46ccd50b02592ba : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4c9cccdca1fa451aa065358cec91d6bf : SUCCESS in 5m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c28c28682a7745dc828bafab3760c932 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36036c2e907041ab8a15a0e9a7a6da2b : SUCCESS in 6m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/39333a1edae94550810b2b4837ba20ff : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a93264bec69649d8b10d9c1a72777ac2 : SUCCESS in 5m 48s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/1d8054e10400493f8beba2e58fd56e5e : SUCCESS in 35m 24s","accounts_in_message":[],"_revision_number":2},{"id":"25d033ecd4b441cb271c8b5f40c0a9f6b10e1f75","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-12-24 04:32:14.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6298ecaa40aa9b7be60c552fe194bcf5a8c458cd","revisions":{"38fc849bfd39056e03f128b884860b14644fcbf8":{"kind":"REWORK","_number":1,"created":"2020-07-13 09:59:34.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/55/740655/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/740655/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/740655/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/740655/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/740655/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/740655/1"}}},"commit":{"parents":[{"commit":"6307c84e83530ab27124ab9e6a8117ee44069614","subject":"Make greande jobs n-v for EM and oldest stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6307c84e83530ab27124ab9e6a8117ee44069614"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-07-13 09:55:45.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-07-13 09:59:27.000000000","tz":120},"subject":"Revert \"Make greande jobs n-v for EM and oldest stable\"","message":"Revert \"Make greande jobs n-v for EM and oldest stable\"\n\nThis reverts commit 6307c84e83530ab27124ab9e6a8117ee44069614.\n\nNOTE(elod.illes): devstack in Rocky is fixed and grenade jobs started\nto pass in Stein. We can enable grenades again to restore testing as it\nwas before.\n\nChange-Id: I9693b71038f91d976ac82e768697001a4dc80e05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/38fc849bfd39056e03f128b884860b14644fcbf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/38fc849bfd39056e03f128b884860b14644fcbf8"}]},"branch":"refs/heads/stable/stein"},"6298ecaa40aa9b7be60c552fe194bcf5a8c458cd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-18 20:35:19.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/55/740655/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/55/740655/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/740655/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/740655/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/55/740655/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/55/740655/2"}}},"commit":{"parents":[{"commit":"28f74c81e76ce4e1f84920f52a3ae5530f0097c2","subject":"Merge \"Update vagrant config\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/28f74c81e76ce4e1f84920f52a3ae5530f0097c2"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-07-13 09:55:45.000000000","tz":120},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-18 20:34:45.000000000","tz":-420},"subject":"Revert \"Make greande jobs n-v for EM and oldest stable\"","message":"Revert \"Make greande jobs n-v for EM and oldest stable\"\n\nThis reverts commit 6307c84e83530ab27124ab9e6a8117ee44069614.\n\nNOTE(elod.illes): devstack in Rocky is fixed and grenade jobs started\nto pass in Stein. We can enable grenades again to restore testing as it\nwas before.\n\nChange-Id: I9693b71038f91d976ac82e768697001a4dc80e05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6298ecaa40aa9b7be60c552fe194bcf5a8c458cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6298ecaa40aa9b7be60c552fe194bcf5a8c458cd"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
