)]}'
{"id":"openstack%2Fdesignate~741054","triplet_id":"openstack%2Fdesignate~master~I7ce84ac9cf5bc5f6fca47dc79c51fe0becf96ac6","project":"openstack/designate","branch":"master","topic":"bug/1886685","hashtags":[],"change_id":"I7ce84ac9cf5bc5f6fca47dc79c51fe0becf96ac6","subject":"Fix multi messages AXFR with TSIG","status":"MERGED","created":"2020-07-14 21:08:14.000000000","updated":"2020-08-13 16:08:33.000000000","submitted":"2020-08-13 16:07:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741054-1597334821500-2e16d46c","meta_rev_id":"18b57dbae5310414de4dd341d6ca150808563679","_number":741054,"virtual_id_number":741054,"owner":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-13 16:07:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"value":0,"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2020-07-30 19:56:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-30 18:05:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"value":0,"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":1,"date":"2020-08-13 15:12:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"value":0,"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"value":0,"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-23 17:19:42.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-08-13 15:12:39.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-08-13 16:07:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"218f63ce5e7e9584b10ab85d593b6f8349fc411c","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-07-14 21:08:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3bd92ec3e8c97742e4ebf574435a6c264beefde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 22:11:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ec2b3ded3dc2495a92ac2395726c8b1d : SUCCESS in 36m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/23ca34f6908947fc9a7095b351c98d1f : SUCCESS in 36m 07s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6b6205850f96488483519204a4a4c464 : SUCCESS in 56m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2c402781708d46838901666e9c40ec17 : SUCCESS in 34m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/46eb0d61f3864915981bc194fa9fef93 : SUCCESS in 39m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2078520dd9647b3acbe577d48a72719 : SUCCESS in 6m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a558c646ba247cea39791a4552922cb : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ecdf81a459d4413aa481abfed034277 : FAILURE in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19c1756684724e048654680e70df251a : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/593fa137f14d48788a22d7e182c4c765 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d98e37878e3144069c709f5a998f7d9e : SUCCESS in 7m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/04cb37110c6b4625a99602952bb9e155 : SUCCESS in 34m 54s","accounts_in_message":[],"_revision_number":1},{"id":"72005aa97ea1abb020475baf04263a6f9cd9a2b9","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-07-14 22:26:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e347da7ed89b91bfd150842d0367978be7f01490","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 23:34:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d64291488bf8450081a4649264bdb2d7 : SUCCESS in 39m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5c90cde9103c44649ba7489bb3bd8232 : SUCCESS in 38m 26s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/53c93096a05647d1aaf47afff9931733 : SUCCESS in 1h 01m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f10fa6bdf2904e8f87f1d1cb87218e01 : SUCCESS in 38m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bfc19eeca50641698eb092c3cf9e3915 : SUCCESS in 40m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d117d8ea2b2b4db8b622d4b37a9b5a03 : SUCCESS in 5m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ae4cfa327c143819d1cb44b6bac3e01 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90e01296982a445f95abaec7be300dbd : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a7a6d256be547029a58ab1303a951f7 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c75d1a35e6e34b2c8747f00282e9e485 : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06fa4f830bbb4f3b90a4abe957124c6e : SUCCESS in 7m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/78590f175536433ba674fbefce7e9074 : SUCCESS in 40m 53s","accounts_in_message":[],"_revision_number":2},{"id":"13a11fd9d0530d8def4a6cfe8622b02f162a4fc4","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-07-21 18:49:15.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"266ffaec6db0a1060772ca74755941d30a545908","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-07-22 19:52:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7bcb2447b464e509da09dba332bf5e889a76e8f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 21:02:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5c21669de1494b04855648f48b0b1c1e : FAILURE in 49m 43s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/59a31eec237145b0b1ad63d3369042ed : SUCCESS in 47m 11s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/aafd29cc70914afa9c9d4f6cdf5d8068 : SUCCESS in 1h 02m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0bdd86ba652948fa900ab0ab0a05f330 : SUCCESS in 35m 10s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/da45989002624f1f90cb86c35ac64f02 : SUCCESS in 38m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a389b2a10aa141bd8ca101288f3829c4 : SUCCESS in 7m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c2e2d7c03cc498e8a13459d9a1933e1 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17b01fea1c8e4ba4838f4a2c0d6fb53d : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a86a340112f748e4ab0ed8b442883500 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1b5afff2b03472baf73837863747f6e : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cacd0849634648308c1958102d74b9e4 : SUCCESS in 9m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2924e34b37c64535aa06177d3ba0a09f : SUCCESS in 29m 19s","accounts_in_message":[],"_revision_number":3},{"id":"31a27f8c170d039d08d45bd46ecb05db5b35ff2e","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-07-22 22:23:23.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2663a5ade780653ba43d8fea20011a37f614c210","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-07-22 22:32:35.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"fa945a826cd8efc9535b9635e0bdd30f12c425ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 23:27:50.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9f69d22c4ba14db49ec18f370096ea61 : SUCCESS in 35m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/30ab8ad030d44d1ca40ec033353a881b : SUCCESS in 32m 22s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/659e51a7c6fb4bdd836fad7ee95da21a : SUCCESS in 49m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ab1d2720cdb440eaaca13cdd4cf84732 : SUCCESS in 43m 20s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/28114e9bd48143a99cf1a911f7832c77 : FAILURE in 34m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6155b8e1a5474e72a503f659e203fc84 : SUCCESS in 5m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/077f88432ab84d57a7e0131742970771 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6886b8d5ca894bdc8509b38f72ea9c03 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/490accf2d8714a42a1c3375345078858 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f765f16506a4f3fa576f5ad670fd0dd : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/194d6a1933284001890b88c18368cc4a : SUCCESS in 7m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4ca6777746fe4ee08b46955e48489cdb : SUCCESS in 36m 03s","accounts_in_message":[],"_revision_number":3},{"id":"49b434be5bd014d129d174fbbcfc19ce94d5c353","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-07-23 03:34:00.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c8ffcb5fc9c6e069a155229bc7f4ed3e9a15d97f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 04:20:22.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/42c72d1c3f9b45deacebab21d557efc6 : SUCCESS in 44m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0b523f8af96b4a979ea43998189c4c93 : SUCCESS in 37m 51s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/02616ed599c84a16914990345b144118 : SUCCESS in 37m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5c815c0fd11d413b85a5574045182497 : FAILURE in 25m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f894d4b0ec2a48a0b055e15818dbff44 : SUCCESS in 25m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1535a0ac962a458f9b098824716af326 : SUCCESS in 5m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/689f4974297e4319a72adeab9c571e15 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b0eb45d169a42b8b32fc086c09167b7 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6cc0b2506ab84b469baf9467051b98ea : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/240c4efe25c444068edaa42ec5c37a6d : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1fb48568777450cb79ef40ae33c9224 : SUCCESS in 6m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1e0ac584f4f44fe880da7d8a6aa11a1c : SUCCESS in 36m 29s","accounts_in_message":[],"_revision_number":3},{"id":"cb49857e16512fc7307f57f63863086bc833b137","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-07-23 04:24:48.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2934efc90342ecad2e1159f787ce19c05f90233a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 06:20:39.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/afea04f16c3d477abb0ab9db10711958 : SUCCESS in 36m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/58ff628ca5ff49f48654943f2bcfd147 : FAILURE in 37m 31s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8c4a3f546e874f8289df6ff392d72d20 : SUCCESS in 1h 45m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3e9737c08a9b485aa4eb9535dc96e2f2 : SUCCESS in 39m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8cb19a0742474479a42e1c425e1d18c1 : SUCCESS in 27m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c779b6891d844c2bbc454e1b112ce59 : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/089b35cdbb404cf68b5646bbaae8259a : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4206a61d27f43adb0b3461edeb7bc0b : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a361621d3da4b5db1136077398e386e : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/979616f264b04d5b90a4abd5b7bd6066 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16c3405fb30b4a63b911e3e62cdcdede : SUCCESS in 6m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9c40fc07c9b048c7a7158f59c3f9a63a : SUCCESS in 32m 24s","accounts_in_message":[],"_revision_number":3},{"id":"688171b68e5cea3c7131e5f72e604aa7de0eb89b","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-07-23 07:13:26.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1c34f6c782d87701fcaeade25ba904a5af3354db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 08:23:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c29ee82e82384822ac5589e1fde96c30 : SUCCESS in 38m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/46363046ca324588a650515ac9c1073a : SUCCESS in 39m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ae32ac161a384d11b19e28e0b808bf1c : SUCCESS in 1h 02m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/04802b5b1eb14c728b2213f983d3d528 : SUCCESS in 34m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cb9ab645feaa4a90a8fa5180b06972cd : SUCCESS in 36m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ce0a8fe0291457cb749f1d7ffba99f6 : SUCCESS in 7m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39e56e14e4e242e290fc35a775890364 : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4462ab044c68454a819affa0af14d6f2 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f0470635dcd46469f1bf8cf24bf7c73 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77db4544971f400a8ddc0ed463ce319f : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/633fc9fa431440f390c2aebab7d2342b : SUCCESS in 10m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/28cf153203ef4eb88e3f7d5d98925bd0 : SUCCESS in 31m 43s","accounts_in_message":[],"_revision_number":3},{"id":"95fef9713bffa7476b0f1b1b5cbf3eea4e5c53ae","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-07-23 17:19:42.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7bd0c2dd7c6bdcde05c1bcd2db156218c00684ce","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-07-30 18:05:05.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"f4dab09486a3a4ee0114f683a20b0c0350832a10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 19:14:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c5b63ed631044bae9269d7c67af54fe3 : SUCCESS in 40m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fe7f4aa5d8a64504aa7fb73f53ca9950 : SUCCESS in 45m 21s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4bfca39e94d947d4ab6d1c4b4e754cab : SUCCESS in 1h 01m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/24ff689d6b754411b39aa8e01cd2a6f7 : SUCCESS in 45m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6b7649de26d74e7eba910975d59bcadf : SUCCESS in 29m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9e5f5c75703422d84969358727582db : SUCCESS in 4m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8501482d9e6c492ca0fb49026c672b54 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3958f8db34d4236ad885ef7c3ec18f6 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15bcdbf831964cb6bc3c61132c4b0f67 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbd3d6555037471a88d5064a100cfb7d : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/550fe6b952ce4b38a11c02e7c628bbc5 : SUCCESS in 6m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/917fa5508a22467fb386f1982b9b7f03 : SUCCESS in 39m 47s","accounts_in_message":[],"_revision_number":4},{"id":"fda7e06e4405829b88942c4599fd6705e7b7acf9","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-07-30 19:56:29.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM, Thanks!","accounts_in_message":[],"_revision_number":4},{"id":"2be64ed808e9798638261e306ac2f5c854d50209","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-08-13 15:12:39.000000000","message":"Patch Set 4: Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"5015c9603d55f25501aef5697d4bb15c08a0d233","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 15:12:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"697a805175e45ce6f57225d6a3081768be08cb19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 16:07:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"aa9d87c8d61defe85745f6952d81737ffa8a2cc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 16:07:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2e5badcdd7af4754a1f08f965b00249a : SUCCESS in 29m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e4ea7c8685114ca4898eee139a0d37f3 : SUCCESS in 27m 00s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/59053c2457bc47b086bbbdc1daa3f42e : SUCCESS in 50m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5a674459383443518866ef68f1842407 : SUCCESS in 36m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/84dfed1e8e574fbdaa3ace5cb4931018 : SUCCESS in 36m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e21c27cc0e740569092d0ea72ade0ec : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba92d58b1b8741abbc9f75501a9f3521 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07d0e1c9d0e54243a4cab8e7c9ad5580 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d56b6d8de1c4c75ade8e60b37745ccb : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94e340ef44674f018bc06e3c3769e7aa : SUCCESS in 8m 46s","accounts_in_message":[],"_revision_number":4},{"id":"935899b83f68c8320f1d456372e2a59e16ca1fd8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 16:08:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b3bbf848e6642629df20251eb3a4714 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1fa892ea5a6db061d415a3ebe2b94105aaba57fd","revisions":{"7a8474d1d806cdaaa3fd7f5b846c09a3c487552e":{"kind":"REWORK","_number":1,"created":"2020-07-14 21:08:14.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/54/741054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/54/741054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/54/741054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/54/741054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/54/741054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/54/741054/1"}}},"commit":{"parents":[{"commit":"522ab45d51d919e815946d9e4c4d86065e154f34","subject":"Merge \"Fix inspect module deprecation warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/522ab45d51d919e815946d9e4c4d86065e154f34"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-07-14 21:05:05.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-07-14 21:08:08.000000000","tz":0},"subject":"Fix multi messages AXFR with TSIG","message":"Fix multi messages AXFR with TSIG\n\nTSIG sign for multi messages AXFR was using the wrong dnspython\nsigning function which does not support sequence messages in\nresponse which cause the AXFR error below\nWARNING -- Some TSIG could not be validated\n\nChange-Id: I7ce84ac9cf5bc5f6fca47dc79c51fe0becf96ac6\nCloses-Bug: 1886685\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7a8474d1d806cdaaa3fd7f5b846c09a3c487552e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7a8474d1d806cdaaa3fd7f5b846c09a3c487552e"}]},"branch":"refs/heads/master"},"1e1f1e429ed0e9d8d431d5f13c71c1d9e530fe79":{"kind":"REWORK","_number":2,"created":"2020-07-14 22:26:39.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/54/741054/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/54/741054/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/54/741054/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/54/741054/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/54/741054/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/54/741054/2"}}},"commit":{"parents":[{"commit":"522ab45d51d919e815946d9e4c4d86065e154f34","subject":"Merge \"Fix inspect module deprecation warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/522ab45d51d919e815946d9e4c4d86065e154f34"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-07-14 21:05:05.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-07-14 22:25:41.000000000","tz":0},"subject":"Fix multi messages AXFR with TSIG","message":"Fix multi messages AXFR with TSIG\n\nTSIG sign for multi messages AXFR was using the wrong dnspython\nsigning function which does not support sequence messages in\nresponse which cause the AXFR error below\nWARNING -- Some TSIG could not be validated\n\nChange-Id: I7ce84ac9cf5bc5f6fca47dc79c51fe0becf96ac6\nCloses-Bug: 1886685\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1e1f1e429ed0e9d8d431d5f13c71c1d9e530fe79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1e1f1e429ed0e9d8d431d5f13c71c1d9e530fe79"}]},"branch":"refs/heads/master"},"76a06b8943490a02cc7f252d83e6ddc8a9b12f9a":{"kind":"REWORK","_number":3,"created":"2020-07-22 19:52:40.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/54/741054/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/54/741054/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/54/741054/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/54/741054/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/54/741054/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/54/741054/3"}}},"commit":{"parents":[{"commit":"522ab45d51d919e815946d9e4c4d86065e154f34","subject":"Merge \"Fix inspect module deprecation warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/522ab45d51d919e815946d9e4c4d86065e154f34"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-07-14 21:05:05.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-07-22 19:52:27.000000000","tz":0},"subject":"Fix multi messages AXFR with TSIG","message":"Fix multi messages AXFR with TSIG\n\nTSIG sign for multi messages AXFR was using the wrong dnspython\nsigning function which does not support sequence messages in\nresponse which cause the AXFR error below\nWARNING -- Some TSIG could not be validated\n\nChange-Id: I7ce84ac9cf5bc5f6fca47dc79c51fe0becf96ac6\nCloses-Bug: 1886685\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/76a06b8943490a02cc7f252d83e6ddc8a9b12f9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/76a06b8943490a02cc7f252d83e6ddc8a9b12f9a"}]},"branch":"refs/heads/master"},"1fa892ea5a6db061d415a3ebe2b94105aaba57fd":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-07-30 18:05:05.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/54/741054/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/54/741054/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/54/741054/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/54/741054/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/54/741054/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/54/741054/4"}}},"commit":{"parents":[{"commit":"45cb4376a24d244fa677f3736299b4effa092657","subject":"Merge \"Remove translation sections from setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/45cb4376a24d244fa677f3736299b4effa092657"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-07-14 21:05:05.000000000","tz":0},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2020-07-30 18:05:05.000000000","tz":0},"subject":"Fix multi messages AXFR with TSIG","message":"Fix multi messages AXFR with TSIG\n\nTSIG sign for multi messages AXFR was using the wrong dnspython\nsigning function which does not support sequence messages in\nresponse which cause the AXFR error below\nWARNING -- Some TSIG could not be validated\n\nChange-Id: I7ce84ac9cf5bc5f6fca47dc79c51fe0becf96ac6\nCloses-Bug: 1886685\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1fa892ea5a6db061d415a3ebe2b94105aaba57fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1fa892ea5a6db061d415a3ebe2b94105aaba57fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
