)]}'
{"id":"openstack%2Fdesignate~743610","triplet_id":"openstack%2Fdesignate~stable%2Fussuri~I9dddd4130a0cbb29311eeb52e077e216c8c03f3a","project":"openstack/designate","branch":"stable/ussuri","topic":"bug/1879798","hashtags":[],"change_id":"I9dddd4130a0cbb29311eeb52e077e216c8c03f3a","subject":"Update zones masters using pool target masters.","status":"MERGED","created":"2020-07-28 18:26:25.000000000","updated":"2021-03-05 05:37:57.000000000","submitted":"2021-03-04 22:59:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743610-bug/1879798","meta_rev_id":"8b6918cdf766fb2daa6e78ff2c4d0ef6a20d4dcf","_number":743610,"virtual_id_number":743610,"owner":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-04 22:59:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2021-02-26 15:01:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-04 21:45:59.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2021-03-05 05:25:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":1,"date":"2021-03-04 21:59:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2021-03-05 05:25:38.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-30 12:05:54.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-08-04 23:29:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-01 07:55:01.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-03-04 21:45:59.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-03-05 05:25:38.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"1f1d02ee08dfb4088ccb95d6b5d35e8eb807bb01","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2020-07-28 18:26:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"bbb2d2dc34a38a9901fe388dbb5db2101eb68508","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 19:25:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fb6c4e1e5ab9458c93dacdd5313fec25 : SUCCESS in 47m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/32171a3dc2c541af83065311ce55cf6d : SUCCESS in 42m 10s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c0a6e029188440e9b2990db7240dad2f : SUCCESS in 55m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3083991c92e440a1b7d93d3d1cbaebcf : SUCCESS in 46m 32s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/67919e96137743b2bb373cef88234aed : SUCCESS in 48m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7eb4e8f671943988266ff358b9a6221 : SUCCESS in 5m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86ad0e1c509c4a678ef95d96c25b0964 : FAILURE in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9f0aaec311f432f9a7fc865669d424b : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71fe0de9af3b4ac88e7c4a5aee1f9e7e : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/de1735eda20a44eab7253c947a6334e1 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8006e182a8774da6b530c25232216e40 : SUCCESS in 8m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c977d7467ba045438196d4894f90874d : SUCCESS in 45m 50s","accounts_in_message":[],"_revision_number":1},{"id":"0d902873ebf23382d3cac6c376a0078a07d655e3","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-07-30 12:05:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c2e6d9ca3939f55533de33a23c1e922b867841be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 13:13:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a1c5336655eb46a7a3e5ba335611e2db : SUCCESS in 43m 45s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f57f470b7405452ca533017c18f59544 : SUCCESS in 34m 17s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2b6f2424e48541d0a65320adfa43e27d : SUCCESS in 48m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fedd6e080cb84591b53ca4cf2cebed05 : SUCCESS in 40m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/007aa575ab604930be007b145e169c88 : SUCCESS in 38m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eac937c83995481ebfca0244b71a9949 : SUCCESS in 8m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f12aff8e8a9409a9023f23ea5f722e5 : FAILURE in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07cd5cb2bd8442eba653cc747cdb6fb4 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ead846179c884f3f81b042ca41cbda27 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8c4a16fb68c498eb4f68df6f0a8f4a3 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/290180f1c2324f6483805948f38a6167 : SUCCESS in 7m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6e04d077da3242af88918ded2feb07f7 : TIMED_OUT in 1h 03m 29s","accounts_in_message":[],"_revision_number":1},{"id":"67b47c71de98260e60deb8ef508c6f42b7cb9944","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2020-08-04 22:10:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"eef6c08464de19a88acef10eb148246d775c02b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 23:29:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/24b8b2b36ae74106a6babecaaa3147e4 : SUCCESS in 38m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/92890bf2cd384c489f4c89617e375ec6 : SUCCESS in 42m 54s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/86f80d9b1df444a08167b9f7db7a4a0c : SUCCESS in 1h 11m 49s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6596e17eb7824c959eed53513b0e2e34 : SUCCESS in 39m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5f1a105796254c62be49979e661a3714 : SUCCESS in 41m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/084c13810fc145d3898e29d87245db80 : SUCCESS in 5m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/798cdcba64e244cb91779f51670fb918 : FAILURE in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20defe7bed324b2fabda448d8e79cf1b : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6446bebedda942c2826554fa58555bb5 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/911841dee45447c88a46272f9ffa1f8b : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cf65e44021343c9affaef5f70217725 : SUCCESS in 7m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9d696f9e7f694ea6991f60a0dcdacd06 : SUCCESS in 35m 55s","accounts_in_message":[],"_revision_number":2},{"id":"952b6e10a8a7b6c449c2d38a573b4ca76fe543e4","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2020-08-05 22:25:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ab2326c6bd1f9a93e35976a5c95c507f2a88fa57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 23:26:02.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d6fc4fff47f64cc0b5d9b1d740de95e7 : SUCCESS in 32m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3a2ab6192293441f8f138d41951a0964 : SUCCESS in 33m 13s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/df310ef8c3f648ba8975f3f7e58d2ee9 : SUCCESS in 58m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5269efb735254fc9851184be5ad3ea27 : SUCCESS in 31m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c95b99a33cee400a9abdc0d48fd375d0 : SUCCESS in 34m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bab63652107346b28d8160d4e0efe762 : SUCCESS in 5m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56ae0863219445f7aef60659b036a567 : FAILURE in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e317badc4c9242da919cb4b2063bc2ed : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18d5acd4827a4025b462f6ef077c3d26 : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b9a03d7d1e3f42e88841d8f221520c14 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c77cdc53d374aaca665ee73ccfd1b2c : SUCCESS in 7m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c4f582cdcccc471a8a47b41ec83f3645 : SUCCESS in 32m 15s","accounts_in_message":[],"_revision_number":2},{"id":"ab661ae650b966048a07c2bfcc936d1993ec46bd","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2020-08-20 22:45:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"104d35311f9f984fe00b2d52400fabec60a14dda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 00:03:11.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/feda43c538b641ea939abce7e973a108 : SUCCESS in 39m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4bbbcf19a3b54413a192874e7cc9a32c : SUCCESS in 40m 08s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/79c8ab5325b84248804aab5819808060 : SUCCESS in 1h 07m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/73f28f48805a455098fb0a4bfc3cb432 : SUCCESS in 40m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cdcbcf8ca81c4b58861ff7a7887f0f8c : SUCCESS in 37m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f70da306805467b89e40b3f57f525a5 : SUCCESS in 5m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bad05f13de1949b98cac2de0e84d96a0 : FAILURE in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4462e3b201a4579a2b85378efc7679a : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ec8698d14d842379609369da89dc4b4 : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2eb563ad0a5f4c4da0e6f967a865976e : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adcdc7b6f9d942189f9f10e3f7c52540 : SUCCESS in 7m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/828090f1726d4a7886015b9828f0f6ec : SUCCESS in 25m 49s","accounts_in_message":[],"_revision_number":2},{"id":"619ff1253e663cf6a9e12861e9af357d7887da5f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-01 07:55:01.000000000","message":"Patch Set 2:\n\nIf this gets merged, please also cherry pick https://review.opendev.org/#/c/755349/.","accounts_in_message":[],"_revision_number":2},{"id":"a139fdf5eee35317281b8105faee3d4da744df4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-01-13 21:30:52.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"f3594892f3ed0a9bd10cf15d14220b10290d1c45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 22:50:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9d27ebdb3cb44742bd1b6a71a94c54e4 : SUCCESS in 57m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/405a60a14bee4971ad4f0cd07faa74b1 : SUCCESS in 53m 26s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d1b248cde75c4a3bb270b37c39dcce0f : SUCCESS in 1h 11m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1c83a0074c0e4c4fabcc93a80aa088c0 : SUCCESS in 33m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fc5a369462ca4730a6c20ffe9d0e456b : SUCCESS in 57m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb10fc1519b04e20b40f0e2d38f1a4c1 : SUCCESS in 4m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c8eeb46d273445fb4c772b64f27ec5d : FAILURE in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bfaff86838344fa9481a6b2e294b70c : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff1d1086560e4c42b217dd87296d97c2 : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dffc54c64e6149728169cfc3a9132e87 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d847320653248f593ba5a2554ce5d2a : SUCCESS in 7m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/31efe27bb87949fd93d7606509f39960 : SUCCESS in 38m 40s","accounts_in_message":[],"_revision_number":3},{"id":"7005bd069b7ffaf04c2ffb73c8e619217822c932","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-26 13:47:05.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"fb37699f1c7f22c2da468f87eeafeb4ea1e2ba3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 14:43:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d21648949af64f63bf729a4f175d8966 : SUCCESS in 34m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f08d705f2872464abe62fc85a6ffb49b : SUCCESS in 34m 37s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8462fcfc3e534ed28e915916926e2a6e : SUCCESS in 54m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/31a40ed33e634fbdaae4d34851cb3206 : SUCCESS in 33m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0e5159121ab445299e054017173072de : SUCCESS in 33m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06e14ed5ef9847db85da4b807e8aa298 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b69506547b134aa2bb08695d269405ca : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/641d1477ba514cb7ae874e30aa98ef44 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa99a9f909a646a192fc20aa9d927b2a : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2802c163d5ed460589e15f951d5716d4 : SUCCESS in 8m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3277eb66ec624a709e2f380dc51a7874 : SUCCESS in 26m 11s","accounts_in_message":[],"_revision_number":4},{"id":"56da550c462fda30be59bfdd4535dc082e5bdd64","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-26 15:01:15.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"b9848277060a922eea13a8a1afad17aac3f682e3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-04 21:45:59.000000000","message":"Patch Set 4: Code-Review+1\n\nI don\u0027t see anything here to block a backport.","accounts_in_message":[],"_revision_number":4},{"id":"17726526e5724e33ba91ca82550bd80b124dffdc","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-03-04 21:59:25.000000000","message":"Patch Set 4: Workflow+1\n\nThanks Michael!","accounts_in_message":[],"_revision_number":4},{"id":"c67554b05356b6f397715ed4c7399c54d166b202","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 21:59:49.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7ee1382a062ee8d14183556dc62e3ed32c025f56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 22:59:29.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ad31e4264a644a95b9a4e752304b2622 : SUCCESS in 37m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b73a481464604e2689a5ad3327badcd8 : SUCCESS in 30m 22s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ace15a06e1d64fd2ae75efcaf42943de : SUCCESS in 58m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6323d2cb809e4f1194f310995c832ba1 : SUCCESS in 33m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/df416a56514b4a639442327186d3211e : SUCCESS in 36m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f3ea31a1e36479ab430ce2e873e44b1 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5c50578ec694d5b8f3dad33ae5eadcc : SUCCESS in 8m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b95eaafa96c34893bb38c01568549eda : SUCCESS in 9m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b468e342b0d7409cb0069faaa2613226 : SUCCESS in 8m 36s","accounts_in_message":[],"_revision_number":4},{"id":"c8da3787ce45a9e078f9a2c129ecfeb415e7f951","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 22:59:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"17f254ae032f1f3d611f4054a5b64e23da3dfe10","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 23:00:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0882991e5694551b45fb43bbbccfcc2 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":4},{"id":"d662bc5ab1e173f3dcffde1c1f37fd8034c5a43f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-03-05 05:25:38.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8b6918cdf766fb2daa6e78ff2c4d0ef6a20d4dcf","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-03-05 05:37:57.000000000","message":"Patch Set 4:\n\nSorry, completely missed that this was approved and merged. It\u0027s really approved now! ;)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b967e9f706373f1aad6db882c2295fbbe1fadfc9","revisions":{"595ace59853ea9093ded69048fb648964a6fef77":{"kind":"REWORK","_number":1,"created":"2020-07-28 18:26:25.000000000","uploader":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"ref":"refs/changes/10/743610/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/743610/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/743610/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/743610/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/743610/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/743610/1"}}},"commit":{"parents":[{"commit":"d7e8b938d7105d7d51dc3f9000b398a3fd7e2c1a","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/ussuri\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d7e8b938d7105d7d51dc3f9000b398a3fd7e2c1a"}]}],"author":{"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","date":"2020-05-28 18:57:49.000000000","tz":-240},"committer":{"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","date":"2020-07-28 18:26:25.000000000","tz":0},"subject":"Update zones masters using pool target masters.","message":"Update zones masters using pool target masters.\n\nThis change enforces the update of the zone masters\nfor all zones that belongs to a particular pool, using\nthe pool\u0027s defined target(s) masters and forcing a update_zone\ncall.\n\nThis change also, moves the backend base class update_zone\nmethod as an abstract method, allowing to each backend\nimplementation to create its own update logic. For the\ncase of bind9 its extended to allow running a rndc modzone\nwith the new given masters for the zone fixing the behavior\nexposed on LP: #1879798.\n\nFixes-Bug: #1879798\n\nChange-Id: I9dddd4130a0cbb29311eeb52e077e216c8c03f3a\nSigned-off-by: Jorge Niedbalski \u003cjorge.niedbalski@canonical.com\u003e\n(cherry picked from commit 3756fc51e71aaf0ba7cfb9155ca5d1de26ab78bc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/595ace59853ea9093ded69048fb648964a6fef77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/595ace59853ea9093ded69048fb648964a6fef77"}]},"branch":"refs/heads/stable/ussuri"},"8a9301f13e0a2c822e8526454a650702ffd4765a":{"kind":"REWORK","_number":2,"created":"2020-08-04 22:10:01.000000000","uploader":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"ref":"refs/changes/10/743610/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/743610/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/743610/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/743610/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/743610/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/743610/2"}}},"commit":{"parents":[{"commit":"d7e8b938d7105d7d51dc3f9000b398a3fd7e2c1a","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/ussuri\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d7e8b938d7105d7d51dc3f9000b398a3fd7e2c1a"}]}],"author":{"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","date":"2020-05-28 18:57:49.000000000","tz":-240},"committer":{"name":"Jorge Niedbalski","email":"jnr@metaklass.org","date":"2020-08-04 22:09:59.000000000","tz":-240},"subject":"Update zones masters using pool target masters.","message":"Update zones masters using pool target masters.\n\nThis change enforces the update of the zone masters\nfor all zones that belongs to a particular pool, using\nthe pool\u0027s defined target(s) masters and forcing a update_zone\ncall.\n\nThis change also, moves the backend base class update_zone\nmethod as an abstract method, allowing to each backend\nimplementation to create its own update logic. For the\ncase of bind9 its extended to allow running a rndc modzone\nwith the new given masters for the zone fixing the behavior\nexposed on LP: #1879798.\n\nFixes-Bug: #1879798\n\nChange-Id: I9dddd4130a0cbb29311eeb52e077e216c8c03f3a\nSigned-off-by: Jorge Niedbalski \u003cjorge.niedbalski@canonical.com\u003e\n(cherry picked from commit 3756fc51e71aaf0ba7cfb9155ca5d1de26ab78bc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8a9301f13e0a2c822e8526454a650702ffd4765a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8a9301f13e0a2c822e8526454a650702ffd4765a"}]},"branch":"refs/heads/stable/ussuri"},"106fcc13ed18058205771a37d3d63c6a6b766c1f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-13 21:30:52.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/10/743610/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/743610/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/743610/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/743610/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/743610/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/743610/3"}}},"commit":{"parents":[{"commit":"9f28e73823dfc50a7b4fb3f8020520131b452eb2","subject":"Merge \"fix database sync always get the wrong  value\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9f28e73823dfc50a7b4fb3f8020520131b452eb2"}]}],"author":{"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","date":"2020-05-28 18:57:49.000000000","tz":-240},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-01-13 21:30:18.000000000","tz":-420},"subject":"Update zones masters using pool target masters.","message":"Update zones masters using pool target masters.\n\nThis change enforces the update of the zone masters\nfor all zones that belongs to a particular pool, using\nthe pool\u0027s defined target(s) masters and forcing a update_zone\ncall.\n\nThis change also, moves the backend base class update_zone\nmethod as an abstract method, allowing to each backend\nimplementation to create its own update logic. For the\ncase of bind9 its extended to allow running a rndc modzone\nwith the new given masters for the zone fixing the behavior\nexposed on LP: #1879798.\n\nFixes-Bug: #1879798\n\nChange-Id: I9dddd4130a0cbb29311eeb52e077e216c8c03f3a\nSigned-off-by: Jorge Niedbalski \u003cjorge.niedbalski@canonical.com\u003e\n(cherry picked from commit 3756fc51e71aaf0ba7cfb9155ca5d1de26ab78bc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/106fcc13ed18058205771a37d3d63c6a6b766c1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/106fcc13ed18058205771a37d3d63c6a6b766c1f"}]},"branch":"refs/heads/stable/ussuri"},"b967e9f706373f1aad6db882c2295fbbe1fadfc9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-02-26 13:47:05.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/10/743610/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/743610/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/743610/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/743610/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/743610/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/743610/4"}}},"commit":{"parents":[{"commit":"dc2426f7f234dac587b0725ad89f514f0afdcd06","subject":"Fix pdns4 devstack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dc2426f7f234dac587b0725ad89f514f0afdcd06"}]}],"author":{"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","date":"2020-05-28 18:57:49.000000000","tz":-240},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2021-02-26 13:46:54.000000000","tz":-420},"subject":"Update zones masters using pool target masters.","message":"Update zones masters using pool target masters.\n\nThis change enforces the update of the zone masters\nfor all zones that belongs to a particular pool, using\nthe pool\u0027s defined target(s) masters and forcing a update_zone\ncall.\n\nThis change also, moves the backend base class update_zone\nmethod as an abstract method, allowing to each backend\nimplementation to create its own update logic. For the\ncase of bind9 its extended to allow running a rndc modzone\nwith the new given masters for the zone fixing the behavior\nexposed on LP: #1879798.\n\nFixes-Bug: #1879798\n\nChange-Id: I9dddd4130a0cbb29311eeb52e077e216c8c03f3a\nSigned-off-by: Jorge Niedbalski \u003cjorge.niedbalski@canonical.com\u003e\n(cherry picked from commit 3756fc51e71aaf0ba7cfb9155ca5d1de26ab78bc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b967e9f706373f1aad6db882c2295fbbe1fadfc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b967e9f706373f1aad6db882c2295fbbe1fadfc9"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
