)]}'
{"id":"openstack%2Fdesignate~743611","triplet_id":"openstack%2Fdesignate~stable%2Ftrain~I9dddd4130a0cbb29311eeb52e077e216c8c03f3a","project":"openstack/designate","branch":"stable/train","topic":"bug/1879798","hashtags":[],"change_id":"I9dddd4130a0cbb29311eeb52e077e216c8c03f3a","subject":"Update zones masters using pool target masters.","status":"MERGED","created":"2020-07-28 18:26:36.000000000","updated":"2021-03-04 23:48:36.000000000","submitted":"2021-03-04 23:47:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743611-bug/1879798","meta_rev_id":"3fd14771761faefb1f16efaf7bee5d25c4688e9e","_number":743611,"virtual_id_number":743611,"owner":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":32438,"name":"Heather Lemon","email":"heather.lemon@canonical.com","username":"hlemon"},{"value":0,"date":"2020-10-01 07:54:53.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-04 23:47:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2021-01-27 16:50:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":1,"date":"2020-09-16 18:39:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32438,"name":"Heather Lemon","email":"heather.lemon@canonical.com","username":"hlemon"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-28 23:22:56.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":1,"date":"2021-03-04 23:00:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":32438,"name":"Heather Lemon","email":"heather.lemon@canonical.com","username":"hlemon"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":32438,"name":"Heather Lemon","email":"heather.lemon@canonical.com","username":"hlemon"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2021-01-28 23:22:56.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32438,"name":"Heather Lemon","email":"heather.lemon@canonical.com","username":"hlemon"}],"CC":[{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-20 23:53:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-21 13:51:52.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-09-16 18:39:14.000000000","updated_by":{"_account_id":32438,"name":"Heather Lemon","email":"heather.lemon@canonical.com","username":"hlemon"},"reviewer":{"_account_id":32438,"name":"Heather Lemon","email":"heather.lemon@canonical.com","username":"hlemon"},"state":"REVIEWER"},{"updated":"2020-10-01 07:54:53.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-01-28 23:22:04.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-01-29 00:00:34.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"}],"messages":[{"id":"c19a0a97f9db195ea4ee229708011ae2a265d037","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2020-07-28 18:26:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"96d1faa92a3a2a2b050c30c56bba2e7657de2a8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 19:47:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fada6c9d08a741e9bd1fa72d2befedd7 : SUCCESS in 34m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1529992d1c594056ae57cfb960035496 : SUCCESS in 40m 04s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/9edddca049444be2add3ed1acc76c224 : TIMED_OUT in 1h 13m 03s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/5c95f53252694f73bc6bfb3e1ec22025 : SUCCESS in 39m 18s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/7ff5a5612bf84663b39f9294359f128d : FAILURE in 21m 31s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a64abb31db4946b686f43cf41621d7b8 : SUCCESS in 43m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3d633edee4d442b4b6ed1128066ca95e : SUCCESS in 43m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ea10e07acc5748108173a9dd632db2e8 : SUCCESS in 38m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c93590fa00ef4f2f9107ae0e7bdf2350 : SUCCESS in 6m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81893dd534ae4410aba86ea1dad83610 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94abdcdf718c4647b6b4d9d40aa1833c : SUCCESS in 4m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/daa19f719a78489b971e34d707006d3e : FAILURE in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e9b521cbc3e4f528ef019cf48603797 : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b86dac96924b4c40afb0a5b76104b768 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ee92c9b4a8d4581bb91ece767787769 : SUCCESS in 9m 23s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/d49320a39b2c4f9cba4b404fd9b75d9c : SUCCESS in 38m 06s","accounts_in_message":[],"_revision_number":1},{"id":"0c18cde5c7aff099d03b34f6bf28c47d10d4e34e","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-07-30 12:05:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c828489b137c54fbd8fc8362f0d3581cf18de949","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 13:24:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1b15ae55ce8f41689ac05dcba7dce947 : SUCCESS in 43m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c27564e1c2e644c3b9a9ccf54a1a931f : SUCCESS in 39m 18s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/245427962bda48058a86103af1eceed8 : TIMED_OUT in 1h 14m 16s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/8cd3d51d42994277859d0002ea5bf027 : SUCCESS in 55m 31s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/59613dc7865844efb45e6096726a79a4 : FAILURE in 27m 27s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/984e9e9ba58e499cad386ce80519f384 : SUCCESS in 1h 04m 01s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1d7c8476a8c04114820f97ddfe1265ef : SUCCESS in 39m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2490ce055d044fd582137dc615a0b00c : SUCCESS in 45m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8823e77897440eeb432aa7764265812 : SUCCESS in 5m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16172a2d006a4da490f9c656c66e47eb : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a11d522d2f124aeb9c395de8435ef585 : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ae82497f2ffc49c6b1eb08f20097505c : FAILURE in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bb10d80819447aabd6b840def0158c9 : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f957be99dcd74e09aa7655383a1b7d61 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa48f9a18308445d89d108ca807bc3f6 : SUCCESS in 7m 09s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/f8d3450387524ecb8fc973fc060b2289 : TIMED_OUT in 1h 02m 38s","accounts_in_message":[],"_revision_number":1},{"id":"95d30744884108852bf5a42f832a3419883b1bcf","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2020-07-30 15:53:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9cd8c192691e5ce2c6f76dab67a26f85462122b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 17:14:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3155fa002420416babe0be77f0c23042 : SUCCESS in 36m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/193aa9915c674b8e8df3971507137b29 : SUCCESS in 35m 02s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/99984ad6492d4d41821668da0930d621 : TIMED_OUT in 1h 12m 53s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/e865734b8a7c4712aaefd608cc815979 : SUCCESS in 40m 35s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/b42813e4a8584fac98237c8471466069 : FAILURE in 28m 40s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6103537bde4f45b998f5c7fcefe28b3b : SUCCESS in 1h 11m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/44db95e695c843938d5b3f77f6aeb028 : SUCCESS in 44m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3bb89dd6cef04020bb88bc288fd1d416 : SUCCESS in 28m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d819fee2c0724334877dce4c35143346 : SUCCESS in 5m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa7ceb4c92ac46bd93fdbcfde82c1cb5 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16914df0c2894c4d90a63f694d82e008 : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/474101f6013043908ae6a4b025fba1d7 : FAILURE in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58cbafb8fd0e46f1bfcde21a580c866c : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ff552c6cfbe406f8c78888db88cb974 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc28b41bc0b24152b289a26d9eeab698 : SUCCESS in 7m 02s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/1bac60a2d13c4d2fa8e04eae75d3d5f1 : SUCCESS in 40m 29s","accounts_in_message":[],"_revision_number":1},{"id":"b2018ea82d803ab1cf61096a140d48c6e878f01e","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2020-08-04 22:09:40.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"fdb320b6d6ca7e6b9ca84a519e513311aad8262b","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2020-08-04 22:15:52.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"6c0939913f12574cc999ea8ea3aaaafec130defb","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2020-08-04 22:16:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b17b87b74743b7ec70733c3f8b15ac2e2a3aba5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 23:38:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6097563b101b409ebc51209b9818e95e : SUCCESS in 38m 54s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0a611896f2eb4e59bbf5756ff7e269f5 : SUCCESS in 41m 00s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/5607a14f99d14080b56a06e76066427c : TIMED_OUT in 1h 14m 34s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/65c6d041f29c402ca742c008a215a797 : SUCCESS in 40m 03s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/cc97a5c938084144a3caa7754cc1ad06 : FAILURE in 31m 26s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/135f757105494fa3b74cef6ab1833781 : SUCCESS in 1h 03m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/638f085108f34e55969e1cd3fcee84bc : SUCCESS in 41m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0b6931446a42483ea78d3cbe6ca675fd : SUCCESS in 43m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b341f83473c9486dac2cb9155a6f1a93 : SUCCESS in 6m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abe7ad6fe040438dba84841731521686 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b2913f197cb49feb9d0ce26530dbab0 : SUCCESS in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/39f8e17789b24519816d03bc529337ef : FAILURE in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13c9c1087b4041c28b4f76dd694d7971 : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b9e05bce71864522b64a536bcf4b64e0 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e10efd921b0f496a82fff3b539fdc01a : SUCCESS in 7m 10s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/077a95c83c5547d48547d3cf1945b00c : SUCCESS in 45m 21s","accounts_in_message":[],"_revision_number":2},{"id":"06c67e3861fecfd1d0d02afb53b8dc0200e97a26","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2020-08-20 22:43:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"369746950866de60218dd3fce511276b5a6087af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 23:53:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/51d22a5e631d4c7dad7a4a1ad9450765 : SUCCESS in 28m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/297a6fbc48ea4780b1e9ead372bd17b7 : SUCCESS in 29m 10s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/e38dbcf39e984e798bf0428014353b71 : SUCCESS in 46m 44s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/52045df67c71413abbb6fc0f9bca4bd0 : SUCCESS in 35m 44s\n- designate-pdns4-postgres https://zuul.opendev.org/t/openstack/build/fdf938fa3e454137b3649f727ba766bb : FAILURE in 25m 30s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/723f069619384410a839e9138a6b1f43 : SUCCESS in 1h 04m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/152ac162e2264847b3b0a9be77e5185f : SUCCESS in 39m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1c74b533b3d04cc3bd3bf1f5c9ee53d5 : SUCCESS in 36m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bd35ec19af54f73ab19fec374ad62c6 : SUCCESS in 5m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6926216636b5488caeca36264a0db5e3 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2097e289ef5242619f4079b7f53f3542 : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f9d475ff6cb549d097f96bbf8c292f19 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e378980a99e4fd0ad41f528a5dbf44d : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed4ade67091d4f2c9754bbf5b74ae2f4 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d58276dfa854077a7eb61337fa3cef8 : SUCCESS in 8m 38s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/540b75d1fe864303aaffebc52be79e96 : SUCCESS in 35m 18s","accounts_in_message":[],"_revision_number":3},{"id":"b0d409620d2ffbc0fcd910595e31bb9347211b63","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-08-21 13:51:52.000000000","message":"Patch Set 3: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"c0ce4ba9281acd29d865d80bc0aca074992e09c8","author":{"_account_id":32438,"name":"Heather Lemon","email":"heather.lemon@canonical.com","username":"hlemon"},"date":"2020-09-16 18:39:14.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"fe9f7cbb75894e9bcf4c3fbb986533c705204531","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-01 07:54:53.000000000","message":"Patch Set 3:\n\nIf this gets merged, please also cherry pick https://review.opendev.org/#/c/755349/.","accounts_in_message":[],"_revision_number":3},{"id":"573653e4229bb2edf9792f4a2ce05448f53a89eb","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-01-27 16:50:57.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7990cbbcfd3cda44b771f730856d2ea4cb67a4d5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-01-28 23:22:04.000000000","message":"Patch Set 3: Review-Priority+2\n\nI wish we had a release note with this, otherwise looks ok for a backport.","accounts_in_message":[],"_revision_number":3},{"id":"95a2ee17852a27cb6a26e7b46dc11978e1e5b843","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-01-28 23:22:15.000000000","message":"Removed Review-Priority+2 by Michael Johnson \u003cjohnsomor@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":3},{"id":"2e9c73e017eeb1d30d3cd5a67519dca1cacdda30","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-01-28 23:22:56.000000000","message":"Patch Set 3: Code-Review+1\n\nI wish we had a release note for this, but looks ok for a backport.","accounts_in_message":[],"_revision_number":3},{"id":"a93fa0767c0f55df06689b878cf379050b00377d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-01-29 00:00:34.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3: Code-Review+1\n\u003e \n\u003e I wish we had a release note for this, but looks ok for a backport.\n\nI am fine with it as well, but we need to merge the Ussuri patch first.","accounts_in_message":[],"_revision_number":3},{"id":"027f22d74d132c74b616b4f9fa3dfc491cede8f6","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-26 15:38:26.000000000","message":"Patch Set 3:\n\nThe Ussuri patch passed CI and is ready to go","accounts_in_message":[],"_revision_number":3},{"id":"e718b62be341f2b33b458fab52cd8c5234b84148","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-03-04 23:00:06.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"92f65aa265aaf4b42605ec0048c23ae5d39fcc53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 23:00:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"de47bd74510fb59e93ea808aa18db9bc4f2a079b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 23:47:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/45e42dcdb20b48bea63f1e9ca35b2fde : SUCCESS in 30m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/76c9381752f84e1ea6609044b0f021af : SUCCESS in 28m 12s\n- designate-bind9-py27 https://zuul.opendev.org/t/openstack/build/cbd971a8dd18459f9de3059eeb94f945 : SUCCESS in 33m 07s\n- designate-pdns4-py27 https://zuul.opendev.org/t/openstack/build/055c8decf5c143f4a21c6182ac17ddb6 : SUCCESS in 33m 52s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f32323fc89e1470cbff8456810c1309f : SUCCESS in 38m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c884aedbb74245bdabd6600cb36b1a53 : SUCCESS in 42m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/10412b96f54f458d8769b9f619403ce5 : SUCCESS in 44m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b890132508bf42718defd092123fb218 : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2008503bfa894815ab41244f9a34f39e : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66be6056b32b43d0ba0b4a7065bcc675 : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b63bdbb19a24a8bbf4ff14ec8ea7b76 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04c1306ada634eca959fea155b39682f : SUCCESS in 8m 23s","accounts_in_message":[],"_revision_number":3},{"id":"b179546da5ae2d5724b6fc94c4d95257bc7ba43f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 23:47:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"3fd14771761faefb1f16efaf7bee5d25c4688e9e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 23:48:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/469288b136994ebabb61eb7cb2997ed6 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"953492904772933f5f8e265d1ae6cc1e6385fcc6","revisions":{"4d963adaaeb7fa672837cafae6288502c6b24041":{"kind":"REWORK","_number":1,"created":"2020-07-28 18:26:36.000000000","uploader":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"ref":"refs/changes/11/743611/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/743611/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/743611/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/743611/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/743611/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/743611/1"}}},"commit":{"parents":[{"commit":"4fc04ee416d13f7e4e27c6b522b42a3c69c271b5","subject":"Merge \"Adding distributed locking to central\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4fc04ee416d13f7e4e27c6b522b42a3c69c271b5"}]}],"author":{"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","date":"2020-05-28 18:57:49.000000000","tz":-240},"committer":{"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","date":"2020-07-28 18:26:36.000000000","tz":0},"subject":"Update zones masters using pool target masters.","message":"Update zones masters using pool target masters.\n\nThis change enforces the update of the zone masters\nfor all zones that belongs to a particular pool, using\nthe pool\u0027s defined target(s) masters and forcing a update_zone\ncall.\n\nThis change also, moves the backend base class update_zone\nmethod as an abstract method, allowing to each backend\nimplementation to create its own update logic. For the\ncase of bind9 its extended to allow running a rndc modzone\nwith the new given masters for the zone fixing the behavior\nexposed on LP: #1879798.\n\nFixes-Bug: #1879798\n\nChange-Id: I9dddd4130a0cbb29311eeb52e077e216c8c03f3a\nSigned-off-by: Jorge Niedbalski \u003cjorge.niedbalski@canonical.com\u003e\n(cherry picked from commit 3756fc51e71aaf0ba7cfb9155ca5d1de26ab78bc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4d963adaaeb7fa672837cafae6288502c6b24041"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4d963adaaeb7fa672837cafae6288502c6b24041"}]},"branch":"refs/heads/stable/train"},"0eb828e8be3a077f564cf53de5c1fb9e5c1f49ba":{"kind":"REWORK","_number":2,"created":"2020-08-04 22:16:01.000000000","uploader":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"ref":"refs/changes/11/743611/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/743611/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/743611/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/743611/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/743611/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/743611/2"}}},"commit":{"parents":[{"commit":"4fc04ee416d13f7e4e27c6b522b42a3c69c271b5","subject":"Merge \"Adding distributed locking to central\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4fc04ee416d13f7e4e27c6b522b42a3c69c271b5"}]}],"author":{"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","date":"2020-05-28 18:57:49.000000000","tz":-240},"committer":{"name":"Jorge Niedbalski","email":"jnr@metaklass.org","date":"2020-08-04 22:15:21.000000000","tz":-240},"subject":"Update zones masters using pool target masters.","message":"Update zones masters using pool target masters.\n\nThis change enforces the update of the zone masters\nfor all zones that belongs to a particular pool, using\nthe pool\u0027s defined target(s) masters and forcing a update_zone\ncall.\n\nThis change also, moves the backend base class update_zone\nmethod as an abstract method, allowing to each backend\nimplementation to create its own update logic. For the\ncase of bind9 its extended to allow running a rndc modzone\nwith the new given masters for the zone fixing the behavior\nexposed on LP: #1879798.\n\nFixes-Bug: #1879798\n\nChange-Id: I9dddd4130a0cbb29311eeb52e077e216c8c03f3a\nSigned-off-by: Jorge Niedbalski \u003cjorge.niedbalski@canonical.com\u003e\n(cherry picked from commit 3756fc51e71aaf0ba7cfb9155ca5d1de26ab78bc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0eb828e8be3a077f564cf53de5c1fb9e5c1f49ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0eb828e8be3a077f564cf53de5c1fb9e5c1f49ba"}]},"branch":"refs/heads/stable/train"},"953492904772933f5f8e265d1ae6cc1e6385fcc6":{"kind":"REWORK","_number":3,"created":"2020-08-20 22:43:32.000000000","uploader":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"ref":"refs/changes/11/743611/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/743611/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/743611/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/743611/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/743611/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/743611/3"}}},"commit":{"parents":[{"commit":"4fc04ee416d13f7e4e27c6b522b42a3c69c271b5","subject":"Merge \"Adding distributed locking to central\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4fc04ee416d13f7e4e27c6b522b42a3c69c271b5"}]}],"author":{"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","date":"2020-05-28 18:57:49.000000000","tz":-240},"committer":{"name":"Jorge Niedbalski","email":"jnr@metaklass.org","date":"2020-08-20 22:43:03.000000000","tz":-240},"subject":"Update zones masters using pool target masters.","message":"Update zones masters using pool target masters.\n\nThis change enforces the update of the zone masters\nfor all zones that belongs to a particular pool, using\nthe pool\u0027s defined target(s) masters and forcing a update_zone\ncall.\n\nThis change also, moves the backend base class update_zone\nmethod as an abstract method, allowing to each backend\nimplementation to create its own update logic. For the\ncase of bind9 its extended to allow running a rndc modzone\nwith the new given masters for the zone fixing the behavior\nexposed on LP: #1879798.\n\nFixes-Bug: #1879798\n\nChange-Id: I9dddd4130a0cbb29311eeb52e077e216c8c03f3a\nSigned-off-by: Jorge Niedbalski \u003cjorge.niedbalski@canonical.com\u003e\n(cherry picked from commit 3756fc51e71aaf0ba7cfb9155ca5d1de26ab78bc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/953492904772933f5f8e265d1ae6cc1e6385fcc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/953492904772933f5f8e265d1ae6cc1e6385fcc6"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
