)]}'
{"id":"openstack%2Fdesignate~744013","triplet_id":"openstack%2Fdesignate~master~I296c5f085cce27033461be81ca58c85f095df89a","project":"openstack/designate","branch":"master","topic":"restart-if-needed-ib","attention_set":{},"removed_from_attention_set":{"19298":{"account":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"last_update":"2022-11-14 17:24:27.000000000","reason":"Change was submitted"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2022-11-14 17:24:27.000000000","reason":"Change was submitted"},"29543":{"account":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"last_update":"2022-11-14 17:24:27.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-11-09 15:45:46.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2022-11-11 18:13:54.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I296c5f085cce27033461be81ca58c85f095df89a","subject":"Reload Infoblox config after deleting zone","status":"MERGED","created":"2020-07-30 16:35:39.000000000","updated":"2022-11-14 17:25:28.000000000","submitted":"2022-11-14 17:24:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"744013-restart-if-needed-ib","meta_rev_id":"3b6339a9ae73b02cbc37888266067aef1c07d035","_number":744013,"virtual_id_number":744013,"owner":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-14 17:24:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2022-11-14 15:34:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-14 16:31:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-11-11 18:13:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-14 16:31:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-31 14:47:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-24 18:17:53.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-08-24 22:48:27.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-08-24 23:14:38.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-11-07 20:01:50.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2022-11-09 16:41:14.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"}],"messages":[{"id":"6fdd30f16bcaf939cb32bde0d03f3b7a7258a3f5","author":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"date":"2020-07-30 16:35:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe186162c673afd841dfce76156c381621f1b730","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-07-30 18:03:59.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7338f08875664124a39eee84ab05fb378ce6ebe6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 18:13:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5011b2a712124e8eb618a307d9d9e56b : SUCCESS in 35m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bbd59eb64f4944d0a2c65b0e15bbde64 : SUCCESS in 27m 58s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/85a25cff0bf44107b5ce7b4f09591b51 : SUCCESS in 1h 14m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7fd3683cb3b84af59802e82347cd5682 : SUCCESS in 39m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/deb65897893145738f15433ddebd7dfd : SUCCESS in 34m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/095c069d715d41ceaee880461548d38f : FAILURE in 5m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ad57d66b68e4b02bca070d9e9af2f18 : FAILURE in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd19b9cea2504d61938fe48cc5399b4c : FAILURE in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e815b5b2df3c4259b91a5582097abda2 : FAILURE in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b0ac7b4c3994d3b8b3d22ecb3fae50a : FAILURE in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f37f47478054405b453c324222c629d : SUCCESS in 6m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/73f8731585f84acfa23e2cd8cf00aa01 : SUCCESS in 38m 10s","accounts_in_message":[],"_revision_number":1},{"id":"4a7ca2e107486b60a0e71ee3722677542dc5392d","author":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"date":"2020-07-30 19:20:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2112fd585c89ccaf7260bce0ff43c23ed04e45f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 20:28:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f2c079669a074cc59551977418f1855a : SUCCESS in 38m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0a23035128b444a2a64b09e2f530296f : SUCCESS in 34m 13s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0e6583251fd14632b45e29a01a60465b : SUCCESS in 59m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d6058c84e2464a4296432fc6d0635728 : SUCCESS in 27m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/465618852e3e443bb71a920e513bef49 : SUCCESS in 37m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef69db95448e4031a2dcf1055f530bb2 : FAILURE in 5m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/299f08b8aeab4dc099338080953b536a : FAILURE in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/255fb9bb56684d538bc4b10a07f7eb84 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/063fa3e8711a477b93ee482b6eb33546 : FAILURE in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50bdd8ae02e1489c9a4924a14451cae9 : FAILURE in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c4b95d0d6cd4e15a83514bf3969a21d : SUCCESS in 7m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c843e8997a074511af632603eb33553e : SUCCESS in 32m 07s","accounts_in_message":[],"_revision_number":2},{"id":"5bd1cd127df5f1e5b606cc3df19717934015a0cf","author":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"date":"2020-07-30 22:53:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a3f8448f712478831c57b233526816c70f1ca86d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-07-30 23:06:02.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6e002f9c08452217d1454d9d23c6fb1669fff7c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 00:08:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/540dfc83c45d46fba4fcc7c3d44e915e : SUCCESS in 40m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/307c33586ce84a2db5bbaa9b038fef77 : SUCCESS in 38m 20s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7810a5bcbf384602a089645947234ae6 : SUCCESS in 1h 04m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/88476a21fc984a2dba8382b985311c9a : SUCCESS in 32m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1fcdc590f22e4421b7d1f1d4b0a0f798 : SUCCESS in 41m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7d38a84c4574cfcb6f65409fa39722c : SUCCESS in 6m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6fd1da242194a7aacf57a752227c0bd : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce0fbe342fb44ba7bce3bb0b4fc76515 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0830297f798c401f8ac226379906edae : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b586b4c0ac84708a9761037083b7cca : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/515afc9b406149e989794170451fe267 : SUCCESS in 7m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ce616e6023d64843a4e5abd538528444 : SUCCESS in 38m 10s","accounts_in_message":[],"_revision_number":3},{"id":"70d35628f0b10fa642aea69cf60f22fc2410cd8b","author":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"date":"2020-07-31 08:08:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b666020bf26528ed77a84ce0a6c199f260bbaa44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 09:09:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4ca4f3d1bf2d41969180611cd3193518 : SUCCESS in 56m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a61648fbf09a4f69b29966397c55473b : SUCCESS in 37m 25s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/129e7a3fb73a4d60832b119e8638325a : SUCCESS in 55m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/dd5affb19d3c47fcbb4296cfa3743607 : SUCCESS in 43m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a55ee21bef0e4722993b830394f4757b : SUCCESS in 57m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7930c7b9a374a5da2acaebb4221386b : FAILURE in 5m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26845a2c8e8f48deb737626277a5ebbb : FAILURE in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9977798aafc0414386e5b45e217608b0 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c91211bd1214131a29a36b414a55bb6 : FAILURE in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37395e1cd5474f0ca0e63647f19295ab : FAILURE in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c7b5c2f678e4ce29e11ae0db0bd392f : SUCCESS in 7m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c90ca516e1014a55a8ea8c74aabf26f0 : SUCCESS in 32m 25s","accounts_in_message":[],"_revision_number":4},{"id":"d2e2c5a920479a56c108baa028e1e660cb8cd751","author":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"date":"2020-07-31 13:57:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a29ac23d7fe07e8a959c040bf0597ded5a06dd3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 14:47:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f696c5566da1423fab010591f6bfaacc : SUCCESS in 27m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e0d1dd3018c14ea784842c81aff091dc : SUCCESS in 27m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/045d339065634a84b857fb16fd979c5a : SUCCESS in 40m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c47544f3eea84f269e21ab24824fdfda : SUCCESS in 34m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ce96430c209943b38d9d40b86196836c : SUCCESS in 42m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ee0ee9768f545fdb7e8c638bc160281 : SUCCESS in 5m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b09a3014a17941a3a6f63ddfef0bd240 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d96f58369274dc99bd106697cecedcf : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/529c1a42d902441f85f30a8109d96e06 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5464ca49bfe4678b44fb98270ba0629 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b72ee3977a54b4d861d1271f012622f : SUCCESS in 7m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bd94a92225d849eb9d65a1cae9c5a11d : SUCCESS in 38m 25s","accounts_in_message":[],"_revision_number":5},{"id":"191242e1257f618f34b4131f7202e200639f5607","author":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"date":"2020-08-24 09:08:23.000000000","message":"Patch Set 5:\n\nbump","accounts_in_message":[],"_revision_number":5},{"id":"10e016241a3ba98ebf015c810392b3d891183300","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-08-24 18:17:53.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)\n\nThanks for the updates!","accounts_in_message":[],"_revision_number":5},{"id":"75dfb379a03b7f69fec8635a81e8b6c370c4847e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-24 22:26:11.000000000","message":"Patch Set 5: Code-Review+1\n\nLGTM,\nI\u0027m not sure I see how a useful test could be added here without a backend available, but maybe I am missing something.","accounts_in_message":[],"_revision_number":5},{"id":"737a1f257fd360ac6f1d8d9bd83ee87747a02b29","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-08-24 22:48:27.000000000","message":"Patch Set 5:\n\n\u003e LGTM,\n \u003e I\u0027m not sure I see how a useful test could be added here without a\n \u003e backend available, but maybe I am missing something.\n\nI like testing all paths if it isn\u0027t too much trouble to help catch typos in things like log lines (e.g. the initial missing s in the string format would have lead to an unexpected exception).","accounts_in_message":[],"_revision_number":5},{"id":"dcb4b6b0227f6e7c100382a4994a4a46637879f0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-24 23:14:38.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)\n\nYeah, I am picky on coverage as well, but just reading the patch it looked like the ask was to test log lines. I did however miss that the actual call path is not being hit in the unit test, so that should be fixed for coverage.","accounts_in_message":[],"_revision_number":5},{"id":"ab9cf684f1f3c35e1eb8cfacd4c2827358235d6b","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-08-25 22:28:26.000000000","message":"Patch Set 5:\n\n@Michael: Wouldn\u0027t we be able to at least mock the _get_infoblox_object_or_none call and generate an error condition this way? I was more suggesting to test the strings that were not quite correct in some version of this change as opposed to testing the actual backend or the code that talks to it.","accounts_in_message":[],"_revision_number":5},{"id":"6972dc3184b6c3f41baafe452ea41063082866da","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"date":"2022-09-21 14:35:24.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":5},{"id":"cf9cb0a78c9029be44554aaa24ec5e7428850057","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"date":"2022-09-21 14:36:30.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"9df5a1b9f728a8be29ebc9c62dee63391cac815d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 16:02:05.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48cd518e96a2455aaf6d28ff91006aaa\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0e60d78b9ed2475f9c1f41781e27d613 : SUCCESS in 54m 12s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/9600495176da4e44aeba94b70c206899 : SUCCESS in 1h 19m 51s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/719a9b232add487e845d343711677f58 : SUCCESS in 1h 01m 39s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/3922e581b30a4e5f961cf58df2f645b8 : SUCCESS in 43m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/05c9e1641bf34b0d9188db6e3d263d83 : SUCCESS in 43m 22s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a826f0d84218489eae244c123c78dce2 : SUCCESS in 59m 23s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d1cf8c8f4c4f45718938c8bf000ce256 : SUCCESS in 1h 06m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7e243641e7e447b3aeeb84029ce32e48 : SUCCESS in 51m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2ce4ae5c7bc847d398a9ff432cbb4735 : SUCCESS in 53m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff13e16e16df4eecae89105dd082b44d : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f5eed8b95ab42adb57f63f6d66a2d59 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60d202911b0a41849838c9ba8c9d9295 : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a57bf0b02a845a19a83e9ba7b9ab5ab : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96833d080ce84098aee169b7a7dff627 : SUCCESS in 9m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/25d4041afe4d45bb9369d33520d959d0 : SUCCESS in 52m 44s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/16aefa5e07fa49c5a6de696890ef4bc6 : SUCCESS in 5m 21s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1c49aad9141e4a45eedd3cb874beb4c598441e2a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"date":"2022-09-21 20:35:53.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0264e97490e53a4fa4ed36e61968ff3effef782b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 21:57:19.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3005cf3391940b397e0e1c7ff68a11e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/74249977157048e6ae4b832c1b23f8ee : SUCCESS in 49m 01s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d536a80d4c6f4901b8914136cbc62b17 : SUCCESS in 1h 16m 08s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d949cea215114c93b0c69d425cce85eb : SUCCESS in 1h 01m 40s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/b2c0bb5da59f46d6a5cc0ef5271f918e : SUCCESS in 58m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3e68b7120c6e4dba904b72bda106c122 : SUCCESS in 50m 40s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a458ba3d559140928548dcc731b99691 : SUCCESS in 1h 12m 34s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/85b85c7a743e40bb80a0068a0d60301e : SUCCESS in 1h 09m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/19959ea39115473bb51d56fbc7813929 : SUCCESS in 49m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4bf9522fa498428caf8b4af99c81939c : SUCCESS in 51m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e7c39ec542744f48011c35bed938c1d : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cba4d3ad3b9540afba91e9290e7d409b : FAILURE in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d586d0d1e84427cb738dcc2314a569d : SUCCESS in 7m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/51b7bbfe55a348ac8f0b957da77521bf : SUCCESS in 9m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81b296b84f3e42959ae14492031a3cf1 : SUCCESS in 10m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/62c8ca271ecd4656bee20ef2ebe70292 : SUCCESS in 31m 44s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/bb2784c3eff14e81b4beba2741fd2dbf : SUCCESS in 6m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9f5915b2b44d23ac02673ae34c733f082a3839f9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"date":"2022-09-22 16:34:59.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"fa12eeebdb5bf377f104274215a91fa53eb8f4d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 17:52:06.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/37be89209e2445a9be7e979b1f5146d7\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/659f2cb80cbc41abab5d638ab30fccb8 : SUCCESS in 48m 12s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4670e2f0cffe4028b5a1c087efb7dea2 : SUCCESS in 1h 12m 59s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c556ae5a3b034119999db22f09c12def : SUCCESS in 57m 03s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/04e14a329229401594b6b33f00d2639e : SUCCESS in 52m 45s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d5114760deca4e5c99a386a6b7e2e8e4 : SUCCESS in 53m 02s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8488b29307db4adbba8938365970b1cd : SUCCESS in 1h 04m 29s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/531989ac4a4a4d4bb3a9a98ccc23ed77 : SUCCESS in 1h 11m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/45433d63a4a843e89bcdba74c02d783d : SUCCESS in 47m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ed10e563881548e196a071bcb14296b3 : SUCCESS in 52m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4fbdd4455304628b59af63facad86ad : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd979cf54bea4bc5bf377305671974ea : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4537812a8f9e4ccb93d0cdd06ea280e7 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/90f268b8773740819adb164357d89653 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fa0f9c2660a4506bde67ce140e56798 : POST_FAILURE in 19m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ba73e544cbdc4aa68acff81210ac1df5 : SUCCESS in 51m 50s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/13c06bc4b28a49baa36ef24eb7ea555a : SUCCESS in 5m 17s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"524a8e42a741aaa9c0c80b48d345522580e39fa9","author":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"date":"2022-09-23 08:58:19.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7a9f53241cf7543f2898d2f2fd0fb3f627fab35a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 10:19:09.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6efef1c5cdb34d009e1eb434b63b2b76\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b0a0c60bd23046a7a3db53465bba1472 : SUCCESS in 50m 59s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/734bbc327fb642a08253afc9ee2ce479 : SUCCESS in 1h 09m 25s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/72884fda17984c91b58d4671c19a30af : SUCCESS in 1h 03m 46s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/08cd92011d1e4e3cb485928702071ccb : SUCCESS in 41m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4fa026bc800a4a15a20a9b667e6a975c : SUCCESS in 35m 38s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/972a967ee20546739bf42a0eb6c88bcd : SUCCESS in 1h 13m 40s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/aba416e2ac764a0ea396c910f226326e : SUCCESS in 53m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f9b401e1316d4b3da9fe35ebf6ffa0fa : SUCCESS in 44m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a8580eb620da4605a9ed964bed63557f : SUCCESS in 47m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed6d057d031245149ecfcf9ebb6b1ab0 : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fdc882aa88a47238e3923149218397f : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3488841bdc8442598b0c07174f0c6f9e : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/acec0a1b119241249731bdf26b53c31c : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce4f9b56a1e14312bb3fdd932137bdad : SUCCESS in 9m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/425415b74761414c922f9747973ff0a2 : SUCCESS in 46m 18s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5ad1539371114d06a1a330f9b1dace06 : SUCCESS in 5m 13s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4c1ec44c9009065f71be2532bec35b94d1807123","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-11-07 20:01:50.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"70c87ba186529cb46e8d37c216852b31779bcca0","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"date":"2022-11-07 20:40:39.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":8},{"id":"9ab46c6937c8cec2b7aba90d9784eda2732d8ab8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-11-08 18:35:19.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"0dec8bdb900b75ee6151369260ff8af7c88962ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 19:47:58.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5fb77fcabe546349c6b5fbf9f2ad951\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/985f0548e3464794b138f58f4b6a1813 : SUCCESS in 36m 38s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d49a5af1076a4e02ab9b80dee471f5f1 : SUCCESS in 56m 01s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/20a018f611f44b7490532ec4d073ed06 : SUCCESS in 42m 54s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/ba19971e2e4a44db8c2ae895d3cb2a27 : SUCCESS in 37m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9fd374d0f7df4d84baac27722bdc7912 : SUCCESS in 39m 54s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/63edcb4803724430825f57bf73b5b430 : SUCCESS in 59m 10s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e575d6879b9a498e80fab08b69e7161e : SUCCESS in 1h 04m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8cc27b299a0b48af88e95d643ea02f5c : SUCCESS in 37m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/82bea4786be04c48838d352b67a79066 : SUCCESS in 36m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d503fb350804876ae4cbb083c4c5ceb : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fd09b4629b14bdb977d38c2439af314 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47a39c3fcdaa466ba0ad5bddaac16082 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/db6359d10eea4dbe89367b1c469746e7 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da930849c5e24e84b0a1db7fc91dfdb2 : SUCCESS in 7m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0f6684adb38047bcab59263eb001454e : SUCCESS in 33m 44s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ea26e00472614612acb7ce663fa41769 : SUCCESS in 4m 12s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"88c32d7e395485a0fcad24c800a7e52849851c9e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-11-09 15:45:46.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"28037bb2825a6b3e612d89026b7bdfef27f4ab66","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-11-09 16:41:14.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"a670c72bd6894f8739481745c4d974830d6711d7","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-11-11 06:42:49.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"59e78ae0328cff59a6093b306e87bf67d8b7b541","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-11-11 16:03:43.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"e15d5dc9d65d1a84adfed468d6eea41643a8a854","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-11-11 16:04:51.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"86f4a433b46b6e952f017bd054eacec332cf70a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 17:02:56.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a44d3dda3c5e4b4195adadc44476139e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d6d7c7f217d1457e98bcb8c81f7818cd : SUCCESS in 36m 39s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/3e1d5aa374364152bb659f8137d053ef : SUCCESS in 54m 42s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/94ef39ab72e549a6864f2a2ab58ba917 : SUCCESS in 43m 24s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/563b71c375494058944ca24fc3f648fa : SUCCESS in 41m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1e646ac094b048bba1c0c6a6e60aeac8 : SUCCESS in 42m 09s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/4703b3a92dc04656ac18beff549a4867 : SUCCESS in 43m 21s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/335db9344f714bd4a71e1c6d43703c19 : SUCCESS in 57m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/62a11191dae94b13ba76d0facf39b8c3 : SUCCESS in 33m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/43946981e4544e1087c370c1562012bb : SUCCESS in 36m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33d857be4f7442a2887b746e0cf7e73a : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2981b1866434c50be0a6cb916ee4cf8 : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f32d336d5eb8451c9deb8f2019762593 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84a51f80659a48c1a35c6202c7d1efcd : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73874e210ad34a0785876813de39a75d : SUCCESS in 7m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/570690f31f1547d19f8cec4ac0a2be23 : SUCCESS in 37m 27s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/75339d18da4141728523fa6be83250e4 : SUCCESS in 3m 21s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"4a8cef82a1e710cd618df2e11e00f5f269087b3f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-11-11 18:13:54.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"6a2c9b35af59ff34a021dfa7f74061f805b78807","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-11-14 15:34:05.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"2390e9123caf0583f8a478a7f051b3e4d3a19e6b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-11-14 16:31:33.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"9eafd6f338061ad68c0efe1847e2f83ce5775e3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 16:32:15.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"31bf035589278a70a036f4d9269a3131108bc33f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 17:24:27.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/840a2e4f7de94468b8181ecdc1b3d751\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/20c937852034487b83e72179b61b5362 : SUCCESS in 33m 29s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/ea91ba2a85c647c39d09c6aeaa31e39d : SUCCESS in 38m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6d5d44cd2a6e4b53b327f1315419bae8 : SUCCESS in 32m 40s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/193a0d6be52e4cd7ae79d80c7e1732b8 : SUCCESS in 51m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/12d3d8be59974b8f9c706a5e754b62a1 : SUCCESS in 34m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e9e14c874da149fbb24e28355786851b : SUCCESS in 35m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9db30fd286ef4a47b5a3e69656cef630 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cda023a143994ff3b9007f2292405b7d : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/799720794817404b8d95472cde755829 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf18c9c4fad147938185a3ddd209bd05 : SUCCESS in 6m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/de500f551911455bafbd55225210b1ca : SUCCESS in 26m 17s","accounts_in_message":[],"_revision_number":10},{"id":"5df7b28833d3a1ef76d2e4403d296259bb598d6e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 17:24:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"3b6339a9ae73b02cbc37888266067aef1c07d035","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 17:25:28.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1399289a5c2b4b1fa71f4bfed72685ec\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/401452d52a5341dbafa9be1392a4e1c0 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"d0e1bb18721352007c1c0356658ee73d0bed8c06","revisions":{"917cd487c67af3214f4c9d69a6f25d2f52a7e2f5":{"kind":"REWORK","_number":1,"created":"2020-07-30 16:35:39.000000000","uploader":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"ref":"refs/changes/13/744013/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/13/744013/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/13/744013/1"}}},"commit":{"parents":[{"commit":"45cb4376a24d244fa677f3736299b4effa092657","subject":"Merge \"Remove translation sections from setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/45cb4376a24d244fa677f3736299b4effa092657"}]}],"author":{"name":"Scott Solkhon","email":"scottsolkhon@gmail.com","date":"2020-07-30 16:28:18.000000000","tz":60},"committer":{"name":"Scott Solkhon","email":"scottsolkhon@gmail.com","date":"2020-07-30 16:35:16.000000000","tz":60},"subject":"Reload Infoblox config after deleting zone","message":"Reload Infoblox config after deleting zone\n\nIf you delete a zone using the Infoblox backend\nthe zone will be deleted from Infoblox but the\nzone within OpenStack will go into an ERROR state\nunless an operator manually logs into Infoblox\nand issues a reload of the running configuration.\n\nThis commit fixes this bug by issuing a \u0027restart\nif needed\u0027 after deleting the zone. This does not\nhappen when creating zones as \u0027restart if needed\u0027\ncan be passed through as a param into that API call.\n\nChange-Id: I296c5f085cce27033461be81ca58c85f095df89a\nCloses-Bug: #1840253\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/917cd487c67af3214f4c9d69a6f25d2f52a7e2f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/917cd487c67af3214f4c9d69a6f25d2f52a7e2f5"}]},"branch":"refs/heads/master"},"4d53ec3bb5145949a33a855d06d8a8ee272212b1":{"kind":"REWORK","_number":2,"created":"2020-07-30 19:20:50.000000000","uploader":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"ref":"refs/changes/13/744013/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/13/744013/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/13/744013/2"}}},"commit":{"parents":[{"commit":"45cb4376a24d244fa677f3736299b4effa092657","subject":"Merge \"Remove translation sections from setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/45cb4376a24d244fa677f3736299b4effa092657"}]}],"author":{"name":"Scott Solkhon","email":"scottsolkhon@gmail.com","date":"2020-07-30 16:28:18.000000000","tz":60},"committer":{"name":"Scott Solkhon","email":"scottsolkhon@gmail.com","date":"2020-07-30 19:20:30.000000000","tz":60},"subject":"Reload Infoblox config after deleting zone","message":"Reload Infoblox config after deleting zone\n\nIf you delete a zone using the Infoblox backend\nthe zone will be deleted from Infoblox but the\nzone within OpenStack will go into an ERROR state\nunless an operator manually logs into Infoblox\nand issues a reload of the running configuration.\n\nThis commit fixes this bug by issuing a \u0027restart\nif needed\u0027 after deleting the zone. This does not\nhappen when creating zones as \u0027restart if needed\u0027\ncan be passed through as a param into that API call.\n\nChange-Id: I296c5f085cce27033461be81ca58c85f095df89a\nCloses-Bug: #1840253\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4d53ec3bb5145949a33a855d06d8a8ee272212b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4d53ec3bb5145949a33a855d06d8a8ee272212b1"}]},"branch":"refs/heads/master"},"ac8613098a399c35f383eacb62d9dc561451dd06":{"kind":"REWORK","_number":3,"created":"2020-07-30 22:53:03.000000000","uploader":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"ref":"refs/changes/13/744013/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/13/744013/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/13/744013/3"}}},"commit":{"parents":[{"commit":"45cb4376a24d244fa677f3736299b4effa092657","subject":"Merge \"Remove translation sections from setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/45cb4376a24d244fa677f3736299b4effa092657"}]}],"author":{"name":"Scott Solkhon","email":"scottsolkhon@gmail.com","date":"2020-07-30 16:28:18.000000000","tz":60},"committer":{"name":"Scott Solkhon","email":"scottsolkhon@gmail.com","date":"2020-07-30 22:52:49.000000000","tz":60},"subject":"Reload Infoblox config after deleting zone","message":"Reload Infoblox config after deleting zone\n\nIf you delete a zone using the Infoblox backend\nthe zone will be deleted from Infoblox but the\nzone within OpenStack will go into an ERROR state\nunless an operator manually logs into Infoblox\nand issues a reload of the running configuration.\n\nThis commit fixes this bug by issuing a \u0027restart\nif needed\u0027 after deleting the zone. This does not\nhappen when creating zones as \u0027restart if needed\u0027\ncan be passed through as a param into that API call.\n\nChange-Id: I296c5f085cce27033461be81ca58c85f095df89a\nCloses-Bug: #1840253\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ac8613098a399c35f383eacb62d9dc561451dd06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ac8613098a399c35f383eacb62d9dc561451dd06"}]},"branch":"refs/heads/master"},"9f4a36cc257515073a5e5a1186469c67353983bd":{"kind":"REWORK","_number":4,"created":"2020-07-31 08:08:38.000000000","uploader":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"ref":"refs/changes/13/744013/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/13/744013/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/13/744013/4"}}},"commit":{"parents":[{"commit":"45cb4376a24d244fa677f3736299b4effa092657","subject":"Merge \"Remove translation sections from setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/45cb4376a24d244fa677f3736299b4effa092657"}]}],"author":{"name":"Scott Solkhon","email":"scottsolkhon@gmail.com","date":"2020-07-30 16:28:18.000000000","tz":60},"committer":{"name":"Scott Solkhon","email":"scottsolkhon@gmail.com","date":"2020-07-31 08:08:25.000000000","tz":60},"subject":"Reload Infoblox config after deleting zone","message":"Reload Infoblox config after deleting zone\n\nIf you delete a zone using the Infoblox backend\nthe zone will be deleted from Infoblox but the\nzone within OpenStack will go into an ERROR state\nunless an operator manually logs into Infoblox\nand issues a reload of the running configuration.\n\nThis commit fixes this bug by issuing a \u0027restart\nif needed\u0027 after deleting the zone. This does not\nhappen when creating zones as \u0027restart if needed\u0027\ncan be passed through as a param into that API call.\n\nChange-Id: I296c5f085cce27033461be81ca58c85f095df89a\nCloses-Bug: #1840253\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9f4a36cc257515073a5e5a1186469c67353983bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9f4a36cc257515073a5e5a1186469c67353983bd"}]},"branch":"refs/heads/master"},"0bf50b00a6236ecd6c70c16d103a3f432ad86507":{"kind":"REWORK","_number":5,"created":"2020-07-31 13:57:09.000000000","uploader":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"ref":"refs/changes/13/744013/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/13/744013/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/13/744013/5"}}},"commit":{"parents":[{"commit":"45cb4376a24d244fa677f3736299b4effa092657","subject":"Merge \"Remove translation sections from setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/45cb4376a24d244fa677f3736299b4effa092657"}]}],"author":{"name":"Scott Solkhon","email":"scottsolkhon@gmail.com","date":"2020-07-30 16:28:18.000000000","tz":60},"committer":{"name":"Scott Solkhon","email":"scottsolkhon@gmail.com","date":"2020-07-31 13:53:33.000000000","tz":60},"subject":"Reload Infoblox config after deleting zone","message":"Reload Infoblox config after deleting zone\n\nIf you delete a zone using the Infoblox backend\nthe zone will be deleted from Infoblox but the\nzone within OpenStack will go into an ERROR state\nunless an operator manually logs into Infoblox\nand issues a reload of the running configuration.\n\nThis commit fixes this bug by issuing a \u0027restart\nif needed\u0027 after deleting the zone. This does not\nhappen when creating zones as \u0027restart if needed\u0027\ncan be passed through as a param into that API call.\n\nChange-Id: I296c5f085cce27033461be81ca58c85f095df89a\nCloses-Bug: #1840253\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0bf50b00a6236ecd6c70c16d103a3f432ad86507"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0bf50b00a6236ecd6c70c16d103a3f432ad86507"}]},"branch":"refs/heads/master"},"111b83f410fdd97c2b0dacb7727a186c10aed485":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-09-21 14:36:30.000000000","uploader":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"ref":"refs/changes/13/744013/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/13/744013/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/13/744013/6"}}},"commit":{"parents":[{"commit":"e3ac092a303e1f26a31797a6886a97d857f03bae","subject":"Merge \"Fix mdns test_handler setup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e3ac092a303e1f26a31797a6886a97d857f03bae"}]}],"author":{"name":"Scott Solkhon","email":"scottsolkhon@gmail.com","date":"2020-07-30 16:28:18.000000000","tz":60},"committer":{"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","date":"2022-09-21 14:36:30.000000000","tz":0},"subject":"Reload Infoblox config after deleting zone","message":"Reload Infoblox config after deleting zone\n\nIf you delete a zone using the Infoblox backend\nthe zone will be deleted from Infoblox but the\nzone within OpenStack will go into an ERROR state\nunless an operator manually logs into Infoblox\nand issues a reload of the running configuration.\n\nThis commit fixes this bug by issuing a \u0027restart\nif needed\u0027 after deleting the zone. This does not\nhappen when creating zones as \u0027restart if needed\u0027\ncan be passed through as a param into that API call.\n\nChange-Id: I296c5f085cce27033461be81ca58c85f095df89a\nCloses-Bug: #1840253\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/111b83f410fdd97c2b0dacb7727a186c10aed485"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/111b83f410fdd97c2b0dacb7727a186c10aed485"}]},"branch":"refs/heads/master","description":"Rebase"},"1431035fd99805a487310ed08ca42000eb7880cf":{"kind":"REWORK","_number":7,"created":"2022-09-21 20:35:53.000000000","uploader":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"ref":"refs/changes/13/744013/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/13/744013/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/13/744013/7"}}},"commit":{"parents":[{"commit":"e3ac092a303e1f26a31797a6886a97d857f03bae","subject":"Merge \"Fix mdns test_handler setup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e3ac092a303e1f26a31797a6886a97d857f03bae"}]}],"author":{"name":"Scott Solkhon","email":"scottsolkhon@gmail.com","date":"2020-07-30 16:28:18.000000000","tz":60},"committer":{"name":"Scott Solkhon","email":"scottsolkhon@gmail.com","date":"2022-09-21 20:35:51.000000000","tz":60},"subject":"Reload Infoblox config after deleting zone","message":"Reload Infoblox config after deleting zone\n\nIf you delete a zone using the Infoblox backend\nthe zone will be deleted from Infoblox but the\nzone within OpenStack will go into an ERROR state\nunless an operator manually logs into Infoblox\nand issues a reload of the running configuration.\n\nThis commit fixes this bug by issuing a \u0027restart\nif needed\u0027 after deleting the zone. This does not\nhappen when creating zones as \u0027restart if needed\u0027\ncan be passed through as a param into that API call.\n\nChange-Id: I296c5f085cce27033461be81ca58c85f095df89a\nCloses-Bug: #1840253\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1431035fd99805a487310ed08ca42000eb7880cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1431035fd99805a487310ed08ca42000eb7880cf"}]},"branch":"refs/heads/master"},"88b3cd6a1fdc4b087bee21f45892321826fbe3e6":{"kind":"REWORK","_number":8,"created":"2022-09-22 16:34:59.000000000","uploader":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"ref":"refs/changes/13/744013/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/13/744013/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/13/744013/8"}}},"commit":{"parents":[{"commit":"e3ac092a303e1f26a31797a6886a97d857f03bae","subject":"Merge \"Fix mdns test_handler setup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e3ac092a303e1f26a31797a6886a97d857f03bae"}]}],"author":{"name":"Scott Solkhon","email":"scottsolkhon@gmail.com","date":"2020-07-30 16:28:18.000000000","tz":60},"committer":{"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","date":"2022-09-22 16:34:59.000000000","tz":0},"subject":"Reload Infoblox config after deleting zone","message":"Reload Infoblox config after deleting zone\n\nIf you delete a zone using the Infoblox backend\nthe zone will be deleted from Infoblox but the\nzone within OpenStack will go into an ERROR state\nunless an operator manually logs into Infoblox\nand issues a reload of the running configuration.\n\nThis commit fixes this bug by issuing a \u0027restart\nif needed\u0027 after deleting the zone. This does not\nhappen when creating zones as \u0027restart if needed\u0027\ncan be passed through as a param into that API call.\n\nChange-Id: I296c5f085cce27033461be81ca58c85f095df89a\nCloses-Bug: #1840253\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/88b3cd6a1fdc4b087bee21f45892321826fbe3e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/88b3cd6a1fdc4b087bee21f45892321826fbe3e6"}]},"branch":"refs/heads/master"},"70bc4230115768e2e4b0745c9bd514bd501904af":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2022-11-08 18:35:19.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/13/744013/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/13/744013/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/13/744013/9"}}},"commit":{"parents":[{"commit":"f4999fdfcb1211bf343d04ccd28cfc0e1bc87129","subject":"Merge \"Migrate Designate jobs to run on Ubuntu 22.04 (Jammy)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f4999fdfcb1211bf343d04ccd28cfc0e1bc87129"}]}],"author":{"name":"Scott Solkhon","email":"scottsolkhon@gmail.com","date":"2020-07-30 16:28:18.000000000","tz":60},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2022-11-08 18:35:19.000000000","tz":0},"subject":"Reload Infoblox config after deleting zone","message":"Reload Infoblox config after deleting zone\n\nIf you delete a zone using the Infoblox backend\nthe zone will be deleted from Infoblox but the\nzone within OpenStack will go into an ERROR state\nunless an operator manually logs into Infoblox\nand issues a reload of the running configuration.\n\nThis commit fixes this bug by issuing a \u0027restart\nif needed\u0027 after deleting the zone. This does not\nhappen when creating zones as \u0027restart if needed\u0027\ncan be passed through as a param into that API call.\n\nChange-Id: I296c5f085cce27033461be81ca58c85f095df89a\nCloses-Bug: #1840253\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/70bc4230115768e2e4b0745c9bd514bd501904af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/70bc4230115768e2e4b0745c9bd514bd501904af"}]},"branch":"refs/heads/master","description":"Rebase"},"d0e1bb18721352007c1c0356658ee73d0bed8c06":{"kind":"REWORK","_number":10,"created":"2022-11-11 16:03:43.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/13/744013/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/13/744013/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/13/744013/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/13/744013/10"}}},"commit":{"parents":[{"commit":"f4999fdfcb1211bf343d04ccd28cfc0e1bc87129","subject":"Merge \"Migrate Designate jobs to run on Ubuntu 22.04 (Jammy)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f4999fdfcb1211bf343d04ccd28cfc0e1bc87129"}]}],"author":{"name":"Scott Solkhon","email":"scottsolkhon@gmail.com","date":"2020-07-30 16:28:18.000000000","tz":60},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2022-11-11 16:03:20.000000000","tz":-480},"subject":"Reload Infoblox config after deleting zone","message":"Reload Infoblox config after deleting zone\n\nIf you delete a zone using the Infoblox backend\nthe zone will be deleted from Infoblox but the\nzone within OpenStack will go into an ERROR state\nunless an operator manually logs into Infoblox\nand issues a reload of the running configuration.\n\nThis commit fixes this bug by issuing a \u0027restart\nif needed\u0027 after deleting the zone. This does not\nhappen when creating zones as \u0027restart if needed\u0027\ncan be passed through as a param into that API call.\n\nCo-Authored-By: Jay Faulkner \u003cjay@jvf.cc\u003e\nChange-Id: I296c5f085cce27033461be81ca58c85f095df89a\nCloses-Bug: #1840253\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d0e1bb18721352007c1c0356658ee73d0bed8c06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d0e1bb18721352007c1c0356658ee73d0bed8c06"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
