)]}'
{"id":"openstack%2Fdesignate~747260","triplet_id":"openstack%2Fdesignate~master~I306599c0321ea40334750db57f39d9e3a237f13e","project":"openstack/designate","branch":"master","topic":"native-zuulv3-migration","hashtags":[],"change_id":"I306599c0321ea40334750db57f39d9e3a237f13e","subject":"Native Zuul v3 designate-grenade-pdns4 job","status":"MERGED","created":"2020-08-20 17:43:35.000000000","updated":"2020-09-18 19:31:45.000000000","submitted":"2020-09-18 19:27:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747260-1600457243655-cb39844b","meta_rev_id":"cee14e3d7609692a37bb24eab7c1408404a115a0","_number":747260,"virtual_id_number":747260,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"date":"2020-09-18 17:20:28.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-18 19:27:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2020-08-23 16:18:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2020-09-09 06:18:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-08 21:35:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2020-09-09 06:18:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-23 16:18:09.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-09-08 21:35:25.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-09-09 06:18:33.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-09-18 19:27:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"718d19a51c98a6a5a7a21ce8182cce0f969ffb09","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-08-20 17:43:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5af5547c53461ab44868ab5df81c297f114980b7","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-08-20 18:27:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3de7bea6a5002e085c064301593705afb86e21a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 19:33:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/62af711926814316bf6fe5b2230f283a : FAILURE in 30m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7b45461f194f41ce9703b63bc3a56af0 : FAILURE in 34m 55s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/337930744ab4472d85585b410bcc64b6 : FAILURE in 46m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/165bfa1bc6c94891a84aa4bd1c2ea5c5 : SUCCESS in 52m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e918b8b9a2cb4bbe8f47358c3c02e617 : SUCCESS in 54m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1388d26ea3b4226964540171f6c4b52 : SUCCESS in 5m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/844589fbed264b7381c56d2d9e7731db : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2d9098e39b84bcd82380bb46bb779ea : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5090a0393b14f31bafe2ce527c1e4c8 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bfe2f95cd6a405dbe5ca3df042b36bf : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4a77f5ce77140f18f09f737d760905b : SUCCESS in 11m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/de9c7d26fe194685b74414b65cfb16cd : FAILURE in 36m 57s","accounts_in_message":[],"_revision_number":2},{"id":"bd457c9850da667df38aca2659dc8906ddbc626a","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-08-20 21:35:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"884746f22e12609129a1008ff2fcf1fd78e924a3","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-08-20 22:15:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"995b287e7b75a18d6d841bb16eb596363f2fe299","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 23:03:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ab588c81fba24738908fc11da70ae98e : SUCCESS in 39m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/57cf4bec011148e98fbfce55f1543d64 : SUCCESS in 39m 58s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2dd95f6940b64e6a960b90a13eb78e4b : FAILURE in 36m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e7c535b44b7e4069a4974a9343f3ee3c : SUCCESS in 35m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fdfab8feeb1148f0b3ca3aaee90b9122 : SUCCESS in 28m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b70d72c7121e4ffb9fc10662e9cdd912 : SUCCESS in 5m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ceb8c271bb7b4ee59794fb2204e458f0 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c59242be3e44f22bcf3d6ade382e15a : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36c0046fcffa459cb25047452d10c920 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52dc1572efed48fcac61319f95204d8b : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16229491cf444a98a8f17b8d4b44ae55 : SUCCESS in 8m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9652482f195547b8bdc7feae8e260a4b : SUCCESS in 33m 00s","accounts_in_message":[],"_revision_number":4},{"id":"1b6f08c8ec31e3939bcbb27d37f081ac177d72db","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-08-22 22:15:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"97abe258792c5d4fbd6d9346cbb2deed1aa43c95","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-08-22 22:52:07.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"92af744a6b147cd06b57e77cc21999c0f455ee94","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-08-22 22:54:31.000000000","message":"Patch Set 6:\n\nI think I fixed the missing port, but the 10.0.0.1 IP can\u0027t be found now.","accounts_in_message":[],"_revision_number":6},{"id":"79ccfb58b6e9acf758844e8f404e2248daf6394f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 23:39:15.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4c5326df9f714cf698ba0bc0dee50902 : SUCCESS in 43m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0d1c7bfc85eb4171813ea31b17901327 : SUCCESS in 40m 19s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c9518a734124447b84840da669483ec0 : FAILURE in 45m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6a4bf33693154bdd88345092af034e92 : SUCCESS in 38m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7d9e200d39414a03b29a8b57bfa69c94 : SUCCESS in 36m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa5ca9fc098b473dbd8db6846daad032 : SUCCESS in 4m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6392d075b1e94b3fa6db5824a95c580d : SUCCESS in 9m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7ff415e9f3143149a2e4726993b1825 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a7822275279490a96de4a514c0260f3 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30037a41bfb5405b99f6b06079f71fa7 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe6449ecbf554eed9ad8c633d5381391 : SUCCESS in 6m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f6bbb85ff1c14e4cad94dca56d9f4ce4 : FAILURE in 38m 04s","accounts_in_message":[],"_revision_number":6},{"id":"41cd3ffdb0e1d8da45667fed0ff6a06f5f0a4c3d","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-08-23 12:46:53.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"161772bf3dd3a612226bde28b330724f1d07cf58","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-08-23 12:48:34.000000000","message":"Patch Set 7:\n\nThanks for the hint, I tend to forget that variable (even after co-designing it...). Let\u0027s see how it works now.","accounts_in_message":[],"_revision_number":7},{"id":"bdc64688122e62f551b4349d9c090316b08f6e3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 13:23:41.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b16451204833447da7f9089d7764edf9 : SUCCESS in 32m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ae0f0fc11bfa4ef799dff895201114f4 : SUCCESS in 31m 17s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a9b932c3f24e4e418d3ba0b60e27e562 : FAILURE in 34m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/71cbc428005f4bfe8e6bf3ad0064f9c7 : SUCCESS in 33m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7c3aa0486ddf440ba998b5289f56e273 : SUCCESS in 32m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/997931265ba64c1e93376c61c873ad19 : SUCCESS in 5m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57222a54ffbb4147b57e5aa40b036b7e : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7afecba5ec6b4e8481bb76916c303e99 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73aefe84fdd94c88a458608c1521c6b1 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64d8216acd484362aa1d3433521ff372 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/083feb2442bb48e183d0dac79f9047b9 : SUCCESS in 7m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b86813632d3f4a55a15225f8cda5d85b : SUCCESS in 29m 59s","accounts_in_message":[],"_revision_number":7},{"id":"df632c240e0d52b644ad14227a7f65af6ce11c0c","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-08-23 13:48:46.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"3b1de5dcec304102465532ace970de42b3f7b795","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 14:32:35.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9b0ac05f175b490e8e36f744055776ca : SUCCESS in 36m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b0bdd42b7bb946d9962ca45c83c34f29 : SUCCESS in 34m 13s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b61a706fefb44eb0957a5c49c221395e : FAILURE in 42m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fbded85d84fa4cc88412e0bb9eb5a1b2 : SUCCESS in 35m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e46ed2ddd53841a5a33221f2353a62fd : SUCCESS in 36m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc86f8d5cb054cb2ab267fa9f8f3992d : SUCCESS in 5m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83db5d721dab478693c72c82f3718f7b : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e6fa76a3cd44ca59e517045af3ab344 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b81f28d75604a8ba0ebd363aa68326e : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/756263f272da40be91b61b458c763d34 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bc1dddeb1da46109fb3a6ef3a299532 : SUCCESS in 7m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/31ea0bbca5be4f23bc9da6fdbbe0494b : SUCCESS in 31m 19s","accounts_in_message":[],"_revision_number":8},{"id":"e51d67a575ce436b99095eadb796e754a8be35d7","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-08-23 14:54:55.000000000","message":"Uploaded patch set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"1b164f5d5de33a302782c4a8d851e01c13841fe2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 16:03:32.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5c42262dafaa4dc7b735a908941f67e3 : SUCCESS in 38m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f79af1c6d8ef4d6c85c9841cd053f1f4 : SUCCESS in 35m 52s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/63c215d49bf046f3b0fdb881f880d199 : SUCCESS in 1h 01m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0641fee68a63447abd521330cce68c65 : SUCCESS in 42m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e820861c708e44bba784cdaa07e5ea9d : SUCCESS in 41m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f47e9689bb7a448998138fbb01fb2ddf : SUCCESS in 4m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/91bbaa50262a438fa103c8dc163f890c : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f14aa3d5ecf4f479aceb6be2eeddaa0 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4da643c8e7240969a48c72bf7af2dc3 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c9fc855838944cfbbe8186df84dd621 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4284796bc384f6fa70aa67c635f0196 : SUCCESS in 7m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4e8f7c02727d4f48963c4452b9ab6b94 : SUCCESS in 34m 04s","accounts_in_message":[],"_revision_number":9},{"id":"4d59d17fac8636d87ed3f651f9b29b6fa6496998","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-08-23 16:18:09.000000000","message":"Patch Set 9: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":9},{"id":"81ca1203d8d9d5d3b13c712ced6da5da746b5468","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-08-23 16:30:38.000000000","message":"Patch Set 9:\n\nNow we need grenade cores to vote on https://review.opendev.org/#/c/747260/","accounts_in_message":[],"_revision_number":9},{"id":"26eff93fdac25c6aabf12490fb3b1fc3e6048b5f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-08-23 19:18:34.000000000","message":"Patch Set 9: Code-Review-1\n\nSomething seems wrong, the designate-api log doesn\u0027t contain any requests:\nhttps://13bf088df83684f885f6-9cb29527c5d80ca3ff28f256812f71ef.ssl.cf5.rackcdn.com/747260/9/check/designate-grenade-pdns4/63c215d/controller/logs/screen-designate-api.txt","accounts_in_message":[],"_revision_number":9},{"id":"953e7ca0e1e86a17384075ac09384fb393fbc7dd","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-09-08 15:28:55.000000000","message":"Patch Set 9:\n\nNicolas, can you please help with the question, whether it\u0027s expected? I don\u0027t know designate enough","accounts_in_message":[],"_revision_number":9},{"id":"56cb5dc34b520cf6012957181430f2fe6cdbdaf0","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-09-08 21:35:20.000000000","message":"Patch Set 9:\n\n\u003e Something seems wrong, the designate-api log doesn\u0027t contain any\n \u003e requests:\n \u003e https://13bf088df83684f885f6-9cb29527c5d80ca3ff28f256812f71ef.ssl.cf5.rackcdn.com/747260/9/check/designate-grenade-pdns4/63c215d/controller/logs/screen-designate-api.txt\n\nThis could be an old issue. We made the switch to nginx, and the logs look to still be in Apache. Maybe both are running?\nhttps://zuul.opendev.org/t/openstack/build/63c215d49bf046f3b0fdb881f880d199/log/controller/logs/apache/designate-api_log.txt","accounts_in_message":[],"_revision_number":9},{"id":"ae285bdfb5e2613bcd64907a4850961f54291485","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-09-08 21:35:25.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"9e5baa394e21298612cbd24503013c8d76a7e693","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-09-09 06:18:33.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\nOk, so we do have some logs, can clean up later.","accounts_in_message":[],"_revision_number":9},{"id":"f2730a2b48109b1494baa39d0fca679d74a5541f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 06:18:49.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"d0b26e44fb67ce4144147e118bf255d2bfe72ec2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 07:17:03.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c798995b428e4fd7a7426977997f611e : SUCCESS in 27m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/552877606f9d49f4b7ad7fd433e4d760 : SUCCESS in 27m 59s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3a672afff9cd4c33875fe644649bbd53 : SUCCESS in 54m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/952296383c404dc294aec40435afb757 : SUCCESS in 41m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/84e333b150f6402e93624d49cf9100c6 : SUCCESS in 38m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f84d481215b548cfbd7e70a987430778 : FAILURE in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c505370560254996a311983c5526cb3b : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/125996af50c9461396a189a9e89aeda9 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7baa414d3ea34958bde1cc419cd02ddf : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04d5428d1cbe42c9b204ee11046ac752 : SUCCESS in 9m 27s","accounts_in_message":[],"_revision_number":9},{"id":"f78f88b627907055c95284d683e5899b33fda1e8","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-09-09 08:30:38.000000000","message":"Patch Set 9:\n\nOuch, lower-constraints now complain with \n\nModuleNotFoundError: No module named \u0027vine.five\u0027","accounts_in_message":[],"_revision_number":9},{"id":"21e3d889c83c5d9a8626e3101c2c67c71d235337","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-09-18 11:05:47.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"56d038d80cbb7e2a7dbdbfe08c8b22ae5385109c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 12:13:02.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/067fe1e10e5e490b9385cca3e5f519b0 : SUCCESS in 33m 19s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8c2eaf27c0be4a769b01fa1202471d0f : SUCCESS in 33m 30s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/38277486d28444a8a261c724e5d6b221 : SUCCESS in 51m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bc3af06232994e029ae6ff668af2bf8f : FAILURE in 1h 02m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f138737373ff4a068bb4d59d0e15e410 : SUCCESS in 1h 02m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6eb7796741f04cdea74a00fcc8df4c73 : SUCCESS in 5m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/580a9c9be4cf45ce89d0cf91bbe27d9c : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39e8c2e1b01b40c4a544b829eac2d2ec : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ae439dc86084f83ad5856a11b37b873 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c45d0a56ab14a65918e8f279c7d841c : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d3ddd3fcf0645cfa8a5b0c732922c48 : SUCCESS in 15m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6bef7316e86b4f2984d913b07fcb8abe : SUCCESS in 31m 00s","accounts_in_message":[],"_revision_number":9},{"id":"f75001f8cf7f800bf92f9b919ef80075dc4e975f","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-09-18 12:29:34.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"fc7a57d1b0ef8346e057b4e0aaab04279a3c01a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 13:35:32.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/df19f36459054e3384482c4918404a4b : SUCCESS in 37m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/dcd93a0fd4634ba6a014c81bc8e49647 : SUCCESS in 37m 05s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bda8c09283a74a2a93bb91bacb5b9dcf : SUCCESS in 58m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e528f6e5698b419eb8e7f09c72bd8308 : SUCCESS in 43m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0b5ccedd4f1d489e90827ea4988e1edf : SUCCESS in 44m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf16654fb4fd48f7ae25f3184e776855 : SUCCESS in 6m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1cdc17916f2e480ca09b43253444b6cd : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37fe49184f5e49f692afa9d0fa008468 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38bf5f5e07024c1ca554061c2caaa9c2 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be5a79aae57841b5bd74f74841faae1f : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e148bfbce484a168e9c0681bb81676c : SUCCESS in 15m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3b277b1610bc4b6aaa89b4c32fa98464 : SUCCESS in 30m 38s","accounts_in_message":[],"_revision_number":9},{"id":"a6f3814708459403eb57ca0a918de996ad5325d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 13:35:52.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"cee506d21bffd44feab8db14620bdd207d0388a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 14:34:54.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/aaf163ae48a1427aa3fce6bb045ee250 : FAILURE in 43m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bbe15fc5612a4eb6b6eb1ba0de9d22ba : SUCCESS in 39m 16s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/434e7ef97d824afa905380c2c278210e : SUCCESS in 53m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0898c29247ac44079b0384a1ca3ddd1d : SUCCESS in 40m 29s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1fda4afe45b449239f5582e0055c59dc : SUCCESS in 35m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6855a8b3004b4fdd8759b1c2a511e0b6 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ccaa1d4514e46da8640a5e0efd0228d : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0901ce9e3909476fb58268e1c6ce54c8 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f8c1ca630684c4d83dc60d604a4611c : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90aa4ce4e4cc499fb60deb598ea4bc41 : SUCCESS in 15m 27s","accounts_in_message":[],"_revision_number":9},{"id":"5e0e8df3f83053caa71d85aabacebb79d3a819ff","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-09-18 14:53:28.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"7b8e47c8379dd4412f45090ded4b81d13b7f770c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 16:01:14.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1019bc3ae2614311b8597af1ccc8652d : SUCCESS in 35m 29s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/99e0562ee6d8498f87c4be4c09546f81 : SUCCESS in 30m 46s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/61f91bd910c04f58aee0d418e54fd75e : SUCCESS in 1h 04m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3975401f71424628b4c5fb263ef1de52 : SUCCESS in 40m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/dd3e4be1d68743058be990cc3eaaffa1 : SUCCESS in 41m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a729ef06c920470bbc410cf2c9259879 : SUCCESS in 5m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b24736bf728d457dbba44a89d7bc4e91 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38ba43c88284442fbde50f4d3b7f1617 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32a785b3bb4740c59a89a91d2d06613b : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af7aa5ae9dc8420fb6f67a2bcc063987 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/774a904da5194c7ea464921e2fe2b71f : SUCCESS in 14m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b737a7d4815445bfa81d2723c71e6ce0 : SUCCESS in 30m 26s","accounts_in_message":[],"_revision_number":9},{"id":"6f841cef44a9da34f5dff4da3725842fbd9b2961","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 16:01:34.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"a2d8525a4b65ebc835a8c1e249a96aac047cf98d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 17:07:55.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f0f175ded6094123a4da4ca742451798 : SUCCESS in 43m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9235987dfbab459c9273d25cfc75b234 : SUCCESS in 42m 31s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d9a52f2c56b840aaa6dd1b712482e713 : SUCCESS in 58m 53s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/22d3c8fb6e7c46cf98ff942eab7bdd2b : FAILURE in 40m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/efd01611ef0644c68c05ac1255981578 : SUCCESS in 38m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26f98299ade2453193dc864a7e6ae947 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f9a77ecb2254bb8bf3a4a5b17c86b88 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2578f5538fcc4507964a3da0af1f11cc : SUCCESS in 7m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62a567f808464901b310d46dd09f0c66 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee840d6356aa4ffb9ae8b47c620ada78 : SUCCESS in 14m 39s","accounts_in_message":[],"_revision_number":9},{"id":"d93742d36c6ce8425de2f831b7e7cfd878e6ab3e","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-09-18 17:20:28.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"b88a4d5dfa7e69e6174f51be588775c251c4a11b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 18:20:49.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ae85418eebb24c2c9ea3d41a99c688e2 : SUCCESS in 36m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d660574aaba148aaa26f7a7adec1a219 : SUCCESS in 42m 12s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e8dc2aa1b8df4c9ba95d3298b481e86c : SUCCESS in 54m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/095e90da213f4301818c5aa7f693b802 : SUCCESS in 40m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1bd739515df5470c962485a2ede54c4e : SUCCESS in 37m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/598db49306634952a7cc7620b533ad6a : SUCCESS in 6m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19d7da111c9048279806343b06325c39 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb7ec71a83534132ae8dfc15a2ae2871 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c786257d054945d5b79caba637fcbf72 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d52ab0c8667f4f928e81728185ff6980 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a0588e33c9c40779f41a6dc259acb6d : SUCCESS in 17m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/26b30a4af32f4709a0a3e69f963e7694 : SUCCESS in 34m 17s","accounts_in_message":[],"_revision_number":9},{"id":"fb87fe81e59df6fbee851f8c45664b8f46965a17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 18:21:08.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"cd29ccd9f134e98d1701471ac144bb2a874e42a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 19:27:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"d8d832b2e6da4a54224870bc15d58df987d90917","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 19:27:23.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bae8745e293745508278a1fb06c57940 : SUCCESS in 36m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e7291e9d08544a40acfcd0c4ca35d2ef : SUCCESS in 39m 43s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/46c89901d02a49ae8f03b27f3421ff2b : SUCCESS in 59m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/464686beeeb44d2c90c658cb7b205e4a : SUCCESS in 37m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8b1a2e61863a4c74be560ab08c9975ce : SUCCESS in 35m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/655ca22a9b8046948cebc0c908fd0502 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24d00ac5866846a293bf87067b0838a9 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d9b8e26ff9c4f8aa2165c8218b9042b : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6965740208a54358bdac7cffc10891cd : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb5afaaf9cef48fbb17f1d711fe6a6c6 : SUCCESS in 29m 45s","accounts_in_message":[],"_revision_number":9},{"id":"71b284cef4d00d34ad57f179eac523c5c5c406f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 19:31:45.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f46d99f46dfa42eb9b0103f074acd6a9 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"8cbfafdb8dc7ab9c0003af09cf3d9e0d65bb68ff","revisions":{"53d861d3ea96f2233e1f4a26c73643f4a6a2aa31":{"kind":"REWORK","_number":1,"created":"2020-08-20 17:43:35.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/60/747260/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/747260/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/747260/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/747260/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/747260/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/747260/1"}}},"commit":{"parents":[{"commit":"1fa892ea5a6db061d415a3ebe2b94105aaba57fd","subject":"Fix multi messages AXFR with TSIG","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1fa892ea5a6db061d415a3ebe2b94105aaba57fd"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-08-20 17:41:19.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-08-20 17:41:19.000000000","tz":120},"subject":"Native Zuul v3 designate-grenade-pdns4 job","message":"Native Zuul v3 designate-grenade-pdns4 job\n\nIt matches the feature of the existing legacy job.\n\nChange-Id: I306599c0321ea40334750db57f39d9e3a237f13e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/53d861d3ea96f2233e1f4a26c73643f4a6a2aa31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/53d861d3ea96f2233e1f4a26c73643f4a6a2aa31"}]},"branch":"refs/heads/master"},"396885b2d60f988cdecff36223ad7630aaaffe40":{"kind":"REWORK","_number":2,"created":"2020-08-20 18:27:22.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/60/747260/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/747260/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/747260/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/747260/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/747260/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/747260/2"}}},"commit":{"parents":[{"commit":"1fa892ea5a6db061d415a3ebe2b94105aaba57fd","subject":"Fix multi messages AXFR with TSIG","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1fa892ea5a6db061d415a3ebe2b94105aaba57fd"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-08-20 18:27:13.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-08-20 18:27:14.000000000","tz":120},"subject":"Native Zuul v3 designate-grenade-pdns4 job","message":"Native Zuul v3 designate-grenade-pdns4 job\n\nIt matches the feature of the existing legacy job.\n\nChange-Id: I306599c0321ea40334750db57f39d9e3a237f13e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/396885b2d60f988cdecff36223ad7630aaaffe40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/396885b2d60f988cdecff36223ad7630aaaffe40"}]},"branch":"refs/heads/master"},"54bc56014d3c187c5efc5a4484817f6f022f11cf":{"kind":"REWORK","_number":3,"created":"2020-08-20 21:35:01.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/60/747260/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/747260/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/747260/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/747260/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/747260/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/747260/3"}}},"commit":{"parents":[{"commit":"1fa892ea5a6db061d415a3ebe2b94105aaba57fd","subject":"Fix multi messages AXFR with TSIG","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1fa892ea5a6db061d415a3ebe2b94105aaba57fd"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-08-20 21:34:50.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-08-20 21:34:50.000000000","tz":120},"subject":"Native Zuul v3 designate-grenade-pdns4 job","message":"Native Zuul v3 designate-grenade-pdns4 job\n\nIt matches the feature of the existing legacy job.\n\nChange-Id: I306599c0321ea40334750db57f39d9e3a237f13e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/54bc56014d3c187c5efc5a4484817f6f022f11cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/54bc56014d3c187c5efc5a4484817f6f022f11cf"}]},"branch":"refs/heads/master"},"07ea7eac2414de930ce675aec565ab9ee6104329":{"kind":"REWORK","_number":4,"created":"2020-08-20 22:15:38.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/60/747260/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/747260/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/747260/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/747260/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/747260/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/747260/4"}}},"commit":{"parents":[{"commit":"1fa892ea5a6db061d415a3ebe2b94105aaba57fd","subject":"Fix multi messages AXFR with TSIG","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1fa892ea5a6db061d415a3ebe2b94105aaba57fd"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-08-20 22:15:26.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-08-20 22:15:26.000000000","tz":120},"subject":"Native Zuul v3 designate-grenade-pdns4 job","message":"Native Zuul v3 designate-grenade-pdns4 job\n\nIt matches the feature of the existing legacy job.\n\nChange-Id: I306599c0321ea40334750db57f39d9e3a237f13e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/07ea7eac2414de930ce675aec565ab9ee6104329"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/07ea7eac2414de930ce675aec565ab9ee6104329"}]},"branch":"refs/heads/master"},"1046e6d6dcfa7b21520a80488106acec8e805143":{"kind":"REWORK","_number":5,"created":"2020-08-22 22:15:10.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/60/747260/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/747260/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/747260/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/747260/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/747260/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/747260/5"}}},"commit":{"parents":[{"commit":"1fa892ea5a6db061d415a3ebe2b94105aaba57fd","subject":"Fix multi messages AXFR with TSIG","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1fa892ea5a6db061d415a3ebe2b94105aaba57fd"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-08-20 22:15:26.000000000","tz":120},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-08-22 22:14:55.000000000","tz":-360},"subject":"Native Zuul v3 designate-grenade-pdns4 job","message":"Native Zuul v3 designate-grenade-pdns4 job\n\nIt matches the feature of the existing legacy job.\n\nChange-Id: I306599c0321ea40334750db57f39d9e3a237f13e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1046e6d6dcfa7b21520a80488106acec8e805143"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1046e6d6dcfa7b21520a80488106acec8e805143"}]},"branch":"refs/heads/master"},"83f4721fdd50cba2201e34f5b513cdcaded7f148":{"kind":"REWORK","_number":6,"created":"2020-08-22 22:52:07.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/60/747260/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/747260/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/747260/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/747260/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/747260/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/747260/6"}}},"commit":{"parents":[{"commit":"1fa892ea5a6db061d415a3ebe2b94105aaba57fd","subject":"Fix multi messages AXFR with TSIG","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1fa892ea5a6db061d415a3ebe2b94105aaba57fd"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-08-20 22:15:26.000000000","tz":120},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-08-22 22:51:54.000000000","tz":-360},"subject":"Native Zuul v3 designate-grenade-pdns4 job","message":"Native Zuul v3 designate-grenade-pdns4 job\n\nIt matches the feature of the existing legacy job.\n\nChange-Id: I306599c0321ea40334750db57f39d9e3a237f13e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/83f4721fdd50cba2201e34f5b513cdcaded7f148"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/83f4721fdd50cba2201e34f5b513cdcaded7f148"}]},"branch":"refs/heads/master"},"258243020f61e1809e834debb8049446d3ec1a43":{"kind":"REWORK","_number":7,"created":"2020-08-23 12:46:53.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/60/747260/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/747260/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/747260/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/747260/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/747260/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/747260/7"}}},"commit":{"parents":[{"commit":"1fa892ea5a6db061d415a3ebe2b94105aaba57fd","subject":"Fix multi messages AXFR with TSIG","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1fa892ea5a6db061d415a3ebe2b94105aaba57fd"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-08-23 12:46:41.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-08-23 12:46:41.000000000","tz":120},"subject":"Native Zuul v3 designate-grenade-pdns4 job","message":"Native Zuul v3 designate-grenade-pdns4 job\n\nIt matches the feature of the existing legacy job.\n\nChange-Id: I306599c0321ea40334750db57f39d9e3a237f13e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/258243020f61e1809e834debb8049446d3ec1a43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/258243020f61e1809e834debb8049446d3ec1a43"}]},"branch":"refs/heads/master"},"72d0d948d808c321c9255af18d37f5b61c2b8cb8":{"kind":"REWORK","_number":8,"created":"2020-08-23 13:48:46.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/60/747260/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/747260/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/747260/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/747260/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/747260/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/747260/8"}}},"commit":{"parents":[{"commit":"1fa892ea5a6db061d415a3ebe2b94105aaba57fd","subject":"Fix multi messages AXFR with TSIG","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1fa892ea5a6db061d415a3ebe2b94105aaba57fd"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-08-23 13:48:38.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-08-23 13:48:38.000000000","tz":120},"subject":"Native Zuul v3 designate-grenade-pdns4 job","message":"Native Zuul v3 designate-grenade-pdns4 job\n\nIt matches the feature of the existing legacy job.\n\nChange-Id: I306599c0321ea40334750db57f39d9e3a237f13e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/72d0d948d808c321c9255af18d37f5b61c2b8cb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/72d0d948d808c321c9255af18d37f5b61c2b8cb8"}]},"branch":"refs/heads/master"},"8cbfafdb8dc7ab9c0003af09cf3d9e0d65bb68ff":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2020-08-23 14:54:55.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/60/747260/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/747260/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/747260/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/747260/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/747260/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/747260/9"}}},"commit":{"parents":[{"commit":"1fa892ea5a6db061d415a3ebe2b94105aaba57fd","subject":"Fix multi messages AXFR with TSIG","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1fa892ea5a6db061d415a3ebe2b94105aaba57fd"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-08-23 14:54:43.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-08-23 14:54:43.000000000","tz":120},"subject":"Native Zuul v3 designate-grenade-pdns4 job","message":"Native Zuul v3 designate-grenade-pdns4 job\n\nIt matches the feature of the existing legacy job.\n\nDepends-On: https://review.opendev.org/747579\nChange-Id: I306599c0321ea40334750db57f39d9e3a237f13e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8cbfafdb8dc7ab9c0003af09cf3d9e0d65bb68ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8cbfafdb8dc7ab9c0003af09cf3d9e0d65bb68ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
