)]}'
{"id":"openstack%2Fdesignate~750569","triplet_id":"openstack%2Fdesignate~master~I1634aae2b48da9f51449e35cff7ddcd50e514d4f","project":"openstack/designate","branch":"master","topic":"sample_placement","hashtags":[],"change_id":"I1634aae2b48da9f51449e35cff7ddcd50e514d4f","subject":"Add the placement api to vagrant local.conf when nova is enabled","status":"ABANDONED","created":"2020-09-09 05:58:12.000000000","updated":"2022-09-24 05:52:25.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4d7d73954f64d153e935d1ff20f759ab86a7f5e6","_number":750569,"virtual_id_number":750569,"owner":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-09 15:08:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-09 06:08:11.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-09-09 15:08:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"93895e91ba64be1f79feaf4a490aacdfe77d7090","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2020-09-09 05:58:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ae04566ea2050267660a824b808a4d53a0d9fc9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-09-09 06:08:11.000000000","message":"Patch Set 1:\n\n(3 comments)\n\nI\u0027m surprised that this is still working at all, might be a good opportunity to do some further cleanup. Maybe also crosscheck against the config we have in zuul.yaml?","accounts_in_message":[],"_revision_number":1},{"id":"74c89b3d65609ab91bbe147f41e1c86bfc206667","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 06:55:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c12b5c6e90d64376b4425a1cb885966f : SUCCESS in 34m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/907a313ffbc247b48bd318505e398c4b : SUCCESS in 54m 26s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bc7dbe9c7225435a889c9830e441b18d : SUCCESS in 53m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bcf45ccf9cf94ddea215f1f66ad35c40 : SUCCESS in 53m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5f932427de0145bd96bc6528a8d359cd : SUCCESS in 35m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e5a494e89fa46d79b2cd6b5892b1f30 : SUCCESS in 5m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7abeb3e4326248cd9b589c8f26ffd4ef : FAILURE in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b62c254f7a224fc1925a3dad15d672f8 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2098b91983fd46f4a551e8719bacb4a1 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1c684fbc21341c7ae10bc5a655fa515 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e57d63e6c0cb433b8f246310eb2ef46b : SUCCESS in 7m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c1bddf48ba534801ad4e7a22facf0bff : SUCCESS in 25m 44s","accounts_in_message":[],"_revision_number":1},{"id":"a9f28c3365d64115f89be09d94cc69bed4f273fd","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2020-09-09 14:10:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"62eb95c81c8691e6d9e21dcd4be0dcfb27a1ace6","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2020-09-09 14:14:21.000000000","message":"Patch Set 1:\n\n(2 comments)\n\n\u003e I\u0027m surprised that this is still working at all, might be a good opportunity to do some further cleanup. Maybe also crosscheck against the config we have in zuul.yaml?\n\nI\u0027m just working through it enabling features as I go, I\u0027ll check the current gate jobs that\u0027s a good idea. The vagrant instructions themselves are a little dubious because vagrant-mutate seems to be unmaintained (and doesn\u0027t install for me) but I figure just having the local.conf file there as a clear starting point for designate work is useful on its own.","accounts_in_message":[],"_revision_number":1},{"id":"ad7c32cc36f4b35c05b2b233bcb9df29a4eac33e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 15:08:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e36909d2da4041c9aff5f56e2c622078 : SUCCESS in 31m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/78ad95cd4ee24c2585b1f462ebac2864 : SUCCESS in 37m 43s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0bdbf9b5b014479d955e83d3a0818c5b : SUCCESS in 47m 49s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/777c2f528bff47aea0e2119121fb47e4 : SUCCESS in 35m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9e3f626325ff404bbbc368d8d7106ae3 : SUCCESS in 34m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1d8949ff1b24858ab132725aac7ad78 : SUCCESS in 6m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/816d21cb20ad4e1689cd12ac7caaf64e : FAILURE in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70514ab3170c4a7580cef4eaaa157edb : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/022170285c9b4ea295677fba2aa3c6ad : FAILURE in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7317662908d4106acc1c79f4bfc73e8 : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a6ff00240f1486da2ef316ff3858c91 : SUCCESS in 7m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fd7329adc75946f2a0a84f7b3f4c16e0 : SUCCESS in 35m 51s","accounts_in_message":[],"_revision_number":2},{"id":"4d7d73954f64d153e935d1ff20f759ab86a7f5e6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-09-24 05:52:25.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"92fc8aa6031a67d402cef865d39664e96354f80d","revisions":{"554a29f705d607c34a0f86d893651ba70ffb242b":{"kind":"REWORK","_number":1,"created":"2020-09-09 05:58:12.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/69/750569/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/69/750569/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/750569/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/750569/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/750569/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/69/750569/1"}}},"commit":{"parents":[{"commit":"06f147f76ddf5fa198fb7851ef14144bfb0fb9ee","subject":"Merge \"Remove Babel from lower-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06f147f76ddf5fa198fb7851ef14144bfb0fb9ee"}]}],"author":{"name":"michaeltchapman","email":"woppin@gmail.com","date":"2020-09-09 05:57:14.000000000","tz":600},"committer":{"name":"michaeltchapman","email":"woppin@gmail.com","date":"2020-09-09 05:57:14.000000000","tz":600},"subject":"Add the placement api to devstack local.conf when nova is enabled","message":"Add the placement api to devstack local.conf when nova is enabled\n\nWithout the placement api nova-conductor will fail to start and\ndevstack will fail.\n\nChange-Id: I1634aae2b48da9f51449e35cff7ddcd50e514d4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/554a29f705d607c34a0f86d893651ba70ffb242b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/554a29f705d607c34a0f86d893651ba70ffb242b"}]},"branch":"refs/heads/master"},"92fc8aa6031a67d402cef865d39664e96354f80d":{"kind":"REWORK","_number":2,"created":"2020-09-09 14:10:50.000000000","uploader":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"ref":"refs/changes/69/750569/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/69/750569/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/750569/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/750569/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/750569/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/69/750569/2"}}},"commit":{"parents":[{"commit":"06f147f76ddf5fa198fb7851ef14144bfb0fb9ee","subject":"Merge \"Remove Babel from lower-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06f147f76ddf5fa198fb7851ef14144bfb0fb9ee"}]}],"author":{"name":"michaeltchapman","email":"woppin@gmail.com","date":"2020-09-09 05:57:14.000000000","tz":600},"committer":{"name":"michaeltchapman","email":"woppin@gmail.com","date":"2020-09-09 14:10:32.000000000","tz":600},"subject":"Add the placement api to vagrant local.conf when nova is enabled","message":"Add the placement api to vagrant local.conf when nova is enabled\n\nWithout the placement api nova-conductor will fail to start and\ndevstack will fail.\n\nChange-Id: I1634aae2b48da9f51449e35cff7ddcd50e514d4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/92fc8aa6031a67d402cef865d39664e96354f80d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/92fc8aa6031a67d402cef865d39664e96354f80d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
