)]}'
{"id":"openstack%2Fdesignate~751245","triplet_id":"openstack%2Fdesignate~master~I500c932a502db24332aad95c54713b7056bc1df4","project":"openstack/designate","branch":"master","hashtags":[],"change_id":"I500c932a502db24332aad95c54713b7056bc1df4","subject":"fix database sync always get the wrong  value","status":"MERGED","created":"2020-09-11 11:07:55.000000000","updated":"2020-09-29 20:54:53.000000000","submitted":"2020-09-21 19:37:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751245-1600717059993-309c905f","meta_rev_id":"cc25dfad368ea87c39cb0f4fca6e33e584bec797","_number":751245,"virtual_id_number":751245,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-21 19:37:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-21 13:17:46.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2020-09-21 18:35:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":2,"date":"2020-09-21 17:45:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":1,"date":"2020-09-21 18:35:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-21 17:45:06.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-09-21 18:35:13.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-09-21 19:37:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2e4d7b02ba70e251fd0c270662d13d7b96dd19b7","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-09-11 11:07:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7df79fe948a6bb3cd050102af7bee81edd0f84c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 12:13:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dd6d64bd993e4eb5b860a88016a5861c : FAILURE in 41m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/03231571029b43739f846b8f5613f3cc : SUCCESS in 42m 04s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bf94edf501444101b86ec67986af4bcc : SUCCESS in 57m 28s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/db5626e00af34e0ab6dc8f63af569241 : SUCCESS in 30m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d8837c521b0543f3aab399d19aef5a8e : SUCCESS in 45m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05a9c0e2309644d6a336fcc82f970858 : SUCCESS in 5m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d400cfcaf27c4fc9a84d5d14e1a9939e : FAILURE in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57eb0798fc8c4da49b946de2e0db45e7 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34f91ec989744b8084df277ceb8dc986 : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/218c079e13734e6892db20fc584e6adb : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bfcfa6664494ee7beb803c7e346a420 : SUCCESS in 23m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0102344da2e84174a654a457796d3e72 : SUCCESS in 33m 05s","accounts_in_message":[],"_revision_number":1},{"id":"0bb252aa8eb68f006b8ba8a43428954ce5ec882f","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-09-14 03:11:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cc0e83ca57b5ecc68101546213cdcb2bac51a99e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 04:08:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/09799d2d0af04893b6095a9eca28ed75 : SUCCESS in 35m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8bee28c3046149c7b499d94f86943896 : FAILURE in 33m 38s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c76c3b20e20344728b9485c505760db3 : SUCCESS in 45m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/06da49a734ec416084dc7b2ba0f439aa : SUCCESS in 26m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/856a0b96b70f4e81a094cf148c0dcf18 : SUCCESS in 36m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ef360f2dc0049e4b22e4fb0af8ab242 : SUCCESS in 5m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3ef76fe1d17460c8a44d7dc8bab59f9 : FAILURE in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f114e6e6fce446cbfbe4ac18636d8bc : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b53ef7680c6e48a7a09a703fa44f69b1 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a08b1d30dce346069bd16d04ce0763ae : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/193f61aed0f1447894dc5e1b16e0f160 : SUCCESS in 14m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f467e7528db6463cb3364175ee31b014 : SUCCESS in 31m 03s","accounts_in_message":[],"_revision_number":1},{"id":"eeb93443e140fa8675f6df6d6e4e1852f81f5a36","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-09-20 11:15:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"14e183a71d2b2efadcf58737cd84fce63029e856","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-20 12:14:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fda53c61d79440c4bc0356287c7c5c5a : SUCCESS in 39m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6c8868eb81b041a382f93023201494c5 : SUCCESS in 35m 16s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/df1d3b15a7b2414798eb2b7a43cf194e : SUCCESS in 54m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a022bbf1112d4db5b104ecbe3947b9df : SUCCESS in 38m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ddbd27002a0d43ab8715060dc9a81bea : SUCCESS in 34m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10bde99fafee49c18c51e07ab6d5dd58 : SUCCESS in 4m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c758e246d8954298b68a5391630e891a : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f68b6bc9c214e7ab94f7ae85c80b545 : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f33c9c9c74b5422d97cb0560265127ef : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ccc68dac2644890b2225504662016b9 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fe33ec1d7cf4380a43406e1d254e0fd : SUCCESS in 13m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e6b36297dc2e4588ab594dab547d1ea4 : SUCCESS in 32m 53s","accounts_in_message":[],"_revision_number":1},{"id":"7b56bc3c9c3bf56cdeb45aca37812ea80afe79da","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-09-20 18:49:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2db9e0f8d301ded5105c76ec117d0b039fb55add","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-09-21 13:17:46.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nAnother similar issue is at line https://github.com/openstack/designate/blob/master/designate/manage/pool.py#L90, default_pool_id doesn\u0027t get the value from config file.","accounts_in_message":[],"_revision_number":1},{"id":"6ba03762dc3e1ff79f020c78c3473ec1ab00f1cf","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-09-21 17:45:06.000000000","message":"Patch Set 1: Code-Review+2\n\nInteresting. I will try to figure out why you would need that, but fine as it is now.","accounts_in_message":[],"_revision_number":1},{"id":"7a4b4b8ab4b97fdfbc05908d38f2dec3d4651991","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-21 18:35:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"e9f04658bd46ac315ac4c8df405f083bbd9b8a2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 18:35:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fe763ca000c504c6a8236a4ba012bcabce71d616","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 19:37:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/df9424a617bc41659b2813ecd581d131 : SUCCESS in 45m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ecbf9bfa2da54e32bb692c331104c04f : SUCCESS in 46m 27s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/95ae7a8d90f542cb8f03b827baca883b : SUCCESS in 55m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fbd4051816264420949a051b13c624c8 : SUCCESS in 41m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/541189ed6152400197a376e8cf9cda1d : SUCCESS in 36m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87f4f877ab3e46c88972b69dd9a81e48 : SUCCESS in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fdf088afec143ea8debb7991982e0ae : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/626934e843594d32b361bb2ea0cde47b : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/026b45ea2c3f4f86ab12b7f89033207c : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15995d4f2b2a41f2abc5934b6c5eb765 : SUCCESS in 17m 08s","accounts_in_message":[],"_revision_number":1},{"id":"71e6e9a6508607ad5292dade9f0e385e35e3beeb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 19:37:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"49d54fdecaad12bdda19fc3c46514d73a84a430d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 19:39:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d2bea2db20540b68a1f134905812680 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1},{"id":"4cda0e05d3598c35926aaf79952f3d50f97b8a9f","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-09-22 03:34:38.000000000","message":"Patch Set 1:\n\n\u003e Interesting. I will try to figure out why you would need that, but\n \u003e fine as it is now.\n\nthanks, in kolla deployment, we create a random UUID for default pool id, as we don\u0027t use the value from config file ,so it will encounter a error. \n\n\nif you know why oslo config not works in manage command ,please tell me .thanks very much.","accounts_in_message":[],"_revision_number":1},{"id":"6688a31d9a9d4f54f887f8d5291e52597da55ac2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-29 16:43:26.000000000","message":"Patch Set 1:\n\nIt would be nice to backport this.","accounts_in_message":[],"_revision_number":1},{"id":"6e449c9ed9037dd70e96b4638b1c64a12801522d","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-29 20:53:59.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 9cf3b8ea4867e417f6e2f74b0647b71268b14848","accounts_in_message":[],"_revision_number":1},{"id":"f3376c2bff0dda453e9ef74d0056ba4b0c95ba9a","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-29 20:54:17.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit b47d4306a0979b2a7801d2be03e566b1e1efaf80","accounts_in_message":[],"_revision_number":1},{"id":"de242bc5faba897f16ad3fa7624b26cc02e936e4","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-29 20:54:32.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 9d08244fbd29aa79dc11ae4100de787e95aec50e","accounts_in_message":[],"_revision_number":1},{"id":"6eea9230eb241f6b47a159b97fc9406ec4430853","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-29 20:54:41.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 46778238e4263f4bb45fa84540bf391d1a6cce2a","accounts_in_message":[],"_revision_number":1},{"id":"cc25dfad368ea87c39cb0f4fca6e33e584bec797","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-29 20:54:53.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit 79aed2183d08f988f61a40dd4c7efc487184181d","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"609437e73deff55e9f5548538f093d01f0addd1b","revisions":{"609437e73deff55e9f5548538f093d01f0addd1b":{"kind":"REWORK","_number":1,"created":"2020-09-11 11:07:55.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/45/751245/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/45/751245/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/751245/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/751245/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/45/751245/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/45/751245/1"}}},"commit":{"parents":[{"commit":"06f147f76ddf5fa198fb7851ef14144bfb0fb9ee","subject":"Merge \"Remove Babel from lower-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06f147f76ddf5fa198fb7851ef14144bfb0fb9ee"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2020-09-11 10:57:49.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2020-09-11 11:07:52.000000000","tz":480},"subject":"fix database sync always get the wrong  value","message":"fix database sync always get the wrong  value\n\nwe create a default pool  when exec database sync, and get the\ndefault_pool_id from designate.conf, but we did not register this\nopt before using  it, so we always get the default value which is\nwrong\n\nChange-Id: I500c932a502db24332aad95c54713b7056bc1df4\nCloses-Bug: 1881277\nRelated-Bug: https://bugs.launchpad.net/kolla-ansible/+bug/1855877\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/609437e73deff55e9f5548538f093d01f0addd1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/609437e73deff55e9f5548538f093d01f0addd1b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
