)]}'
{"id":"openstack%2Fdesignate~752568","triplet_id":"openstack%2Fdesignate~master~Icac35fa4d4f31e30a9d069dacf9f3a87aef76531","project":"openstack/designate","branch":"master","topic":"adding_bindep_txt","hashtags":[],"change_id":"Icac35fa4d4f31e30a9d069dacf9f3a87aef76531","subject":"Adding bindep file to pre-install system dependencies","status":"ABANDONED","created":"2020-09-17 20:36:39.000000000","updated":"2020-09-18 06:55:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6804d6f62e0e4e6ed25aeaef19a88a0f19068040","_number":752568,"virtual_id_number":752568,"owner":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-18 03:20:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2020-09-18 05:20:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-18 03:20:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-18 05:20:59.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"cfe9579203fb022542e4b0c51c9c794ebc4947a4","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-09-17 20:36:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba6118a6ef619f0d0b137db495e3844db0aa2bd3","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-09-17 21:04:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cd34e253dd326915235027bc3db69676b3067c6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 22:09:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ec7aaf8499c741259ccaf4effd456a02 : SUCCESS in 4m 40s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b324d1cb820b4765b6cafaa58d7e673b : SUCCESS in 51m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/13d8f213efa040808cc84cf70f462ac3 : SUCCESS in 31m 21s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/075166c06bf94b11b8a51974e7ede023 : SUCCESS in 1h 03m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/992d52c57e5649a682ce6766a07700f3 : SUCCESS in 32m 11s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/19886d62b18b44188e0a7ad7bc93734c : SUCCESS in 52m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca90652329e4490688753c6d8e0bbb40 : SUCCESS in 6m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09d5c74b75394651868af7a6b2ba4a31 : FAILURE in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7bdb6586e4645ac9f96f5b774ee3f87 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10ac7c76f8af4a259a52d17630dc1ca6 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7b1a8059b134b309307b052b7db7a8c : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75a95ccf20704cef813881650e4b003a : SUCCESS in 24m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5a9479a34c54eb2aa36462dd6f7fa27 : SUCCESS in 4m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/937387b04b4441b1ac7b3bf67957dc87 : POST_FAILURE in 42m 05s","accounts_in_message":[],"_revision_number":2},{"id":"800ab8cd0647fe550103b65485ba06b2ae14f40a","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-09-17 22:47:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c85810afb39f735031cc3554220a8ee9a0f6e929","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 23:58:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2647aeea7929449e9fde3ac7dad51f6a : SUCCESS in 3m 42s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0aba63806ae84926a76c9e760371d554 : SUCCESS in 40m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/857e8a8db1cb4c768669be69cda9a882 : SUCCESS in 37m 20s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/78b9632d4cc749a08b78c9f078fa081e : SUCCESS in 1h 04m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5300a7e25e304a67b6219352b3fd64a5 : SUCCESS in 34m 51s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/dcd2784c14d54255827ecd6ad87c9f59 : SUCCESS in 41m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9db1381ca26140438d2723e81e2355f8 : SUCCESS in 5m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/760e25d30f2c4c3ea5bfd67170ab3f35 : FAILURE in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6baa2887f9c64e83bffebb473fd6a770 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8526267ab3204b06aad38da3150cfba8 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ad013f15bcb4d388969c44621ca51b1 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99b7480c2b904a4ba25c10158f554835 : SUCCESS in 17m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c14f60c8cccd4ae9a84a8a223cf2b101 : SUCCESS in 3m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b9b555bc323748b484d761d4b36fd1b2 : SUCCESS in 36m 46s","accounts_in_message":[],"_revision_number":3},{"id":"1b8744bfb0948de54b6bc509dc9b1fdabf38f27e","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-09-18 00:26:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"15e43e17bd42bc0e4cfe39e8285d4f2f2a25c286","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 01:24:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/38c9937d9e754200be904c342f445fad : SUCCESS in 6m 30s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5e0e7c2f97e543aca885fbe5992d33bc : SUCCESS in 37m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d6fef610017848d4bc181d6b344eab6f : SUCCESS in 39m 04s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/464fe652289e47f1ba0e98ff985ee62b : SUCCESS in 55m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d1187334baa449459ec83e0975c5e623 : SUCCESS in 35m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bf34882cbeae4d4ca85cd36276bff870 : SUCCESS in 34m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a023eeddba0c4d998d34fd09644cd9e7 : SUCCESS in 5m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a612b7eedad4c63ab9459e81e5282ae : FAILURE in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17f016fa224c4629be8bfa1f5c3b5c27 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f0d3f8d649945a2a78f22b4839fd8ee : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5c594ef19794f7dad4517141d8a894a : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f197238354e425a8848c3e60a8fc125 : SUCCESS in 17m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/871064cd50e74bf38184c96dc5b3ccf0 : SUCCESS in 2m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8282e3a5dbec44b1926e29e57a7063b6 : SUCCESS in 4m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/80b401f46bb94614bdabb3d9d834942d : SUCCESS in 32m 47s","accounts_in_message":[],"_revision_number":4},{"id":"e573c9676b93ff9b8b44472f176883e0fcbcb96b","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-09-18 02:16:52.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d900b80e6aa2af280b51cd5c4501615db29b844e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 03:20:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9dbf1dac3507453aa8b30dcd969a3662 : SUCCESS in 4m 01s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d656d912a484436089031ad1b1f6b080 : SUCCESS in 40m 42s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/79559cff13134355858837a2b16f33ad : SUCCESS in 37m 50s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8c85e11f04b84b5fb0cdcaee7f96a06b : SUCCESS in 57m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e6b78559d6224310b5b02a7a59e167ce : SUCCESS in 38m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/30c0478fb05d4fbbafb0ba0eb4904c99 : SUCCESS in 41m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03453a4cf6594f6db537f3d056725135 : SUCCESS in 5m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d697e21ada2745bf8afa3eb1afab662d : FAILURE in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c45682fccf0945e09bd0f1c28d718833 : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/597167f1fe3a4ca5b4b46c9391948a56 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ede7b17743c4b09853c13817ae0eccd : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51a0bf3e0fa1461280e5cb7654c8fbdd : SUCCESS in 13m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/51df83f6144e475496899fffdd0d0091 : FAILURE in 2m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a149faccd6f4988ac6c9aab9b9aa01b : SUCCESS in 3m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/91baa2a82c2649278cf4b8dfa7f979f7 : SUCCESS in 33m 09s","accounts_in_message":[],"_revision_number":5},{"id":"7cc3ce5efd08929c1b3ab92f55a5605129eb3f57","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-09-18 05:20:59.000000000","message":"Patch Set 5: Code-Review-1\n\nDuplicate of https://review.opendev.org/751655","accounts_in_message":[],"_revision_number":5},{"id":"7d86fd51a705149f317a34b80779c3cd595b7001","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-09-18 06:55:54.000000000","message":"Abandoned\n\nDuplicate","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f53ce3af8f69750796c2ccbaa34e9bbdeac91d3e","revisions":{"41c3d35e8866350ad219d264acc5c72cb9493a68":{"kind":"REWORK","_number":1,"created":"2020-09-17 20:36:39.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/68/752568/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/752568/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/752568/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/752568/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/752568/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/752568/1"}}},"commit":{"parents":[{"commit":"06f147f76ddf5fa198fb7851ef14144bfb0fb9ee","subject":"Merge \"Remove Babel from lower-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06f147f76ddf5fa198fb7851ef14144bfb0fb9ee"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-17 20:31:59.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-17 20:36:31.000000000","tz":0},"subject":"Adding bindep file to pre-install system dependencies","message":"Adding bindep file to pre-install system dependencies\n\ndesignate lower-constraints is trying to pip install cffi\u003d\u003d1.11.5\nwith python3.8 (the default python3 on ubuntu 20.04 lts), but cffi\ndidn\u0027t publish prebuilt cp38 manylinux1 wheels for that release so\nit\u0027s having to be built from the sdist instead, and designate doesn\u0027t\nhave any bindep file saying to preinstall libffi to build against,\nso the compiler is erroring out\n\nChange-Id: Icac35fa4d4f31e30a9d069dacf9f3a87aef76531\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/41c3d35e8866350ad219d264acc5c72cb9493a68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/41c3d35e8866350ad219d264acc5c72cb9493a68"}]},"branch":"refs/heads/master"},"ac8ddb9db527c4df80d69fd92b554a7f171281e5":{"kind":"REWORK","_number":2,"created":"2020-09-17 21:04:00.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/68/752568/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/752568/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/752568/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/752568/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/752568/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/752568/2"}}},"commit":{"parents":[{"commit":"06f147f76ddf5fa198fb7851ef14144bfb0fb9ee","subject":"Merge \"Remove Babel from lower-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06f147f76ddf5fa198fb7851ef14144bfb0fb9ee"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-17 20:31:59.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-17 21:03:32.000000000","tz":0},"subject":"Adding bindep file to pre-install system dependencies","message":"Adding bindep file to pre-install system dependencies\n\ndesignate lower-constraints is trying to pip install cffi\u003d\u003d1.11.5\nwith python3.8 (the default python3 on ubuntu 20.04 lts), but cffi\ndidn\u0027t publish prebuilt cp38 manylinux1 wheels for that release so\nit\u0027s having to be built from the sdist instead, and designate doesn\u0027t\nhave any bindep file saying to preinstall libffi to build against,\nso the compiler is erroring out\n\nChange-Id: Icac35fa4d4f31e30a9d069dacf9f3a87aef76531\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ac8ddb9db527c4df80d69fd92b554a7f171281e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ac8ddb9db527c4df80d69fd92b554a7f171281e5"}]},"branch":"refs/heads/master"},"603f3848b09960b8e8f4fd83643938d76dd2112b":{"kind":"REWORK","_number":3,"created":"2020-09-17 22:47:05.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/68/752568/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/752568/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/752568/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/752568/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/752568/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/752568/3"}}},"commit":{"parents":[{"commit":"06f147f76ddf5fa198fb7851ef14144bfb0fb9ee","subject":"Merge \"Remove Babel from lower-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06f147f76ddf5fa198fb7851ef14144bfb0fb9ee"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-17 20:31:59.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-17 22:46:27.000000000","tz":0},"subject":"Adding bindep file to pre-install system dependencies","message":"Adding bindep file to pre-install system dependencies\n\ndesignate lower-constraints is trying to pip install cffi\u003d\u003d1.11.5\nwith python3.8 (the default python3 on ubuntu 20.04 lts), but cffi\ndidn\u0027t publish prebuilt cp38 manylinux1 wheels for that release so\nit\u0027s having to be built from the sdist instead, and designate doesn\u0027t\nhave any bindep file saying to preinstall libffi to build against,\nso the compiler is erroring out\n\nChange-Id: Icac35fa4d4f31e30a9d069dacf9f3a87aef76531\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/603f3848b09960b8e8f4fd83643938d76dd2112b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/603f3848b09960b8e8f4fd83643938d76dd2112b"}]},"branch":"refs/heads/master"},"3bd84955fbdc3cbb145dcfb175f046f9a69f9779":{"kind":"REWORK","_number":4,"created":"2020-09-18 00:26:24.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/68/752568/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/752568/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/752568/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/752568/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/752568/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/752568/4"}}},"commit":{"parents":[{"commit":"06f147f76ddf5fa198fb7851ef14144bfb0fb9ee","subject":"Merge \"Remove Babel from lower-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06f147f76ddf5fa198fb7851ef14144bfb0fb9ee"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-17 20:31:59.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-18 00:23:49.000000000","tz":0},"subject":"Adding bindep file to pre-install system dependencies","message":"Adding bindep file to pre-install system dependencies\n\ndesignate lower-constraints is trying to pip install cffi\u003d\u003d1.11.5\nwith python3.8 (the default python3 on ubuntu 20.04 lts), but cffi\ndidn\u0027t publish prebuilt cp38 manylinux1 wheels for that release so\nit\u0027s having to be built from the sdist instead, and designate doesn\u0027t\nhave any bindep file saying to preinstall libffi to build against,\nso the compiler is erroring out\n\nChange-Id: Icac35fa4d4f31e30a9d069dacf9f3a87aef76531\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bd84955fbdc3cbb145dcfb175f046f9a69f9779"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bd84955fbdc3cbb145dcfb175f046f9a69f9779"}]},"branch":"refs/heads/master"},"f53ce3af8f69750796c2ccbaa34e9bbdeac91d3e":{"kind":"REWORK","_number":5,"created":"2020-09-18 02:16:52.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/68/752568/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/68/752568/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/752568/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/752568/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/68/752568/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/68/752568/5"}}},"commit":{"parents":[{"commit":"06f147f76ddf5fa198fb7851ef14144bfb0fb9ee","subject":"Merge \"Remove Babel from lower-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06f147f76ddf5fa198fb7851ef14144bfb0fb9ee"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-17 20:31:59.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-18 02:16:23.000000000","tz":0},"subject":"Adding bindep file to pre-install system dependencies","message":"Adding bindep file to pre-install system dependencies\n\ndesignate lower-constraints is trying to pip install cffi\u003d\u003d1.11.5\nwith python3.8 (the default python3 on ubuntu 20.04 lts), but cffi\ndidn\u0027t publish prebuilt cp38 manylinux1 wheels for that release so\nit\u0027s having to be built from the sdist instead, and designate doesn\u0027t\nhave any bindep file saying to preinstall libffi to build against,\nso the compiler is erroring out\n\nChange the lower constrains to match with python 3.7\ngreenlet support for python 3.7 started with 0.4.16\n\nChange-Id: Icac35fa4d4f31e30a9d069dacf9f3a87aef76531\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f53ce3af8f69750796c2ccbaa34e9bbdeac91d3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f53ce3af8f69750796c2ccbaa34e9bbdeac91d3e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
