)]}'
{"id":"openstack%2Fdesignate~752640","triplet_id":"openstack%2Fdesignate~master~Ia1a0c9102fcc930d96b6e4c0f2c5ef4bef70c5be","project":"openstack/designate","branch":"master","topic":"bug/1714088","hashtags":[],"change_id":"Ia1a0c9102fcc930d96b6e4c0f2c5ef4bef70c5be","subject":"Fix resource links in API endpoints","status":"MERGED","created":"2020-09-18 08:09:33.000000000","updated":"2020-11-16 23:00:41.000000000","submitted":"2020-11-16 22:59:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752640-1605567563135-28503777","meta_rev_id":"0c585a2126a2044ffee581f57826edf20ae3ba29","_number":752640,"virtual_id_number":752640,"owner":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},{"value":0,"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-16 22:59:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2020-11-16 22:06:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},{"value":1,"date":"2020-11-09 11:09:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2020-11-02 02:49:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":1,"date":"2020-11-16 22:06:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},{"value":0,"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},{"value":0,"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-10-15 16:08:35.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-14 23:41:09.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-11-02 02:49:08.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-11-09 11:09:20.000000000","updated_by":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"reviewer":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"state":"REVIEWER"},{"updated":"2020-11-16 22:06:24.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-11-16 22:59:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e543ccaa34cdbb8479b8f1b530478d0fbda30cd5","author":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"date":"2020-09-18 08:09:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52288e18adffa7bb12ffb2f4ef2694ba4bc4abea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 09:10:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/444eb2b8b55c4f0f8f9f66183910fe95 : SUCCESS in 38m 45s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6599b0d278934e30a948f4b9be183bf4 : FAILURE in 35m 25s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c7e73a4563b6496d9b870bc74cb0b313 : SUCCESS in 53m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7e76965f304c491fb59608f3fb4137fe : SUCCESS in 40m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d3b6ffd1df7e4e888d7ffe0db714e399 : SUCCESS in 38m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43574188ab674ec1b7414e8532f1f394 : SUCCESS in 6m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e82dfbceb09456e8105e2491f2bf414 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7604fca71a640d28f75593fd90e5110 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3c1bfc26c4040b180b989fadca7d615 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89f6c13cd330471abf4d135eefba3068 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b9b11cc824e42749273795869a924ea : SUCCESS in 17m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bf4f34c739c44db5b89737ff27fe5979 : SUCCESS in 34m 46s","accounts_in_message":[],"_revision_number":1},{"id":"b78c7e04939c8a589e1a26c4d3685ecc05fcccef","author":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"date":"2020-09-18 09:51:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d43b8533fdd89c984f05dde8789e99fbf6981e2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 10:52:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ee4dfd370fa4408a8baaf92afaec3890 : SUCCESS in 37m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4aa13a593b89464fa024d2e3c9485e8c : SUCCESS in 32m 49s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/098cce2494494004bda4ca9763a721e7 : SUCCESS in 52m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/76a02c3486d14619812e9326dc4bcecc : SUCCESS in 43m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f9694097be6e4e86bc98a5b64f3c2c79 : SUCCESS in 44m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd917a83848f44d8af7ace637e1401c5 : SUCCESS in 5m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ef6213ae502409ba104939047af10ab : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59638b02816c4f9293f6fafec6e9a6f2 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fdd40091b84341ebb517147d7b2f018a : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/211046308fb74fa49c60bc0c809b0851 : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e3c29c3a28f4a34bdaa5f72d1f1d8d1 : SUCCESS in 14m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5a4204916d944775a1b4e828917bca38 : SUCCESS in 33m 28s","accounts_in_message":[],"_revision_number":1},{"id":"2a1e75b37889732f417c3ac367cb0f43eff76b9b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-09-20 18:51:18.000000000","message":"Patch Set 1:\n\nCould you add a test or two as well?","accounts_in_message":[],"_revision_number":1},{"id":"238b2379ac4ad0c6bb69a20009028106afa5616b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-09-20 18:51:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"34903dfcee6d49e0d745ec94c05684a728bd9e41","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-21 18:35:58.000000000","message":"Patch Set 1: Code-Review+1\n\nI second Erik\u0027s suggestion regarding tests.","accounts_in_message":[],"_revision_number":1},{"id":"1e020622b4ff290356c08ee3811ad5c17fe768da","author":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"date":"2020-10-15 16:08:35.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"0ad7b0662bfe42826c431d6eee498bfb835555b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 17:15:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/07ca5b72b86d4cbaba3712ba0ba49264 : SUCCESS in 38m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/45be73aa33af4720bacb20fda3a93605 : FAILURE in 5m 13s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/abe1fb0a5b1b41359b3f631161ac177f : SUCCESS in 45m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/811f93ca05554787b962f5d367a1b54b : SUCCESS in 1h 04m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5177b006842a4c979cb0def6e2f0e407 : SUCCESS in 58m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/731fcc4216b6448f9a26e2045b455920 : SUCCESS in 5m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a3e38550ae64ca2b6cb60c81f7cb50d : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4797e0fb265e49119e315474ecc913f9 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9efef7440d0947789d96f83b2c02f9d5 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67f63cb9dd2f4062b5abb6fd4459d3ae : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9afe86eccbca48a4a7d14cc612e72a18 : SUCCESS in 7m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2389e1da532e48a28504713df5d41700 : SUCCESS in 41m 01s","accounts_in_message":[],"_revision_number":2},{"id":"7e04747027a6c78eea6cec4b2d0fab02f4c6f680","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-10-18 20:39:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1033675a82ffde26306e812af2feafea80b47664","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-18 22:18:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/36e99e41a7ac4a769e79a1faf43922b3 : SUCCESS in 34m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e89c71d2a9d84d4ab89f5c4a757b6f10 : FAILURE in 6m 40s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a9d7d4e542eb4cdfbc14938d76277ee7 : SUCCESS in 45m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c07ca6d3b1154825ba9280f6c8417fd1 : SUCCESS in 1h 23m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b5be2ca81cd14a3b84cac245e90c9845 : SUCCESS in 57m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78e6fb9882bb4f598f7ff144216416ed : SUCCESS in 7m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42ba2eae884746bdb98973a7f7d21e6c : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a04ed8bba85406bb4e68466699df130 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e885235369f44dc9a4d564b0f72f7b0 : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7759f896e3b4ce487dc71f70c78fe1b : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a85afa36e3b474e847c5a1c83aaa761 : SUCCESS in 9m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ae2d1db77adb4eb58c1650d90e3238ff : SUCCESS in 34m 20s","accounts_in_message":[],"_revision_number":2},{"id":"fa4a80162e1afa835bcedb92efb3216462d7a913","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-11-02 02:47:50.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f80ec790d9ac03bf5b4a9062018bfbf31df2af28","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-11-02 02:49:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f0ce0761c42d1b7f1a24a7a1b626d2ef7bce0001","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 03:44:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a323c5b1b9984f1fb6f46104b6dfdec0 : SUCCESS in 40m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c9e6993a7f48413ea83520ae3f68bf61 : SUCCESS in 41m 48s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b04c6dfde56f46d68a1946d48cb73922 : SUCCESS in 54m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/26573454ef8d49a1985ef21cde1408c9 : SUCCESS in 36m 51s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f35e5e0a5f88496a88483dbd433edb36 : SUCCESS in 35m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20a3a60bea314bb19db3d0d31b59c28e : SUCCESS in 4m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b87ec85ccd54586a6271ded27eefbb4 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99877f7fdcf542998a0da38cb6393715 : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fee0c665a51c4215a8e0dbb0f5273180 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b68aced320f14b358bbd631ba62d26f0 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53f3d2ea903b44f7884460af52172466 : SUCCESS in 6m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/943797543fae4a2a824b69aa2b1f8c93 : SUCCESS in 39m 54s","accounts_in_message":[],"_revision_number":2},{"id":"2d802707897b92e712017c104390c592eab7bbaf","author":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"date":"2020-11-09 11:09:20.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2880a4563a530569ac11bf08bf94e3b3bfe8be64","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-16 22:06:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ad1cda073c7f4ffcbda82aad04587550aa4b3cdd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 22:06:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"30991fb59b1a3b077c67de2e5277b4527e8ab80e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 22:59:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4f41cf86e56d4ebe84f18e6241e918e2 : SUCCESS in 37m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6279a737be674619baa2d3129c266840 : SUCCESS in 42m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/af7bb256e4b34028ac257af2a0e23b59 : SUCCESS in 48m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/db0512e721bc4b278bf811afdf97d4f6 : SUCCESS in 39m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/26aeb1cce87b470999ead1503c88441e : SUCCESS in 46m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f59acc92000541059c41888936040611 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dfe39c408d345c69cb7f8627a21d706 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/943f6297344043f3b9281a53aab88c85 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49ecc213e1634376bf9487818e2447b5 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9780661c9e74ce2a75a8fd1d49489ea : SUCCESS in 7m 40s","accounts_in_message":[],"_revision_number":2},{"id":"46a25f8a78ee51846329ec0dc413b47d717ef764","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 22:59:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e6f19341914ca3e1d7f5623f3d9a666d0a4a8663","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 23:00:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76b24e14f2a247eb8b9a5f13cde433be : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b823eb35071214336bf8e5c96f64444c09e27e41","revisions":{"a654df1d8265be5a8c57b9fe308f15eb37c7861d":{"kind":"REWORK","_number":1,"created":"2020-09-18 08:09:33.000000000","uploader":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"ref":"refs/changes/40/752640/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/40/752640/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/40/752640/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/40/752640/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/40/752640/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/40/752640/1"}}},"commit":{"parents":[{"commit":"28037784eb4c2546bbe61e1a8bb7d34a2b8301f3","subject":"Fix l-c testing for ubuntu focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/28037784eb4c2546bbe61e1a8bb7d34a2b8301f3"}]}],"author":{"name":"Igor Malinovskiy","email":"u.glide@gmail.com","date":"2020-09-18 08:06:19.000000000","tz":180},"committer":{"name":"Igor Malinovskiy","email":"u.glide@gmail.com","date":"2020-09-18 08:09:12.000000000","tz":0},"subject":"Fix resource links in API endpoints","message":"Fix resource links in API endpoints\n\nChange-Id: Ia1a0c9102fcc930d96b6e4c0f2c5ef4bef70c5be\nCloses-Bug: #1896192\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a654df1d8265be5a8c57b9fe308f15eb37c7861d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a654df1d8265be5a8c57b9fe308f15eb37c7861d"}]},"branch":"refs/heads/master"},"b823eb35071214336bf8e5c96f64444c09e27e41":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-10-15 16:08:35.000000000","uploader":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"ref":"refs/changes/40/752640/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/40/752640/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/40/752640/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/40/752640/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/40/752640/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/40/752640/2"}}},"commit":{"parents":[{"commit":"9eae67abc6e5aaac061c4576a82f54f3653cef4a","subject":"Merge \"Remove the unused coding style modules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9eae67abc6e5aaac061c4576a82f54f3653cef4a"}]}],"author":{"name":"Igor Malinovskiy","email":"u.glide@gmail.com","date":"2020-09-18 08:06:19.000000000","tz":180},"committer":{"name":"Igor Malinovskiy","email":"u.glide@gmail.com","date":"2020-10-15 16:07:45.000000000","tz":180},"subject":"Fix resource links in API endpoints","message":"Fix resource links in API endpoints\n\nChange-Id: Ia1a0c9102fcc930d96b6e4c0f2c5ef4bef70c5be\nCloses-Bug: #1896192\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b823eb35071214336bf8e5c96f64444c09e27e41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b823eb35071214336bf8e5c96f64444c09e27e41"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
