)]}'
{"id":"openstack%2Fdesignate~753495","triplet_id":"openstack%2Fdesignate~master~I3a6acbe9dac8bbf91a4926c806a1fabac028cda3","project":"openstack/designate","branch":"master","hashtags":[],"change_id":"I3a6acbe9dac8bbf91a4926c806a1fabac028cda3","subject":"Remove install unnecessary packages","status":"MERGED","created":"2020-09-23 03:45:56.000000000","updated":"2020-09-30 20:22:04.000000000","submitted":"2020-09-30 20:20:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753495-1601497219217-7cf8cb66","meta_rev_id":"6eceac6131a2e44cc07fdb3560514b3d769c47d9","_number":753495,"virtual_id_number":753495,"owner":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-30 20:20:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"recommended":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"all":[{"value":1,"date":"2020-09-23 14:10:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":2,"date":"2020-09-23 14:30:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-28 05:42:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":0,"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":1,"date":"2020-09-28 12:27:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-23 14:10:02.000000000","updated_by":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"reviewer":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"state":"REVIEWER"},{"updated":"2020-09-28 05:42:28.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2020-09-28 12:27:17.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-09-30 20:20:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"76c9c7bd5c1fa9d005b2b4d5c3415442510113e5","author":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"date":"2020-09-23 03:45:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d6f27650e72fa212623c446e01715a835aea2e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 04:53:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/89a9ea94ece449de91428fd13dfa43c7 : SUCCESS in 3m 31s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/04d3f3f979ee4b198b26e4dda3fc120c : SUCCESS in 38m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0863bc5f2a574942aed002c6159d23d4 : SUCCESS in 36m 06s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6f2760a72a8e459bad570bca1da3ce60 : SUCCESS in 1h 01m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9acd9872185c4e998ec39dc14ac53353 : SUCCESS in 37m 11s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b670e085cdae45d1bd0e1210197f42f2 : SUCCESS in 38m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39cad6d671a84d66a927c3abd31c1637 : SUCCESS in 4m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/991c7712e56d4c588f2b0c266229c60f : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccb5e0fbab42465091a2b220d3e0ae52 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37587376c950464c8209487f4f8866d6 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fe76bed8ff346819624af7d03d12893 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8331dd9de05c48eeb640223768c1d2d2 : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2894abd961d43449c8ce126ad7e6e66 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"415741b35e5900e4ff5d5138e7aacb956204fd21","author":{"_account_id":6994,"name":"Michael Chapman","email":"woppin@gmail.com","username":"michaeltchapman"},"date":"2020-09-23 14:10:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6d07decb896466946bf9af00002ff80caa732226","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-23 14:30:56.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"05f158e612cfd8d59f2f849f269f84fe11199cb6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-09-28 05:42:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ccca2d64e914026d879db1b66ab23ef084905037","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-28 12:27:17.000000000","message":"Patch Set 1: Workflow+1\n\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"784ef098ee537ca3bbcb79e6be284da25ecdad22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 12:27:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8b009fd488d34ea781dd2ddf24d7663002d8bed1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 13:18:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6f13fe22e3ca4a1f8534e0d42d6ed68e : SUCCESS in 5m 03s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/56e8bd2494a44087868fefb6ccf384e5 : SUCCESS in 36m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e9abf272e7804f0d904ceb946accdf67 : FAILURE in 18m 35s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1fc1697baaed4e78aa2c268d57faf28d : SUCCESS in 48m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/69f9c7830ebd4b1aa05c95bd6b3411b1 : FAILURE in 19m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/15cde50326824c8499a23bec113e78fe : SUCCESS in 33m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc86f66a4eff4242affd173e173823a9 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50e7cfe427054b3b8fe040a929f4c823 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f8a62d5e0b24ce3a1a97baaa9609ec1 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fe0b75b903d409abbf6008473a4a538 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a8c8406a0ad4f3b9bac907ab19d685c : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ba16046a98c416e99a7888108981403 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"6298ac2e5082d6d4a934d48d616f410a7451f710","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-28 14:33:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d37a1a619a12e6548885f585095788b9bb068196","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 15:28:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e8382f22f86e45c88b3b3ff17727bd98 : SUCCESS in 5m 26s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c4f69d86c7dd4a86bdd293d1e05e7f1e : SUCCESS in 46m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1032bc8ecc2841e99a3f449497d21eb3 : FAILURE in 20m 06s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/366ab1029ad240cdaa09761384f76e64 : SUCCESS in 52m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b7ac310d4717480bb9c8113372bbfb96 : FAILURE in 21m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c6b0dc1414b648bca0acafb8ebdb9b73 : SUCCESS in 41m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b775bf6b22e94f86b5066b91fe04493a : SUCCESS in 6m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f7a2f4977ca4ea18bb1f23b39ef8ff0 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b99c94c9c90d4dcf9645c869440f8620 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28b4dc71dae44263b1802a3404c69d92 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6ba6bfe557841d58d9136ce31812068 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f84a1cb7b49049eb8f94413b81724d3b : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/049ef41a2c2044509bc0c7e36d1f77cc : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":1},{"id":"f0ee8e3e80dc3be076157ab02ef385b1803721b4","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2020-09-30 18:14:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"defb69974e0add41a8e6f728da0e67a713d88e0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 19:22:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c2196e7903bf4b629d51baf1a42fa97a : SUCCESS in 3m 48s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/153f94bf116a4aacbbd6edb88d377bd0 : SUCCESS in 35m 14s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d00651cee0e54488be286dc69d370a7d : SUCCESS in 38m 40s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f3f037a952c043d5ace44ee0cd4a9f0d : SUCCESS in 1h 00m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e81b074ede6d4353a675b668f437fea1 : SUCCESS in 47m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f43392069ee44b7680d68ec0af9a1e73 : SUCCESS in 37m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a44d76377a84a5db5b4c2b05d94bdb0 : SUCCESS in 5m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/312446a4d40d4e8db5adefd84cf7cabc : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad00b9e060ae4e26b02f9338cdb09612 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46594b918ebc4890a021247e4b8bbd4a : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/441d457e7d6a4d8d95e8e4a6c7c74628 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a40d962aac994443bdd87a2d251ba85a : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/149d1d8a3018431d9fade3f37e5581c9 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":1},{"id":"4b9183ca0bef9fb5a9d1d3cd9b86ba6ded0751d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 19:22:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c4647a65efa1fc4111251ac2c6baafe1adce76f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 20:20:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"be8c218d96aff0e5d7a5d5fca1a1196110bef93e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 20:20:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/45a567b8528b47e3ab3e1463144e1686 : SUCCESS in 3m 47s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/86e9bcd41c68429bab4580f58cc96bde : SUCCESS in 43m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0b5dc3af29ee41a4b0bffb0e4cea573d : SUCCESS in 38m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9ddad55fb46647fe92d84b0e2f20ed1e : SUCCESS in 55m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/73f9833ad68043cc811e45bf8a7d1479 : SUCCESS in 43m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6ca341662a9b4728ac8d8c43afa2c631 : SUCCESS in 49m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57182200a73a4fbf8f51a76e1e88e26c : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47a21c15498544caa7de978222211a12 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ef76eb52c934ae488880736e00378ac : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4f64a928f3d4ed7807590c4d4c66b19 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe231eaf9e464b708a3961d48a1a9ef4 : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed9f53acab6e4f19b6791ed0abdd86a7 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"6eceac6131a2e44cc07fdb3560514b3d769c47d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 20:22:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/85561c8a82864b1eb58dc8857640d36c : SUCCESS in 1m 02s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e3588efc994405782b3311167eb08dc : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56b0f24545b84b08ac16558b1747a045 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc0802e3592b5dde1e84eada2a881b79daabe355","revisions":{"fc0802e3592b5dde1e84eada2a881b79daabe355":{"kind":"REWORK","_number":1,"created":"2020-09-23 03:45:56.000000000","uploader":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"ref":"refs/changes/95/753495/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/95/753495/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/95/753495/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/95/753495/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/95/753495/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/95/753495/1"}}},"commit":{"parents":[{"commit":"2168b4367d483ce420afee4bc1e5fca371cc40b8","subject":"Merge \"fix database sync always get the wrong  value\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2168b4367d483ce420afee4bc1e5fca371cc40b8"}]}],"author":{"name":"maaoyu","email":"maaoyu@inspur.com","date":"2020-09-23 03:45:15.000000000","tz":480},"committer":{"name":"maaoyu","email":"maaoyu@inspur.com","date":"2020-09-23 03:45:15.000000000","tz":480},"subject":"Remove install unnecessary packages","message":"Remove install unnecessary packages\n\nThe docs requirements migrated to doc/requirements.txt\nwe need not install things from requirements.txt.\n\nChange-Id: I3a6acbe9dac8bbf91a4926c806a1fabac028cda3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc0802e3592b5dde1e84eada2a881b79daabe355"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc0802e3592b5dde1e84eada2a881b79daabe355"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
