)]}'
{"id":"openstack%2Fdesignate~753511","triplet_id":"openstack%2Fdesignate~master~I5d53296caa61d708c6d5d972c9145962931e1df1","project":"openstack/designate","branch":"master","hashtags":[],"change_id":"I5d53296caa61d708c6d5d972c9145962931e1df1","subject":"Remove install unnecessary packages","status":"ABANDONED","created":"2020-09-23 06:15:25.000000000","updated":"2020-09-24 06:56:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4b255db8d54480c5c348a0117aa33505feffaed7","_number":753511,"virtual_id_number":753511,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-23 07:32:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":-2,"date":"2020-09-23 13:29:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-23 07:32:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-23 13:29:10.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"}],"messages":[{"id":"4c931f11577387ba555255b6c75e867a0718610a","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2020-09-23 06:15:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eaf2b691ea90b194cbb24065f7f2de94768a5e85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 07:32:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e9e4196c0c54421ba945ba6cd639f291 : SUCCESS in 6m 03s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e6e84b57344b4c1e8557c0cd78452d88 : SUCCESS in 33m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2d71f61542c24994aff0bced776e825b : SUCCESS in 35m 17s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7af98d0590c64cca942ab8893d14af3f : SUCCESS in 1h 03m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/749a928bc7f942c49e5ade8de8e353b0 : SUCCESS in 39m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/77c8d7908a6e4fd5b98ced42d3d332c7 : SUCCESS in 40m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9457787f64845a780c75edf646a4d85 : SUCCESS in 6m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ae1a0a4c0aa477e95b79f8c00a3bc04 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf4dc75fe6c84ec1b353b5edbd5fb9c0 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8df92d90304041bfaff6df641a832dba : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74c47ff848a044cd8e28c4b9d4efdd7e : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7326ee1689748bbb163edd3b81e6f09 : SUCCESS in 8m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d574b3cef7dd4df9a790f6557688d9c9 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"4e49d1b2fce775be1a5d532c78250a9bad1fb92f","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-23 13:29:10.000000000","message":"Patch Set 1: Code-Review-2\n\nDuplicate of 753495","accounts_in_message":[],"_revision_number":1},{"id":"4b255db8d54480c5c348a0117aa33505feffaed7","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2020-09-24 06:56:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4ed7c10aeff2b6e29c289a6352f444bea359253d","revisions":{"4ed7c10aeff2b6e29c289a6352f444bea359253d":{"kind":"REWORK","_number":1,"created":"2020-09-23 06:15:25.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/11/753511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/753511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/753511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/753511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/753511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/753511/1"}}},"commit":{"parents":[{"commit":"2168b4367d483ce420afee4bc1e5fca371cc40b8","subject":"Merge \"fix database sync always get the wrong  value\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2168b4367d483ce420afee4bc1e5fca371cc40b8"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2020-09-23 06:13:45.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2020-09-23 06:15:17.000000000","tz":480},"subject":"Remove install unnecessary packages","message":"Remove install unnecessary packages\n\nThe docs  requirements migrated to doc/requirements.txt\nwe need not install things from requirements.txt.\n\nChange-Id: I5d53296caa61d708c6d5d972c9145962931e1df1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4ed7c10aeff2b6e29c289a6352f444bea359253d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4ed7c10aeff2b6e29c289a6352f444bea359253d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
