)]}'
{"id":"openstack%2Fdesignate~754226","triplet_id":"openstack%2Fdesignate~master~Ie5a01f8df00edb7d67ab2db86b889ecac0d0efb0","project":"openstack/designate","branch":"master","topic":"bug/1875939","hashtags":[],"change_id":"Ie5a01f8df00edb7d67ab2db86b889ecac0d0efb0","subject":"Adding split horizon documentation","status":"ABANDONED","created":"2020-09-24 22:38:02.000000000","updated":"2022-09-24 05:53:22.000000000","total_comment_count":36,"unresolved_comment_count":16,"has_review_started":true,"meta_rev_id":"861cbe63ce894fb1a3030d5c5637136bb6b551ac","_number":754226,"virtual_id_number":754226,"owner":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-19 22:59:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-29 06:44:55.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-10-02 19:33:40.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-10-02 19:48:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-07 23:48:23.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-02-26 20:20:08.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"e1ccff970acf0bf952655c5e026353513d38e4fb","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-09-24 22:38:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7aa867772b971b79b76d93d83733e6a98f4087be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 22:49:25.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e40e630f9f884953b3e2b5db34435435 : FAILURE in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/933ca6c6dd79422a8dd7ab91ddcbf0c6 : FAILURE in 5m 34s","accounts_in_message":[],"_revision_number":1},{"id":"5eadba28dbc53d915cb69e6231fa88e87edaed15","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-09-24 23:04:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6924ea771c3f58ad248c019003eff86681899348","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 23:14:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb01fc2b179e4262a85ed904767ec680 : FAILURE in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21ed74a4a7ca449b9c171dc9756fca19 : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":2},{"id":"f34ca962fb373d96966feddce5645466f23636b8","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-09-24 23:35:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d30c5067ef8662c1379dd94489feff8f8653fad9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 23:48:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a280fa0ce374270ab53bc71390896e1 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/291321b85a514957b19ed4c153bf4c71 : SUCCESS in 8m 19s","accounts_in_message":[],"_revision_number":3},{"id":"2e47cd00213ac7d3c1b9cbd7bfb28f7cb9c618d5","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-25 01:57:40.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThanks for adding the documentation! Could you have a look at the copyright statement? Thanks!","accounts_in_message":[],"_revision_number":3},{"id":"f7fbc6e5c5df129af71793d65af236849d72bb1f","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-09-25 16:46:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"107911a2780af6be4a80efb314369eb6223c2651","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-09-25 16:47:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"59a0c5d1e3e9b2cc56503354bcd149f71bf5d45d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 17:12:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af606f008a3244709bea90f81e056364 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61f8dfdc73c9403cb3f9dddfa0586995 : SUCCESS in 8m 20s","accounts_in_message":[],"_revision_number":4},{"id":"853a5e1a0ce4636ae963a49c324536b06a0e60f8","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-25 18:03:44.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"82e6bc887a6ade79eca0dcd0e2151d4ef0fc75ac","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-25 18:04:18.000000000","message":"Patch Set 5:\n\nThanks hamza! I made some very minor edits and uploaded another change set. Please have a look.","accounts_in_message":[],"_revision_number":5},{"id":"ce2ebab1275c5e2616cc944a90191bd12e98bddc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 18:16:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6599327dc3e74b80a12b54ef3e7703be : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d649623ba594d409e52513faaa7d3a9 : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":5},{"id":"6b0b654d8f41829492f6872c062904e26fa16596","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-25 19:16:23.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c7738735406f5879c0a2be38f6c32532c41f641b","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-25 19:18:54.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8b822357f10fb9aa21a85fda042a744819977f53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 19:34:00.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4aa2fc52b36c4f738a3aeab98dfc0513 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cf25341fc204001a374e02bdfa2a1af : SUCCESS in 6m 59s","accounts_in_message":[],"_revision_number":7},{"id":"9664a7085b3d2b71962aad1b3b498712c33574f2","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-09-28 16:34:33.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9bfd232b46cf594aaeb71353696ab7c588af740d","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-09-28 16:40:05.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"220e0d3ad96be06fdec0fb43d5c5434208ffd5a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 16:56:18.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06c13445656148fc941d63e685cdaf47 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ebaf39fa4dd4deab57c391264c4abab : SUCCESS in 8m 30s","accounts_in_message":[],"_revision_number":9},{"id":"2369840b465c28ae5f1d5305ed171cb05d1393b3","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-09-28 17:03:14.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"8c4a02e3223c143e76c05f52c8a0c9193db13b25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 17:19:19.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a82651adceff46bba18668ba6c40c1b6 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaecc79ff7f34848973ddcaeef7464c4 : SUCCESS in 9m 35s","accounts_in_message":[],"_revision_number":10},{"id":"56b71d4c4dbdb1a7573fab9531d685b6b5ae09fa","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-09-28 22:41:30.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)\n\nHamza, did you accidentally change my changes back? Could you have a look at change set 7 and base your edits on that? Thanks!","accounts_in_message":[],"_revision_number":10},{"id":"38f9053c1a460a009a18ed76bd50a0d963363fc1","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-09-29 01:27:51.000000000","message":"Patch Set 10:\n\nsorry nicolas i forgot to update my version of the change","accounts_in_message":[],"_revision_number":10},{"id":"2d2da3ee0bbbe349f3ba5c7224b14fb02e7d7cb9","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-09-29 01:36:03.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"20e94ebf975eda850065507536a8464963cbdcb3","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-09-29 01:36:53.000000000","message":"Patch Set 11:\n\nnicolas can u please re-do the request or just point me to change it","accounts_in_message":[],"_revision_number":11},{"id":"25e6a401c793dda4ec9699235232f1577830b7fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 01:43:50.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b423b59c447498a8d5a6d079d3c211a : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe72193562de4f9ab836c267611a0902 : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":11},{"id":"30d9e3f4818a59156887c8ea19a7c2829abf6b8f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-09-29 06:44:55.000000000","message":"Patch Set 11:\n\nI\u0027d be interested in seeing a complete example showing how this is to be used in the real world. Configuring things is nice, but how will this apply to an end user?","accounts_in_message":[],"_revision_number":11},{"id":"d40b9fee3cd0b7c2c102c0deac8c1629ecffeebf","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-09-30 22:20:47.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"816d39b34b7343a281cf16672d91da00209caec4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 22:31:56.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee83ceb8caf244228c36b2b7364379d7 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afbc15fe1a754d8aaf2e85d0d71b9d8f : SUCCESS in 7m 35s","accounts_in_message":[],"_revision_number":12},{"id":"902a4890e91b0cc4c4fa17a137da66ef2782d5c6","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-02 19:26:38.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"c67607a7c96e6af53eb44872cd931c8a930d1c55","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-02 19:33:00.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"b187d297f284be1fe12f644e3a9d7da35bc01db6","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-02 19:33:40.000000000","message":"Patch Set 14:\n\nI made another stab at this hamza. Please have a look. Before you modify anything please make sure you are using the most recent change set. Thanks.","accounts_in_message":[],"_revision_number":14},{"id":"e872e5cd652bd29b6c71ac48958193e628b7c1ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 19:48:41.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfa0af011ea6406f96129a10356f9e61 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ba342d997cc42aa937ad2d453a590d8 : SUCCESS in 7m 47s","accounts_in_message":[],"_revision_number":14},{"id":"af3cb6b5c1d99aab33f8179141f13a0debabada7","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-10-02 22:17:44.000000000","message":"Patch Set 14:\n\n\u003e I made another stab at this hamza. Please have a look. Before you\n \u003e modify anything please make sure you are using the most recent\n \u003e change set. Thanks.\n\n:) i updated my branch thanks for the update","accounts_in_message":[],"_revision_number":14},{"id":"763b3b3a2d13c99335ab1a119b8dfd4fc258168a","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-11-19 19:48:28.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"d551db7924196b67eb662576dde6462727b937a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 20:10:03.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5572722811d846fbb158afba2b4ee534 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8b2393e24104d288c4a7b7d5c6261eb : SUCCESS in 15m 52s","accounts_in_message":[],"_revision_number":14},{"id":"fbad0d79f582dc8cd4651efd5acf12f555fde76e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-12-07 23:48:23.000000000","message":"Patch Set 14: Code-Review-1\n\n(9 comments)\n\nThis is a good start, but I think we need to expand on a few topics. This is a complicated feature that can have security implications if not configured correctly, so I would prefer to error on the side of too much information than not enough.","accounts_in_message":[],"_revision_number":14},{"id":"1ec65f401aff0e553e4f382fb4371773f49f43de","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-12-08 00:23:37.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"fe1d5df78d897a5e6bbcedc353289a6099ed47df","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-12-08 01:33:35.000000000","message":"Patch Set 14:\n\n(5 comments)","accounts_in_message":[],"_revision_number":14},{"id":"4faf6cf5094399ae843b1570ce0336b5012b19aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-12-08 22:46:30.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"e76c54dc9ed5fd1877857022db7a7ccd2156fa64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 22:52:15.000000000","message":"Patch Set 15: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a36da70501e4991b698adb89953e1db : FAILURE in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3e82d126ca34e06aa9610730fc24235 : FAILURE in 4m 34s","accounts_in_message":[],"_revision_number":15},{"id":"a084541dba2b849de451b0e174ea2416b9c72d7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-12-18 18:23:23.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"4553d3551f5d5bf60068fe8f031d155846e91681","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 18:43:10.000000000","message":"Patch Set 16: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71a2cf8208fc4d2484d8f701fbfc4204 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91ec22c82812448e9dbc1f648cb92370 : FAILURE in 10m 19s","accounts_in_message":[],"_revision_number":16},{"id":"c7868057947963df1dda34e907615b6129c7ee56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-12-18 18:53:58.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"d735d84f92a8b858ce6c350eeb501c9a5b9a6ac9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 19:06:57.000000000","message":"Patch Set 17: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35d8635900ed498dbe7cba755f3292ab : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b342689b5c848a8ad335b2987ab1c05 : FAILURE in 5m 14s","accounts_in_message":[],"_revision_number":17},{"id":"540c6f189bf8eb089a3667f545ea46dac0136625","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-12-18 19:22:08.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"d23c67697531fa528209fe4159718c123bf2cc25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 19:37:48.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f7381290f634533b9ecaca2d750c609 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59dab199b04446c5b2995713f589f0d6 : SUCCESS in 13m 17s","accounts_in_message":[],"_revision_number":18},{"id":"da3e225676e55be3cd4cfaacac4e1cefd7741dd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2020-12-18 20:46:09.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"96ea9aeb24776430be3c4fe3732fa50f453e89ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 21:00:00.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b37a9ac1c3ce40bdb4e77d9b5015ff60 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/123e0ce48e5b4b5eaf7747b1ffcbe92b : SUCCESS in 7m 56s","accounts_in_message":[],"_revision_number":19},{"id":"2981ec3f303d1ea9206d5395a5582c027b0c3ec0","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-26 17:43:27.000000000","message":"Patch Set 19: Code-Review-1\n\n(5 comments)\n\nOverall this looks great! Could you have a look at my inline suggestions?","accounts_in_message":[],"_revision_number":19},{"id":"f6452d80a19e4117b3ae54a2434de65f19dd7a3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2021-02-26 17:52:23.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"d6b36781811f7b42a338e1781e44bc4144659d9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 18:02:13.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/081d9f153f72481fad5e2c094c72a9e9 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ee53fe0fbec42baac3f08a9fdb159bb : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":20},{"id":"2070872f7ea9c8aa581663b1c32b39a35a4dfe12","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2021-02-26 20:00:09.000000000","message":"Patch Set 20: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":20},{"id":"c039c755cda3e775c8e05b3b3a3b22f9af87f92a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2021-02-26 20:20:08.000000000","message":"Patch Set 20: Code-Review+2","accounts_in_message":[],"_revision_number":20},{"id":"2953066558daad841dec7b64790cde2050d06495","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-19 22:29:03.000000000","message":"Patch Set 20: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":20},{"id":"7afec9b0b57a8367b6b1e153b72e87f61138fb5f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-19 22:35:00.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"33602f4b422d5655f9c40d8a1f3f9806fb993c9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2021-04-19 22:48:08.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"6ea2519b38635e00f7cb25881b04881b252dc325","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2021-04-19 22:51:45.000000000","message":"Patch Set 21:\n\n(2 comments)","accounts_in_message":[],"_revision_number":21},{"id":"ee173280735eba854c128967b05766d04eccb508","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2021-04-19 22:57:02.000000000","message":"Patch Set 21:\n\n(4 comments)","accounts_in_message":[],"_revision_number":21},{"id":"23e12b4602030577269a1a8be06fa25476c3e9e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 22:59:02.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41db311ba97c48e2a3eb8ddc35a3b65c : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea8c68ac216c41d5843ecad3cd164a47 : SUCCESS in 9m 28s","accounts_in_message":[],"_revision_number":21},{"id":"861cbe63ce894fb1a3030d5c5637136bb6b551ac","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-09-24 05:53:22.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":21}],"current_revision_number":21,"current_revision":"f19906865245742e8900532782324f2e9384ed05","revisions":{"506e708ef3c19ed0c75f8bd1d26a096d924d1771":{"kind":"REWORK","_number":1,"created":"2020-09-24 22:38:02.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/26/754226/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/754226/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/754226/1"}}},"commit":{"parents":[{"commit":"2168b4367d483ce420afee4bc1e5fca371cc40b8","subject":"Merge \"fix database sync always get the wrong  value\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2168b4367d483ce420afee4bc1e5fca371cc40b8"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-24 22:36:08.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-24 22:37:56.000000000","tz":0},"subject":"Adding split horizon documentation","message":"Adding split horizon documentation\n\nThe documentaion explains in details how enable the feature\nand configure it to work\n\nChange-Id: Ie5a01f8df00edb7d67ab2db86b889ecac0d0efb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/506e708ef3c19ed0c75f8bd1d26a096d924d1771"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/506e708ef3c19ed0c75f8bd1d26a096d924d1771"}]},"branch":"refs/heads/master"},"1966adf7afbc5ff87958051dd703bf1acdf6da27":{"kind":"REWORK","_number":2,"created":"2020-09-24 23:04:49.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/26/754226/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/754226/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/754226/2"}}},"commit":{"parents":[{"commit":"2168b4367d483ce420afee4bc1e5fca371cc40b8","subject":"Merge \"fix database sync always get the wrong  value\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2168b4367d483ce420afee4bc1e5fca371cc40b8"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-24 22:36:08.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-24 23:04:31.000000000","tz":0},"subject":"Adding split horizon documentation","message":"Adding split horizon documentation\n\nThe documentaion explains in details how enable the feature\nand configure it to work\n\nChange-Id: Ie5a01f8df00edb7d67ab2db86b889ecac0d0efb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1966adf7afbc5ff87958051dd703bf1acdf6da27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1966adf7afbc5ff87958051dd703bf1acdf6da27"}]},"branch":"refs/heads/master"},"d79e16b1b5c0bb23af149b06a8c00badb4e807a2":{"kind":"REWORK","_number":3,"created":"2020-09-24 23:35:02.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/26/754226/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/754226/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/754226/3"}}},"commit":{"parents":[{"commit":"2168b4367d483ce420afee4bc1e5fca371cc40b8","subject":"Merge \"fix database sync always get the wrong  value\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2168b4367d483ce420afee4bc1e5fca371cc40b8"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-24 22:36:08.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-24 23:34:36.000000000","tz":0},"subject":"Adding split horizon documentation","message":"Adding split horizon documentation\n\nThe documentaion explains in details how enable the feature\nand configure it to work\n\nChange-Id: Ie5a01f8df00edb7d67ab2db86b889ecac0d0efb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d79e16b1b5c0bb23af149b06a8c00badb4e807a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d79e16b1b5c0bb23af149b06a8c00badb4e807a2"}]},"branch":"refs/heads/master"},"afc4e103e3b2159dc98ca4098c8eaba08277a237":{"kind":"REWORK","_number":4,"created":"2020-09-25 16:46:32.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/26/754226/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/754226/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/754226/4"}}},"commit":{"parents":[{"commit":"2168b4367d483ce420afee4bc1e5fca371cc40b8","subject":"Merge \"fix database sync always get the wrong  value\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2168b4367d483ce420afee4bc1e5fca371cc40b8"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-24 22:36:08.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-25 16:46:13.000000000","tz":0},"subject":"Adding split horizon documentation","message":"Adding split horizon documentation\n\nThe documentaion explains in details how enable the feature\nand configure it to work\n\nChange-Id: Ie5a01f8df00edb7d67ab2db86b889ecac0d0efb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/afc4e103e3b2159dc98ca4098c8eaba08277a237"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/afc4e103e3b2159dc98ca4098c8eaba08277a237"}]},"branch":"refs/heads/master"},"b81c0e5865d62fd56b94f5d2899cb8b1eb08c099":{"kind":"REWORK","_number":5,"created":"2020-09-25 18:03:44.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/26/754226/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/754226/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/754226/5"}}},"commit":{"parents":[{"commit":"2168b4367d483ce420afee4bc1e5fca371cc40b8","subject":"Merge \"fix database sync always get the wrong  value\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2168b4367d483ce420afee4bc1e5fca371cc40b8"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-24 22:36:08.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-09-25 18:03:31.000000000","tz":-360},"subject":"Adding split horizon documentation","message":"Adding split horizon documentation\n\nThe documentaion explains in details how enable the feature and\nconfigure it to work\n\nChange-Id: Ie5a01f8df00edb7d67ab2db86b889ecac0d0efb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b81c0e5865d62fd56b94f5d2899cb8b1eb08c099"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b81c0e5865d62fd56b94f5d2899cb8b1eb08c099"}]},"branch":"refs/heads/master"},"6153847138124edea469cb4cb95afd8656584272":{"kind":"REWORK","_number":6,"created":"2020-09-25 19:16:23.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/26/754226/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/754226/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/754226/6"}}},"commit":{"parents":[{"commit":"2168b4367d483ce420afee4bc1e5fca371cc40b8","subject":"Merge \"fix database sync always get the wrong  value\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2168b4367d483ce420afee4bc1e5fca371cc40b8"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-24 22:36:08.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-09-25 19:16:12.000000000","tz":-360},"subject":"Adding split horizon documentation","message":"Adding split horizon documentation\n\nThe documentaion explains in details how enable the feature and\nconfigure it to work\n\nChange-Id: Ie5a01f8df00edb7d67ab2db86b889ecac0d0efb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6153847138124edea469cb4cb95afd8656584272"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6153847138124edea469cb4cb95afd8656584272"}]},"branch":"refs/heads/master"},"c7fbbcb1e5031243d9d55e57ebbbd293c8e4a962":{"kind":"REWORK","_number":7,"created":"2020-09-25 19:18:54.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/26/754226/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/754226/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/754226/7"}}},"commit":{"parents":[{"commit":"2168b4367d483ce420afee4bc1e5fca371cc40b8","subject":"Merge \"fix database sync always get the wrong  value\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2168b4367d483ce420afee4bc1e5fca371cc40b8"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-24 22:36:08.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-09-25 19:18:48.000000000","tz":-360},"subject":"Adding split horizon documentation","message":"Adding split horizon documentation\n\nThe documentaion explains in details how enable the feature and\nconfigure it to work\n\nChange-Id: Ie5a01f8df00edb7d67ab2db86b889ecac0d0efb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c7fbbcb1e5031243d9d55e57ebbbd293c8e4a962"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c7fbbcb1e5031243d9d55e57ebbbd293c8e4a962"}]},"branch":"refs/heads/master"},"88eb48ca361afb9dfbdc40bdadd980b3e69ae39c":{"kind":"REWORK","_number":8,"created":"2020-09-28 16:34:33.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/26/754226/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/754226/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/754226/8"}}},"commit":{"parents":[{"commit":"2168b4367d483ce420afee4bc1e5fca371cc40b8","subject":"Merge \"fix database sync always get the wrong  value\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2168b4367d483ce420afee4bc1e5fca371cc40b8"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-24 22:36:08.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-28 16:34:11.000000000","tz":0},"subject":"Adding split horizon documentation","message":"Adding split horizon documentation\n\nThe documentaion explains in details how enable the feature\nand configure it to work\n\nChange-Id: Ie5a01f8df00edb7d67ab2db86b889ecac0d0efb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/88eb48ca361afb9dfbdc40bdadd980b3e69ae39c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/88eb48ca361afb9dfbdc40bdadd980b3e69ae39c"}]},"branch":"refs/heads/master"},"9563aa1272bab59c3dc269574beb51b539fef241":{"kind":"REWORK","_number":9,"created":"2020-09-28 16:40:05.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/26/754226/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/754226/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/754226/9"}}},"commit":{"parents":[{"commit":"2168b4367d483ce420afee4bc1e5fca371cc40b8","subject":"Merge \"fix database sync always get the wrong  value\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2168b4367d483ce420afee4bc1e5fca371cc40b8"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-24 22:36:08.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-28 16:39:41.000000000","tz":0},"subject":"Adding split horizon documentation","message":"Adding split horizon documentation\n\nThe documentaion explains in details how enable the feature\nand configure it to work\n\nChange-Id: Ie5a01f8df00edb7d67ab2db86b889ecac0d0efb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9563aa1272bab59c3dc269574beb51b539fef241"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9563aa1272bab59c3dc269574beb51b539fef241"}]},"branch":"refs/heads/master"},"b15d9ada83f3d4123cdfa3a717ea501e0e7272b2":{"kind":"REWORK","_number":10,"created":"2020-09-28 17:03:14.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/26/754226/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/754226/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/754226/10"}}},"commit":{"parents":[{"commit":"2168b4367d483ce420afee4bc1e5fca371cc40b8","subject":"Merge \"fix database sync always get the wrong  value\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2168b4367d483ce420afee4bc1e5fca371cc40b8"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-24 22:36:08.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-28 17:02:47.000000000","tz":0},"subject":"Adding split horizon documentation","message":"Adding split horizon documentation\n\nThe documentaion explains in details how enable the feature\nand configure it to work\n\nChange-Id: Ie5a01f8df00edb7d67ab2db86b889ecac0d0efb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b15d9ada83f3d4123cdfa3a717ea501e0e7272b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b15d9ada83f3d4123cdfa3a717ea501e0e7272b2"}]},"branch":"refs/heads/master"},"71ee1a5b1e184450cb18ca9a9f790c2279f7e026":{"kind":"REWORK","_number":11,"created":"2020-09-29 01:36:03.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/26/754226/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/754226/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/754226/11"}}},"commit":{"parents":[{"commit":"2168b4367d483ce420afee4bc1e5fca371cc40b8","subject":"Merge \"fix database sync always get the wrong  value\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2168b4367d483ce420afee4bc1e5fca371cc40b8"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-24 22:36:08.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-29 01:35:42.000000000","tz":0},"subject":"Adding split horizon documentation","message":"Adding split horizon documentation\n\nThe documentaion explains in details how enable the feature\nand configure it to work\n\nChange-Id: Ie5a01f8df00edb7d67ab2db86b889ecac0d0efb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/71ee1a5b1e184450cb18ca9a9f790c2279f7e026"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/71ee1a5b1e184450cb18ca9a9f790c2279f7e026"}]},"branch":"refs/heads/master"},"26da658a1b8562a4875771235ea169563664608e":{"kind":"REWORK","_number":12,"created":"2020-09-30 22:20:47.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/26/754226/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/754226/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/754226/12"}}},"commit":{"parents":[{"commit":"2168b4367d483ce420afee4bc1e5fca371cc40b8","subject":"Merge \"fix database sync always get the wrong  value\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2168b4367d483ce420afee4bc1e5fca371cc40b8"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-24 22:36:08.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-30 22:20:17.000000000","tz":0},"subject":"Adding split horizon documentation","message":"Adding split horizon documentation\n\nThe documentaion explains in details how enable the feature\nand configure it to work\n\nChange-Id: Ie5a01f8df00edb7d67ab2db86b889ecac0d0efb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/26da658a1b8562a4875771235ea169563664608e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/26da658a1b8562a4875771235ea169563664608e"}]},"branch":"refs/heads/master"},"4420cc592b2cd5e3e96f7489b7141637fbef3d46":{"kind":"REWORK","_number":13,"created":"2020-10-02 19:26:38.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/26/754226/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/754226/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/754226/13"}}},"commit":{"parents":[{"commit":"2168b4367d483ce420afee4bc1e5fca371cc40b8","subject":"Merge \"fix database sync always get the wrong  value\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2168b4367d483ce420afee4bc1e5fca371cc40b8"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-24 22:36:08.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-10-02 19:25:47.000000000","tz":-360},"subject":"Adding split horizon documentation","message":"Adding split horizon documentation\n\nThe documentaion explains in detail how to enable the split horizon\nDNS feature and configure it.\n\nChange-Id: Ie5a01f8df00edb7d67ab2db86b889ecac0d0efb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4420cc592b2cd5e3e96f7489b7141637fbef3d46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4420cc592b2cd5e3e96f7489b7141637fbef3d46"}]},"branch":"refs/heads/master"},"7cdc165088db693ee273fb7dd16c23c0d587eaf2":{"kind":"REWORK","_number":14,"created":"2020-10-02 19:33:00.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/26/754226/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/754226/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/754226/14"}}},"commit":{"parents":[{"commit":"2168b4367d483ce420afee4bc1e5fca371cc40b8","subject":"Merge \"fix database sync always get the wrong  value\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2168b4367d483ce420afee4bc1e5fca371cc40b8"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-24 22:36:08.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-10-02 19:32:53.000000000","tz":-360},"subject":"Adding split horizon documentation","message":"Adding split horizon documentation\n\nThe documentaion explains in detail how to enable the split horizon\nDNS feature and configure it.\n\nChange-Id: Ie5a01f8df00edb7d67ab2db86b889ecac0d0efb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7cdc165088db693ee273fb7dd16c23c0d587eaf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7cdc165088db693ee273fb7dd16c23c0d587eaf2"}]},"branch":"refs/heads/master"},"d533b312e9807e3b6024ddf9d5040ff644d72211":{"kind":"REWORK","_number":15,"created":"2020-12-08 22:46:30.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/26/754226/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/754226/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/754226/15"}}},"commit":{"parents":[{"commit":"2168b4367d483ce420afee4bc1e5fca371cc40b8","subject":"Merge \"fix database sync always get the wrong  value\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2168b4367d483ce420afee4bc1e5fca371cc40b8"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-24 22:36:08.000000000","tz":0},"committer":{"name":"Hamza Alqtaishat","email":"hamalq@verizonmedia.com","date":"2020-12-08 22:45:55.000000000","tz":0},"subject":"Adding split horizon documentation","message":"Adding split horizon documentation\n\nThe documentaion explains in detail how to enable the split horizon\nDNS feature and configure it.\n\nChange-Id: Ie5a01f8df00edb7d67ab2db86b889ecac0d0efb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d533b312e9807e3b6024ddf9d5040ff644d72211"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d533b312e9807e3b6024ddf9d5040ff644d72211"}]},"branch":"refs/heads/master"},"c1712dd5549878db0ff5940ccafd6d00777db733":{"kind":"REWORK","_number":16,"created":"2020-12-18 18:23:23.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/26/754226/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/754226/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/754226/16"}}},"commit":{"parents":[{"commit":"1a9c81a957f7dfcc71d05da4f82c834e2498532b","subject":"Add support for split view feature to designate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1a9c81a957f7dfcc71d05da4f82c834e2498532b"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-24 22:36:08.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-12-18 18:22:53.000000000","tz":0},"subject":"Adding split horizon documentation","message":"Adding split horizon documentation\n\nThe documentaion explains in detail how to enable the split horizon\nDNS feature and configure it.\n\nChange-Id: Ie5a01f8df00edb7d67ab2db86b889ecac0d0efb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c1712dd5549878db0ff5940ccafd6d00777db733"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c1712dd5549878db0ff5940ccafd6d00777db733"}]},"branch":"refs/heads/master"},"5a6a313a1168db2815160450ea7a7b8aec8f5ec7":{"kind":"REWORK","_number":17,"created":"2020-12-18 18:53:58.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/26/754226/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/754226/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/754226/17"}}},"commit":{"parents":[{"commit":"1a9c81a957f7dfcc71d05da4f82c834e2498532b","subject":"Add support for split view feature to designate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1a9c81a957f7dfcc71d05da4f82c834e2498532b"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-24 22:36:08.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-12-18 18:53:33.000000000","tz":0},"subject":"Adding split horizon documentation","message":"Adding split horizon documentation\n\nThe documentaion explains in detail how to enable the split horizon\nDNS feature and configure it.\n\nChange-Id: Ie5a01f8df00edb7d67ab2db86b889ecac0d0efb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5a6a313a1168db2815160450ea7a7b8aec8f5ec7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5a6a313a1168db2815160450ea7a7b8aec8f5ec7"}]},"branch":"refs/heads/master"},"cb0416bc3aab6ed32de407e34cf85846b2c7d97b":{"kind":"REWORK","_number":18,"created":"2020-12-18 19:22:08.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/26/754226/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/754226/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/754226/18"}}},"commit":{"parents":[{"commit":"1a9c81a957f7dfcc71d05da4f82c834e2498532b","subject":"Add support for split view feature to designate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1a9c81a957f7dfcc71d05da4f82c834e2498532b"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-24 22:36:08.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-12-18 19:21:42.000000000","tz":0},"subject":"Adding split horizon documentation","message":"Adding split horizon documentation\n\nThe documentaion explains in detail how to enable the split horizon\nDNS feature and configure it.\n\nChange-Id: Ie5a01f8df00edb7d67ab2db86b889ecac0d0efb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cb0416bc3aab6ed32de407e34cf85846b2c7d97b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cb0416bc3aab6ed32de407e34cf85846b2c7d97b"}]},"branch":"refs/heads/master"},"77430354e3210850e2d176755c5f198b28ae4a19":{"kind":"REWORK","_number":19,"created":"2020-12-18 20:46:09.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/26/754226/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/754226/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/754226/19"}}},"commit":{"parents":[{"commit":"1a9c81a957f7dfcc71d05da4f82c834e2498532b","subject":"Add support for split view feature to designate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1a9c81a957f7dfcc71d05da4f82c834e2498532b"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-24 22:36:08.000000000","tz":0},"committer":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-12-18 20:45:33.000000000","tz":0},"subject":"Adding split horizon documentation","message":"Adding split horizon documentation\n\nThe documentaion explains in detail how to enable the split horizon\nDNS feature and configure it.\n\nChange-Id: Ie5a01f8df00edb7d67ab2db86b889ecac0d0efb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/77430354e3210850e2d176755c5f198b28ae4a19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/77430354e3210850e2d176755c5f198b28ae4a19"}]},"branch":"refs/heads/master"},"3517c67de6d5a90bfd2f8ef6a474574971cfc9f8":{"kind":"REWORK","_number":20,"created":"2021-02-26 17:52:23.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/26/754226/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/754226/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/754226/20"}}},"commit":{"parents":[{"commit":"1a9c81a957f7dfcc71d05da4f82c834e2498532b","subject":"Add support for split view feature to designate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1a9c81a957f7dfcc71d05da4f82c834e2498532b"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-24 22:36:08.000000000","tz":0},"committer":{"name":"hamza","email":"hamalq@verizonmedia.com","date":"2021-02-26 17:42:18.000000000","tz":0},"subject":"Adding split horizon documentation","message":"Adding split horizon documentation\n\nThe documentaion explains in detail how to enable the split horizon\nDNS feature and configure it.\n\nChange-Id: Ie5a01f8df00edb7d67ab2db86b889ecac0d0efb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3517c67de6d5a90bfd2f8ef6a474574971cfc9f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3517c67de6d5a90bfd2f8ef6a474574971cfc9f8"}]},"branch":"refs/heads/master"},"f19906865245742e8900532782324f2e9384ed05":{"kind":"REWORK","_number":21,"created":"2021-04-19 22:48:08.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/26/754226/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/754226/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/754226/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/754226/21"}}},"commit":{"parents":[{"commit":"1a9c81a957f7dfcc71d05da4f82c834e2498532b","subject":"Add support for split view feature to designate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1a9c81a957f7dfcc71d05da4f82c834e2498532b"}]}],"author":{"name":"hamalq","email":"hamalq@verizonmedia.com","date":"2020-09-24 22:36:08.000000000","tz":0},"committer":{"name":"hamza","email":"hamalq@verizonmedia.com","date":"2021-04-19 22:35:33.000000000","tz":0},"subject":"Adding split horizon documentation","message":"Adding split horizon documentation\n\nThe documentaion explains in detail how to enable the split horizon\nDNS feature and configure it.\n\nChange-Id: Ie5a01f8df00edb7d67ab2db86b889ecac0d0efb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f19906865245742e8900532782324f2e9384ed05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f19906865245742e8900532782324f2e9384ed05"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
