)]}'
{"id":"openstack%2Fdesignate~754957","triplet_id":"openstack%2Fdesignate~master~I9e9badabd588b94055eccb8298d685b2ccf60095","project":"openstack/designate","branch":"master","topic":"754957","hashtags":[],"change_id":"I9e9badabd588b94055eccb8298d685b2ccf60095","subject":"Remove the unused coding style modules","status":"MERGED","created":"2020-09-29 09:57:01.000000000","updated":"2020-10-15 03:12:51.000000000","submitted":"2020-10-15 03:11:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754957-1602731476876-ffcc631f","meta_rev_id":"fd08ee0b21bfc2b7edd4c749f04af503bff56044","_number":754957,"virtual_id_number":754957,"owner":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"date":"2020-10-15 00:56:42.000000000","_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-15 03:11:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2020-10-14 15:49:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-14 22:50:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-14 22:50:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-14 15:49:02.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-10-14 22:50:07.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-10-15 03:11:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b54b5cc049e5a936c591a20fbd181920dbcbcc8","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-09-29 09:57:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f01c72c71e1510648d6857150a02672a5c236440","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 11:03:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6685733cfb4e4720af6bba99d606c3e6 : SUCCESS in 37m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/26d481af0db34847835c1a55a99b9f6d : SUCCESS in 36m 42s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c9dcfe33f3224cd988532f746124f8f7 : SUCCESS in 57m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c5cca124a5634e368e086efc34169618 : SUCCESS in 39m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7faa9127969e4e078571f011a83ae21d : SUCCESS in 42m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de3440ede8f847488b9ca9a0978f1a04 : SUCCESS in 5m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12cf2254fc6146ed87d5373d5e1c69d8 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53d84473e64046f785f4c12cd4152b92 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d00a73cc9cc5441e9fcf3503431d89fa : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5012223e8494033ba863a271885b320 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ce4cfb7ac4b48099e6401fd985bd7bd : SUCCESS in 8m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b7d048cb963147658fc20840bf1d9a5d : SUCCESS in 3m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9fb4f6bb494342d681a6e5c65a1f89cf : SUCCESS in 35m 20s","accounts_in_message":[],"_revision_number":1},{"id":"1e424d7e48cd0e0dd4ef6ee207dab5262ddb52cc","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-13 22:27:27.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThanks Shuming!","accounts_in_message":[],"_revision_number":1},{"id":"36858513987dee56bffa4994610c707375b923c4","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-10-14 02:09:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7e6cc3053240f8c1d346cf2de933178aef4ff61b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 03:03:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a5ddb6d25c2349eba622d1845d3e5a70 : SUCCESS in 5m 39s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/56e1b3026d9945a68efe6c6f654a73eb : SUCCESS in 45m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/21b7b32a1b77431c84ba855fbd118ebc : SUCCESS in 33m 07s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c2e02bd5139f451cab466d560250b59d : SUCCESS in 51m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e8368f4178414ec1a2d96458bb690855 : FAILURE in 24m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/953095910856421f8a481c214fdf406f : SUCCESS in 37m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3089f3984ff42d3b2aec8ab52d9c62f : SUCCESS in 4m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c5f13ad5b944a3ba19e1abbc4ce0d96 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9498069db0c44a8b959610d6eac8770f : FAILURE in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2bc93bf650d43918bf1779ed727f04d : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/484f02bbae724c90926ab45365dc9ad1 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9db4f7c9e7e24719a03d2718dcf58287 : SUCCESS in 8m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/976e70bac8ae456486c710f07325b55b : SUCCESS in 3m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/435fae907f1a4a24b9e96dd39e722aae : SUCCESS in 4m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6f082b75285a4d32b3ae83b6d8de695b : SUCCESS in 38m 06s","accounts_in_message":[],"_revision_number":2},{"id":"0c8c39c28408b79f3dad4e1d8b922875e422c4eb","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-10-14 03:07:09.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a655f758b9e377125e51311be42478c2b3d6aca2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 04:06:34.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/96437dd56c964714ac44b357106d25c9 : SUCCESS in 3m 31s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cfe823bf5c1749e1be7705921da676e5 : SUCCESS in 42m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1081b9b5e6244f0a96cea729b7834f19 : SUCCESS in 40m 54s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ca59cf7cdef94955a406b4a5df0976f3 : SUCCESS in 56m 10s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/064d8ed50e1b4c52937328c0797683dd : SUCCESS in 41m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8faca1cd0c824dc183d9d2003aa33ba7 : SUCCESS in 38m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93a971d85ca047d5a8853886ecd1e505 : SUCCESS in 4m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9d93b86258d4711a9565ba3d3743e43 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d22784eb8dfd485d9a4216139aceca1e : FAILURE in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd21cb770f174fb89ea5ff46986a19b5 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b37bdaeeae364ac68876dd2044d8fd9d : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84fe061ecf9742bea485aa4aa8e18ffc : SUCCESS in 6m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2f4e3e1790114b55b96b2d9c83ee40d0 : SUCCESS in 1m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e61feb5c8e444aafb6fff69bf682ab94 : SUCCESS in 3m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0f7f8c6167594b71a42b96e2d3ec5b8e : SUCCESS in 37m 41s","accounts_in_message":[],"_revision_number":2},{"id":"4d97e3774ebfcc5f318fea96dae01da1d7688002","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-14 12:16:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"12394479727deab0e26bd3c64cad8ce5a3316ca2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 13:26:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8259b1abe47649b6b9a4c2039c5e51a1 : SUCCESS in 39m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ab6f5c3b30ee419d90453e652b6b576c : SUCCESS in 39m 52s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/63afcac5c5b2487392b60b12a3516326 : SUCCESS in 51m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3b9df845fb43417ab53a61e645ab7abf : SUCCESS in 1h 04m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e078f053ae314c8f9b3644711e5349e5 : SUCCESS in 47m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b05dd26ddd1549a3b04dd21f1ecd44b6 : SUCCESS in 6m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a460b86d3074db6b448df89592aedc9 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/152d13ae1f944b15bc2b2f23e1c47aca : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/165d167e34f94a6e8aecce8f8a0e956e : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a05f010fe3f3400a89a3fb49b79ae8fc : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c7e26e723914b10850d268158532e7d : SUCCESS in 7m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/61ebae5962d54372979ee245cc4a2991 : SUCCESS in 2m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8714e7e3ab58444a80ba80dfef22595b : SUCCESS in 32m 43s","accounts_in_message":[],"_revision_number":3},{"id":"f8aff92e6b3a8c979ff03bba9197c7b087f1a616","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-14 14:22:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e716469033fab744243b9b5514232624e901b1a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 15:26:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/763172baf7544f5782cbb7784fe72fb7 : SUCCESS in 40m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ec3c3569b91e47b1aa9a90c8c6199a14 : SUCCESS in 46m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5c5fd38400b24ffdb8fb208f22c8e7a3 : SUCCESS in 57m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6ed9c01c2e704b49903c228e2924e510 : SUCCESS in 42m 33s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6420954893a843a8a16253227c352550 : SUCCESS in 42m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7bb11579bb9041de99780f39a71d350e : SUCCESS in 6m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd203bcfe541411fb365bf9e48a0f64e : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42321fd2799b4c6db93d0ddc5f676868 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2801c8b318964376b83a25e8de11158c : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62df01cf994741158f8fcc294e667999 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3988540bc92f47ec9644aab6b3c9e435 : SUCCESS in 6m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c0eebce2c0c84707a4d09576fb75ce7b : SUCCESS in 2m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ec112b3142b44a72bc13151fac519191 : SUCCESS in 42m 08s","accounts_in_message":[],"_revision_number":4},{"id":"390f08e4e620a42bc3ec66156f0a8850a24fc9fa","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-14 15:49:02.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0399a96ec5cc9fcb84d0ba4155d9bd453ab507da","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-14 22:50:07.000000000","message":"Patch Set 4: Workflow+1 Review-Priority+2\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"5b9d5c0322a6f90a20656ad5e1c85e6569039cab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 22:50:20.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"76cd092d635ed0b4ee954e24de84f76d98d71965","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 00:14:47.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f4388841883d4af2b698601f490d7c59 : TIMED_OUT in 1h 13m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/895c6f137d614a4e8a35c72ce7469ea6 : SUCCESS in 37m 46s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f5431d31ec50417eb049bb2a3a2f4abb : SUCCESS in 1h 14m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3003586447c7406e978dd0758090a259 : SUCCESS in 56m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/760176a8beb643fdaae5535e98d4b517 : SUCCESS in 39m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2740c7b7d264f64b18c218cb9b2d275 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/917eec4887b843448c8f44ee648176cf : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2923611cd46d4da9ac84032d13d64440 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b3eb00c9f2745868229a93c0c90a72b : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6475923904af4dcdbcb9e281fc3618ff : SUCCESS in 11m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f2a77f837f0b463b898def12dbafbe10 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":4},{"id":"e6fc904bc4ffb889d5323c4727fa47bbda4eebb2","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-10-15 00:56:42.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"43be727d9f3e359d72f53e6fe54737f471ca69cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 02:08:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ca414d4a6f764d23b657df63282b03ac : SUCCESS in 33m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bd99c37fc63945c7ae71557dc5e3fd5b : SUCCESS in 33m 13s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/77ea9e19718244fd99eb16b92399672a : SUCCESS in 58m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d038f135bd4d4613b6219877ed138555 : SUCCESS in 35m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a9a7bdd005054039ae9bdc672d40a641 : SUCCESS in 36m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65a765d226184aa7ab9c38620c72f629 : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff2af18b3b3946e6b8948dcfebcef5c9 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/172a7b4958044df8b6aded43aac3fc93 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd254dc99a074eaca6be84892066e437 : SUCCESS in 8m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dab08b2c5f4845a4973094053aab7e68 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ddbd6053e1943ef9e5a1d8d8a29188b : SUCCESS in 7m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f97087b015f04738af0c9a82a95db559 : SUCCESS in 2m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f9d184441fb143fdb9e4f6e570d948a6 : SUCCESS in 37m 46s","accounts_in_message":[],"_revision_number":4},{"id":"11b0cb6cb284c00117c4cd8fcc933d0bd0b0c426","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 02:09:04.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3244228445054affb74e7bdd93f701e5d817df28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 03:11:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"2d6c293c054fc68edfa4331bd337c51c972bcf6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 03:11:16.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1d5354eeebfd4663a4b9a82f61827b8c : SUCCESS in 36m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c091254f0a9043b1b191e1806f78d59d : SUCCESS in 37m 59s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4f755e5a66a347f99263fda23e8da6a6 : SUCCESS in 50m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0ea3f5ba06044460a3b90136784e49a0 : SUCCESS in 40m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d3ae7814868d4d44ab661bc3f6c746e2 : SUCCESS in 40m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d36f6a865111417a910589f7ae67a934 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e39269642f3a46519dfc99811812e7fe : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98a1a4f101444dcbaca4e0aeb5f7d5bf : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f842b1fedcb4762aac28229df47218c : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddf5a2d111d64f80bec08501137ff0db : SUCCESS in 8m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/83eb635187a940029078a39ccf64a60a : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":4},{"id":"723c52421861954bed0b158e6d2231994719a134","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 03:12:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/592d20aaa3574e3cbfd8d2727e267eb1 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0d277e9cbb42404b1b1082afdfa5bcfc49691e6f","revisions":{"0850d2d3e53897dcf10afea588bf97a1eff3b815":{"kind":"REWORK","_number":1,"created":"2020-09-29 09:57:01.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/57/754957/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/57/754957/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/754957/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/754957/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/754957/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/57/754957/1"}}},"commit":{"parents":[{"commit":"27dbb22c9e66f904845c808de94f7341de69d895","subject":"Merge \"Add link to openstack client commands in cli doc section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/27dbb22c9e66f904845c808de94f7341de69d895"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-09-29 09:56:03.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-09-29 09:56:57.000000000","tz":480},"subject":"Remove the unused coding style modules","message":"Remove the unused coding style modules\n\nPython modules related to coding style checks (listed in blacklist.txt in\nopenstack/requirements repo) are dropped from lower-constraints.txt\nas they are not actually used in tests (other than pep8).\n\nmore info: https://github.com/openstack/requirements/blob/master/blacklist.txt\n\nChange-Id: I9e9badabd588b94055eccb8298d685b2ccf60095\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0850d2d3e53897dcf10afea588bf97a1eff3b815"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0850d2d3e53897dcf10afea588bf97a1eff3b815"}]},"branch":"refs/heads/master"},"f0b9e5470f68e3407590f9162f42b6092e40a29f":{"kind":"REWORK","_number":2,"created":"2020-10-14 02:09:09.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/57/754957/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/57/754957/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/754957/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/754957/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/754957/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/57/754957/2"}}},"commit":{"parents":[{"commit":"27dbb22c9e66f904845c808de94f7341de69d895","subject":"Merge \"Add link to openstack client commands in cli doc section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/27dbb22c9e66f904845c808de94f7341de69d895"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-09-29 09:56:03.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-10-14 02:08:29.000000000","tz":480},"subject":"Remove the unused coding style modules","message":"Remove the unused coding style modules\n\nPython modules related to coding style checks (listed in blacklist.txt in\nopenstack/requirements repo) are dropped from lower-constraints.txt\nas they are not actually used in tests (other than pep8).\n\nmore info: https://github.com/openstack/requirements/blob/master/blacklist.txt\n\nChange-Id: I9e9badabd588b94055eccb8298d685b2ccf60095\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0b9e5470f68e3407590f9162f42b6092e40a29f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0b9e5470f68e3407590f9162f42b6092e40a29f"}]},"branch":"refs/heads/master"},"ea6fff8a3ac2463ba8055a33a2908df0fc554cfa":{"kind":"REWORK","_number":3,"created":"2020-10-14 12:16:30.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/57/754957/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/57/754957/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/754957/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/754957/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/754957/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/57/754957/3"}}},"commit":{"parents":[{"commit":"27dbb22c9e66f904845c808de94f7341de69d895","subject":"Merge \"Add link to openstack client commands in cli doc section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/27dbb22c9e66f904845c808de94f7341de69d895"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-09-29 09:56:03.000000000","tz":480},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-10-14 12:16:15.000000000","tz":-360},"subject":"Remove the unused coding style modules","message":"Remove the unused coding style modules\n\nPython modules related to coding style checks (listed in blacklist.txt\nin openstack/requirements repo) are dropped from lower-constraints.txt\nas they are not actually used in tests (other than pep8).\n\nmore info: https://github.com/openstack/requirements/blob/master/blacklist.txt\n\nChange-Id: I9e9badabd588b94055eccb8298d685b2ccf60095\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ea6fff8a3ac2463ba8055a33a2908df0fc554cfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ea6fff8a3ac2463ba8055a33a2908df0fc554cfa"}]},"branch":"refs/heads/master"},"0d277e9cbb42404b1b1082afdfa5bcfc49691e6f":{"kind":"REWORK","_number":4,"created":"2020-10-14 14:22:40.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/57/754957/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/57/754957/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/754957/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/754957/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/57/754957/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/57/754957/4"}}},"commit":{"parents":[{"commit":"27dbb22c9e66f904845c808de94f7341de69d895","subject":"Merge \"Add link to openstack client commands in cli doc section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/27dbb22c9e66f904845c808de94f7341de69d895"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2020-09-29 09:56:03.000000000","tz":480},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2020-10-14 14:21:37.000000000","tz":-360},"subject":"Remove the unused coding style modules","message":"Remove the unused coding style modules\n\nPython modules related to coding style checks (listed in blacklist.txt\nin openstack/requirements repo) are dropped from lower-constraints.txt\nas they are not actually used in tests (other than pep8).\n\nmore info: https://github.com/openstack/requirements/blob/master/blacklist.txt\n\nChange-Id: I9e9badabd588b94055eccb8298d685b2ccf60095\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0d277e9cbb42404b1b1082afdfa5bcfc49691e6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0d277e9cbb42404b1b1082afdfa5bcfc49691e6f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
